)]}'
{"id":"openstack%2Fmanila-tempest-plugin~665011","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I11887de48c31bc5ea4c097e36c0f0da238d613b2","project":"openstack/manila-tempest-plugin","branch":"master","topic":"TrivialFix","hashtags":[],"change_id":"I11887de48c31bc5ea4c097e36c0f0da238d613b2","subject":"Clarify help text of network options","status":"MERGED","created":"2019-06-12 20:19:00.000000000","updated":"2019-06-25 21:58:53.000000000","submitted":"2019-06-25 21:58:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665011-1561499933807-e8d46088","meta_rev_id":"982477d8319adff7234e6da2ef6a3b9de39e69ec","_number":665011,"virtual_id_number":665011,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-25 18:56:22.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2019-06-12 23:38:49.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-06-25 21:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-25 11:13:25.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-25 19:03:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-25 11:13:25.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 23:38:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-25 11:13:25.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-06-25 19:03:58.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2019-06-25 21:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49a187acd4a7cb0a0fd2a50b721db86561d518cd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-12 20:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"920590722b3e00ee4772ce1e23f8cc519e50b6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 22:16:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/11/665011/1/check/openstack-tox-pep8/ff3e495/ : SUCCESS in 4m 58s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-mysql-generic/410b98a/ : FAILURE in 1h 17m 40s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-postgres-container/c04779b/ : SUCCESS in 50m 37s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-postgres-zfsonlinux/4ab1691/ : FAILURE in 1h 07m 36s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/20e33c2/ : SUCCESS in 1h 01m 17s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-generic-no-share-servers/a50a39f/ : SUCCESS in 1h 13m 55s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-scenario/b77db52/ : FAILURE in 1h 49m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-cephfs-native/fb8bdd9/ : SUCCESS in 25m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/78e87be/ : SUCCESS in 23m 50s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-dummy/771dad1/ : SUCCESS in 39m 20s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-lvm/fbe21ad/ : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5b749ff993562c91feba4e6a1653260c433fe613","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-12 23:38:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/11/665011/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8e40dc9/ : SUCCESS in 1h 10m 51s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/11/665011/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/c182735/ : SUCCESS in 1h 22m 08s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"5d7c513f0731519ce3fc86108b849b8d7ec5f86c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-25 11:13:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"49fc7e63909dd5a6296a64f7c041859de32699c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 11:13:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6dcbe3e7f81b1b2edf9cf935f88cff3964b2cc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 14:03:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/665011/1/gate/openstack-tox-pep8/4989fd9/ : SUCCESS in 17m 50s\n- manila-tempest-minimal-dsvm-dummy finger://ze11.openstack.org/13b643d9938c4b188f9a4cb7210fc0b1 : RETRY_LIMIT in 35m 44s\n- manila-tempest-minimal-dsvm-lvm finger://ze03.openstack.org/b85d34dfc8a04dcabac53b5ea380bcea : RETRY_LIMIT in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c32a3b98c2054dfae1ea0faea07455da46f2a132","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-25 18:56:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"414b7c4400dcbca46dfc92a2665749fcb9e63f31","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2019-06-25 19:03:58.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e6a303a477d67306f9d51abd23504b8b3c363cf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/11/665011/1/check/openstack-tox-pep8/0102ecd/ : SUCCESS in 2m 44s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-mysql-generic/9b80d65/ : SUCCESS in 1h 25m 41s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-postgres-container/49ef153/ : SUCCESS in 49m 57s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-postgres-zfsonlinux/9a3cbe7/ : FAILURE in 1h 05m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/ebb0baf/ : SUCCESS in 59m 24s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-generic-no-share-servers/5e75941/ : SUCCESS in 1h 00m 20s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/11/665011/1/check/manila-tempest-dsvm-scenario/fee6d99/ : FAILURE in 1h 51m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-cephfs-native/b0d22f2/ : SUCCESS in 24m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/cc6e40c/ : FAILURE in 1h 18m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-dummy/c1178a8/ : SUCCESS in 53m 38s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/11/665011/1/check/manila-tempest-minimal-dsvm-lvm/7f8bff6/ : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b691d301cfc856d382d5d772d2bf3d3481fd395b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:55:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffa8b3f4b68bfd193bd9fc4dd4513f664c5bf49a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 21:58:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f765c467331eee5d5add2633bf5a41132be06c3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 21:58:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/11/665011/1/gate/openstack-tox-pep8/7fe5195/ : SUCCESS in 3m 15s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/11/665011/1/gate/manila-tempest-minimal-dsvm-dummy/ab5c741/ : SUCCESS in 38m 33s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/11/665011/1/gate/manila-tempest-minimal-dsvm-lvm/5290540/ : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90c354b526ddbaf2ab0bb69d9c34ef85ae570d9c","revisions":{"90c354b526ddbaf2ab0bb69d9c34ef85ae570d9c":{"kind":"REWORK","_number":1,"created":"2019-06-12 20:19:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/11/665011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/665011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/665011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/665011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/665011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/665011/1"}}},"commit":{"parents":[{"commit":"a4904978b4c28e1acbfb12d26db63085325f94c3","subject":"Merge \"Scenario test: Create/shrink share and write data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a4904978b4c28e1acbfb12d26db63085325f94c3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-06-12 20:18:51.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-06-12 20:18:51.000000000","tz":-420},"subject":"Clarify help text of network options","message":"Clarify help text of network options\n\nTrivialFix\n\nChange-Id: I11887de48c31bc5ea4c097e36c0f0da238d613b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90c354b526ddbaf2ab0bb69d9c34ef85ae570d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90c354b526ddbaf2ab0bb69d9c34ef85ae570d9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
