)]}'
{"id":"openstack%2Fmanila-tempest-plugin~668597","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I48b0f21fe75fae623e7e513cf2890f2c10a80317","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/scenario-tests","hashtags":[],"change_id":"I48b0f21fe75fae623e7e513cf2890f2c10a80317","subject":"Fix floatingip allocation","status":"MERGED","created":"2019-07-02 05:42:31.000000000","updated":"2020-04-25 10:28:56.000000000","submitted":"2020-04-25 02:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668597-1587783554067-a0c18567","meta_rev_id":"849455c7fa9b30d8a7549a6602ac249833742383","_number":668597,"virtual_id_number":668597,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-04-24 23:53:04.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-25 03:59:16.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-25 02:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-04-25 01:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":2,"date":"2020-04-25 00:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-25 00:50:04.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2020-04-25 01:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 23:53:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-25 00:36:24.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-04-25 00:50:04.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-04-25 01:54:26.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-25 02:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-25 03:59:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20608fdcbafbcbec058d3ef9edb65a74c8dc5d84","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2019-07-02 05:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1872f98e513e3e9aae6c7a6e253fb545ab83f13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 07:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/668597/1/check/openstack-tox-pep8/676c5f2/ : SUCCESS in 4m 29s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/97/668597/1/check/manila-tempest-dsvm-mysql-generic/e5263a7/ : FAILURE in 1h 51m 29s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/97/668597/1/check/manila-tempest-dsvm-postgres-container/82a55ac/ : SUCCESS in 41m 58s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/97/668597/1/check/manila-tempest-dsvm-postgres-zfsonlinux/95ec008/ : FAILURE in 53m 18s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/97/668597/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/7e0dd78/ : SUCCESS in 1h 19m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/97/668597/1/check/manila-tempest-dsvm-generic-no-share-servers/a51c3ac/ : SUCCESS in 1h 30m 15s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/97/668597/1/check/manila-tempest-dsvm-scenario/1aecde4/ : FAILURE in 1h 36m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/97/668597/1/check/manila-tempest-minimal-dsvm-cephfs-native/75da29b/ : SUCCESS in 40m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/97/668597/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/6fe9c6e/ : FAILURE in 1h 09m 18s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/97/668597/1/check/manila-tempest-minimal-dsvm-dummy/c6f227f/ : SUCCESS in 36m 08s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/97/668597/1/check/manila-tempest-minimal-dsvm-lvm/066f13c/ : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c403b782279390d1f932aae217a01b7e7ac94df7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-02 09:03:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/97/668597/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5ac5f91/ : SUCCESS in 1h 12m 30s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/97/668597/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/891ac51/ : FAILURE in 1h 22m 11s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7a0506a80aa25146b83fd2748cc6acc36f1e60d5","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2019-07-02 11:55:43.000000000","message":"Patch Set 1:\n\nnetapp-ci-recheck","accounts_in_message":[],"_revision_number":1},{"id":"220734369d1687602e4818f721fdd37ae27286e7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-02 13:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/97/668597/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e7e2aa2/ : SUCCESS in 1h 21m 56s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/97/668597/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/eb48b6d/ : SUCCESS in 1h 36m 21s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f3c514cd571c305fbc85df9f9e9359944bfe6303","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-23 05:10:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks for your contribution. Please see comment inline..","accounts_in_message":[],"_revision_number":1},{"id":"c62e83bb6e4688dff1dc60ae47a1492a0c328ece","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 04:35:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f6d6626031736e9891e618c15aca916b11df2b3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 04:38:13.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"485b82611782527420ebab7613f8a8152c83e2fa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 05:22:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b99d56581de33537589cf6ae048b1d328eec97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:51:56.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c44de97189b9423fa34315963decbe83 : FAILURE in 3m 18s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0471a85536d74d29b16b8e5dcb10a624 : SUCCESS in 1h 24m 21s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/06f5164dab6c41f9b34fef87b53977d6 : SUCCESS in 55m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f00b13eab9944de3a33e4d6db8898510 : FAILURE in 47m 04s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e08508e9bee6459f858a7736dab40bf2 : SUCCESS in 55m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/32f6ee0b1a2641dba9333c8ca73bda7d : FAILURE in 46m 39s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dcd74a2f8bc24ebbbb14d408c6e86644 : FAILURE in 1h 13m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e7d4ee61922e4362b4c7998e306f7745 : SUCCESS in 25m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a4856f1d808940fd9830bdc1378356d6 : SUCCESS in 1h 19m 01s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7d1207f2935e4ea7bc1645b094396977 : FAILURE in 47m 27s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/1829b9ce410f4e8c8464446a186826c3 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":3},{"id":"d4ed4dd760f3ae307d3da4eb96df29af8b8e71d7","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-24 11:40:31.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ee9e980162822555a2574d08a1cd65e28868bf58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:16:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ab2d38528544f09d0e17f3c6876a39 : SUCCESS in 3m 21s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/eb07c89e6a334db6bb88f3341a52c529 : SUCCESS in 1h 26m 58s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/68830dbebedf4b6083e61bfe11e0d9ab : SUCCESS in 55m 31s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/80e0ebbc2d9e454a914b86c738f3d5eb : FAILURE in 48m 50s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/106c8f8f4a554a5db24d2667bcc29972 : SUCCESS in 53m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/869ff12d7df6496298691f36bfc8840b : FAILURE in 48m 06s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28ec451aad284eaa843da715d0b89c19 : FAILURE in 1h 07m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f13191b179b540c0b4fff3dc9e8941ce : SUCCESS in 25m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5c4295117eba476280467c36b406ed19 : FAILURE in 1h 12m 36s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f87e793225034ed9b5127621c7092281 : SUCCESS in 48m 07s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ee268ec8ce3840189ba347f2dfc9726c : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":4},{"id":"8d7ced8e7068a53368cd9bb2b54bb794e80abb4d","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-24 14:11:44.000000000","message":"Patch Set 4: Code-Review-1\n\nThere is at least one job that is not happy with this change[1].\n\n[1] https://6d6cb428289f3bbfd710-31f87add5c30895815b1e3afff00e73d.ssl.cf1.rackcdn.com/668597/4/check/manila-tempest-dsvm-scenario/28ec451/job-output.txt","accounts_in_message":[],"_revision_number":4},{"id":"a492a4bdb2426b78ac946c801d231f6ca50667d6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-24 15:21:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/97/668597/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2a0b64e/ : FAILURE in 2h 04m 15s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/97/668597/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/bd9d703/ : FAILURE in 1h 32m 52s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7eb1639780f9ca5e43b447a5e01b8f0bc90de288","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 17:06:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9167e379922f351449f9f3840c19b2ef755d7460","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 17:13:29.000000000","message":"Patch Set 5:\n\nHi there Shi Yan! I\u0027m modifying this change and trying to get this in, I hope that\u0027s alright with you :)","accounts_in_message":[],"_revision_number":5},{"id":"2ba1112672a9e5b54e607077a8b9a1bcd8baf85c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 17:28:38.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"9e383a7bc5994d346d4a9a383708cd631135bcb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:18:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696df74dc70b4e2c865631315a5545ec : SUCCESS in 6m 36s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/736c6f6b83a049b89d5b3d4a2671ae40 : FAILURE in 1h 17m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/9cf7e20c41134190a40b99126a060e07 : SUCCESS in 49m 25s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9b277bd49c924ef9a9732c07527a536f : FAILURE in 1h 15m 27s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/dbffc34287b64a96b6c5068a394bf31c : SUCCESS in 58m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e1b14fb1870e4dad9b78058c79114a95 : FAILURE in 55m 23s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3b0cf8f2e674f0db119a37046a9f0f4 : FAILURE in 1h 46m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/206684d99ed94c189b83c70d4da935d0 : SUCCESS in 32m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2bb53b018717416e8a5f0e9a02b6b599 : SUCCESS in 1h 15m 02s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8df5f413e2cb418490e3b15c41d1a8c8 : SUCCESS in 43m 17s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a158d3ede0c947fa8f76654cde241cd5 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":6},{"id":"bb485bf310a3c880a6c75c2e42e4521d9d8ba53f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-24 19:52:36.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks Goutham and Shi, this seems correct. Other jobs failures doesn\u0027t seems to related with the fix anymore. Thanks!","accounts_in_message":[],"_revision_number":6},{"id":"c9c904edc231e24e49d7b2db848f0f5d7b944f61","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-24 19:58:01.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks Goutham and Shi Yan, this looks very good.","accounts_in_message":[],"_revision_number":6},{"id":"d68822e5b5e3f860cd20451edbf032ee87a1688a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-24 20:20:43.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c045e23bb1b8da772ff25f3bd745e6d60032d8e5","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-24 20:21:08.000000000","message":"Patch Set 6:\n\nCool, double check the outputs on the gates, nothing seems related to this change. +WF now :)","accounts_in_message":[],"_revision_number":6},{"id":"0ae1b345ba63aa17c3b2ba002dd09049861461cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 20:22:07.000000000","message":"Patch Set 6: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"80eab2a2217c94e4f56551859544154d4e3e782b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-24 21:22:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/97/668597/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1f2d6cb/ : FAILURE in 2h 02m 45s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/97/668597/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/75de7e0/ : SUCCESS in 1h 30m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"af4cb966941e17e6b84d5bfcd1b148a6c366043d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 23:52:03.000000000","message":"Patch Set 6:\n\nrecheck\n\nThings are supposedly fine now","accounts_in_message":[],"_revision_number":6},{"id":"bec1bb533301289253bbcfa43023b2a82faec8f0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 23:52:26.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"6096c37ec7c8463bffc6f5687a1508aac8c47b62","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-24 23:53:04.000000000","message":"Patch Set 7:\n\nMade a teeny change to the commit message to get rid of that fake Merge Conflict. Sorry Douglas, Vicky - can I have your votes back? :)","accounts_in_message":[],"_revision_number":7},{"id":"e7f4460d4b9861ed601af42b18f69f1d22af27f8","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-25 00:36:24.000000000","message":"Patch Set 7: Code-Review+2\n\nTks, LGTM!","accounts_in_message":[],"_revision_number":7},{"id":"133c327775c1e76688c69dd0d751fbf413786a2b","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-25 00:50:04.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"6227fa9636fc7fb8ac0c5c1c03ec57161854f3af","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-25 01:47:03.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM. Thanks!","accounts_in_message":[],"_revision_number":7},{"id":"de36a5e2b39520760aa82db4187287f6110c2184","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-25 01:54:26.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3fb7d8e9a47f3238998b7a4dcb67132ccd57bc8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:58:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62be24e5cb04a99aca9c78a4e9b08ea : SUCCESS in 4m 51s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2c75bc6933794335994e7cda7ac40383 : SUCCESS in 1h 30m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6071d503d1374fb4aafbaf4ea69e5bdd : SUCCESS in 1h 09m 56s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/64e3dc8875bb49828fd571975c03f8a5 : FAILURE in 1h 06m 51s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/8f6e34800d594cd9b05f906cb546f0c9 : SUCCESS in 55m 01s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4742471b7c5e43e483fcdc9868153314 : FAILURE in 47m 24s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e1918a599f6406e9d82b338f440f915 : FAILURE in 1h 52m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ffb993178c8b4d09a03e0504d9a5335d : SUCCESS in 24m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7226a3b6a0994d3687ca18e8fcbf736f : SUCCESS in 1h 18m 58s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/179bc91f753a48439fae4f38d9aed293 : SUCCESS in 40m 51s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/6f8df0b9a68b4d1688c1628edda47f12 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d7736baa79f1738c24c56cf4a6d34e4fb5629dae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:59:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a05ccabbb161a818cf72e1d750ee5cdfc471a293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 02:59:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3d7dae7d1145799aa19e6cead642b8 : SUCCESS in 3m 02s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7d746748df1c4b7ab338b2e6cd4dd9f5 : SUCCESS in 44m 15s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/de0cb4b8182646d0aa4e0ad6dfe2848b : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":7},{"id":"7f7f0b184793de0679a205b148fdc383ed5ad5d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 02:59:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f8bc72effb9ff8251f5d7cfcba9122d86dd1c3b8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-25 03:59:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/97/668597/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/98dd9a1/ : FAILURE in 2h 00m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/97/668597/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5d5bea0/ : FAILURE in 1h 27m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"95df9da64aa8e22858f38ce0c9bed5046fad6884","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-25 10:28:56.000000000","message":"Patch Set 6:\n\nThanks!","accounts_in_message":[],"_revision_number":6}],"current_revision_number":7,"current_revision":"d3432669d2e782734017cc9308e9fd34768f5bb7","revisions":{"9ef2dde8d93d0a0727ce26d344ce0dceff17464a":{"kind":"REWORK","_number":1,"created":"2019-07-02 05:42:31.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/97/668597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/1"}}},"commit":{"parents":[{"commit":"67d4e7fb4d681b85b6bf8e380a977ff4507234c3","subject":"Merge \"Cast block count to integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/67d4e7fb4d681b85b6bf8e380a977ff4507234c3"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 05:40:20.000000000","tz":600},"subject":"Change to use neutron api to create floating ip","message":"Change to use neutron api to create floating ip\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ef2dde8d93d0a0727ce26d344ce0dceff17464a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ef2dde8d93d0a0727ce26d344ce0dceff17464a"}]},"branch":"refs/heads/master"},"dcdad0b551d86a8a387fc86f7d07540ea6d3fc04":{"kind":"REWORK","_number":2,"created":"2020-04-24 04:35:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/97/668597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/2"}}},"commit":{"parents":[{"commit":"32bd346f8b36ea19ff5f905d9be753a36f945004","subject":"Merge \"Remove experimental flag from share groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32bd346f8b36ea19ff5f905d9be753a36f945004"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-24 04:34:50.000000000","tz":-420},"subject":"Change to use neutron api to create floating ip","message":"Change to use neutron api to create floating ip\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dcdad0b551d86a8a387fc86f7d07540ea6d3fc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dcdad0b551d86a8a387fc86f7d07540ea6d3fc04"}]},"branch":"refs/heads/master"},"945fdb64aa72b24338d5fb33a3370bce3326c9af":{"kind":"REWORK","_number":3,"created":"2020-04-24 05:22:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/97/668597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/3"}}},"commit":{"parents":[{"commit":"32bd346f8b36ea19ff5f905d9be753a36f945004","subject":"Merge \"Remove experimental flag from share groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32bd346f8b36ea19ff5f905d9be753a36f945004"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-24 05:22:13.000000000","tz":-420},"subject":"Change to use neutron api to create floating ip","message":"Change to use neutron api to create floating ip\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/945fdb64aa72b24338d5fb33a3370bce3326c9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/945fdb64aa72b24338d5fb33a3370bce3326c9af"}]},"branch":"refs/heads/master"},"ff0d9589805ef9f4360752f608bee357db7549ca":{"kind":"REWORK","_number":4,"created":"2020-04-24 11:40:31.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/97/668597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/4"}}},"commit":{"parents":[{"commit":"32bd346f8b36ea19ff5f905d9be753a36f945004","subject":"Merge \"Remove experimental flag from share groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32bd346f8b36ea19ff5f905d9be753a36f945004"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Shi Yan","email":"yanshi.403@gmail.com","date":"2020-04-24 11:40:18.000000000","tz":0},"subject":"Change to use neutron api to create floating ip","message":"Change to use neutron api to create floating ip\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff0d9589805ef9f4360752f608bee357db7549ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff0d9589805ef9f4360752f608bee357db7549ca"}]},"branch":"refs/heads/master"},"55e2de85426d1df149691a4b3321c6920f7f17ea":{"kind":"REWORK","_number":5,"created":"2020-04-24 17:06:24.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/97/668597/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/5"}}},"commit":{"parents":[{"commit":"32bd346f8b36ea19ff5f905d9be753a36f945004","subject":"Merge \"Remove experimental flag from share groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32bd346f8b36ea19ff5f905d9be753a36f945004"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-24 17:06:12.000000000","tz":-420},"subject":"Fix floatingip allocation","message":"Fix floatingip allocation\n\n- Use neutron\u0027s floating IP provisioning APIs\n  instead of the compute APIs that will proxy\n  the request to neutron anyway.\n- Fix usage of the GET /ports API where an\n  incorrect filter parameter (\"fixed_ip\")\n  was being used (and subsequently ignored\n  by older versions of neutron, see bug #1749820 [1])\n- Refactor the check for ironic (we don\u0027t use\n  baremetal nodes in our testing) to match that\n  in tempest\u0027s own implementation of the port\n  resolution [2]\n- replace assertNotEqual(0, len(somelist))\n  with assertNotEmpty(somelist)\n\n[1] https://launchpad.net/bugs/1749820\n[2] https://opendev.org/openstack/tempest/src/commit/5c191faae78f52f876edba72484ea644d3df04fd/tempest/scenario/manager.py#L958-L994\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\nCo-Authored-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nCo-Authored-by: Shi Yan \u003cyanshi.403@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/55e2de85426d1df149691a4b3321c6920f7f17ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/55e2de85426d1df149691a4b3321c6920f7f17ea"}]},"branch":"refs/heads/master"},"0d919188555ae738373ed49b7fb2583be2f9bced":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-04-24 17:28:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/97/668597/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/6"}}},"commit":{"parents":[{"commit":"32bd346f8b36ea19ff5f905d9be753a36f945004","subject":"Merge \"Remove experimental flag from share groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32bd346f8b36ea19ff5f905d9be753a36f945004"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-24 17:24:20.000000000","tz":-420},"subject":"Fix floatingip allocation","message":"Fix floatingip allocation\n\n- Use neutron\u0027s floating IP provisioning APIs instead of the\n  compute APIs that will proxy the request to neutron anyway.\n  Another advantage of doing this is because we\u0027ll be using the\n  external network to create a floating IP upon, and don\u0027t need\n  to configure yet another \"floating_pool_name\" config opt to\n  get things to work. When used, in most environments, the\n  external network is the same as the floating pool.\n- Fix usage of the GET /ports API where an incorrect filter\n  parameter (\"fixed_ip\")  was being used (and subsequently ignored\n  by older versions of neutron, see bug #1749820 [1])\n- Refactor the check for ironic (we don\u0027t use baremetal nodes in\n  our testing) to match that in tempest\u0027s own implementation of the\n  port resolution [2]\n- replace assertNotEqual(0, len(somelist)) with\n  assertNotEmpty(somelist)\n\n[1] https://launchpad.net/bugs/1749820\n[2] https://opendev.org/openstack/tempest/src/commit/5c191faae78f52f876edba72484ea644d3df04fd/tempest/scenario/manager.py#L958-L994\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\nCo-Authored-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nCo-Authored-by: Shi Yan \u003cyanshi.403@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d919188555ae738373ed49b7fb2583be2f9bced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d919188555ae738373ed49b7fb2583be2f9bced"}]},"branch":"refs/heads/master"},"d3432669d2e782734017cc9308e9fd34768f5bb7":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-04-24 23:52:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/97/668597/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/97/668597/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/97/668597/7"}}},"commit":{"parents":[{"commit":"32bd346f8b36ea19ff5f905d9be753a36f945004","subject":"Merge \"Remove experimental flag from share groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/32bd346f8b36ea19ff5f905d9be753a36f945004"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2019-07-02 04:58:30.000000000","tz":600},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-24 23:52:24.000000000","tz":0},"subject":"Fix floatingip allocation","message":"Fix floatingip allocation\n\n- Use neutron\u0027s floating IP provisioning APIs instead of the\n  compute APIs that will proxy the request to neutron anyway.\n  Another advantage of doing this is because we\u0027ll be using the\n  external network to create a floating IP upon, and don\u0027t need\n  to configure yet another \"floating_pool_name\" config opt to\n  get things to work. When used, in most environments, the\n  external network is the same as the floating pool.\n- Fix usage of the GET /ports API where an incorrect filter\n  parameter (\"fixed_ip\")  was being used (and subsequently ignored\n  by older versions of neutron, see bug #1749820 [1])\n- Refactor the check for ironic (we don\u0027t use baremetal nodes in\n  our testing) to match that in tempest\u0027s own implementation of the\n  port resolution [2]\n- replace assertNotEqual(0, len(somelist)) with\n  assertNotEmpty(somelist)\n\n\n[1] https://launchpad.net/bugs/1749820\n[2] https://opendev.org/openstack/tempest/src/commit/5c191faae78f52f876edba72484ea644d3df04fd/tempest/scenario/manager.py#L958-L994\n\nChange-Id: I48b0f21fe75fae623e7e513cf2890f2c10a80317\nCo-Authored-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nCo-Authored-by: Shi Yan \u003cyanshi.403@gmail.com\u003e\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3432669d2e782734017cc9308e9fd34768f5bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3432669d2e782734017cc9308e9fd34768f5bb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
