)]}'
{"id":"openstack%2Fmanila-tempest-plugin~668622","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1490bf8f424181233aa3ea04a8410052b385699c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1833766","hashtags":[],"change_id":"I1490bf8f424181233aa3ea04a8410052b385699c","subject":"Look for api microversion header in \"Vary\" header data","status":"MERGED","created":"2019-07-02 07:29:48.000000000","updated":"2019-07-03 19:41:36.000000000","submitted":"2019-07-03 19:41:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668622-1562182896522-795c03e7","meta_rev_id":"4064bd09b0872621cb0ec47671677e4b5e54715c","_number":668622,"virtual_id_number":668622,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-03 19:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2019-07-03 18:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2019-07-03 13:34:26.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2019-07-03 18:29:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 10:48:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 13:34:26.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-07-03 18:29:19.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-07-03 19:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cacb8abfdce23ea9de1ee48d26dd0493e088b49e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-07-02 07:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3f4a08702a52b8368fb8c919d50a59e733714bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:21:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/668622/1/check/openstack-tox-pep8/fec3bf2/ : SUCCESS in 2m 51s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/22/668622/1/check/manila-tempest-dsvm-mysql-generic/6a78e09/ : FAILURE in 1h 19m 31s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/22/668622/1/check/manila-tempest-dsvm-postgres-container/2edce86/ : SUCCESS in 49m 29s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/22/668622/1/check/manila-tempest-dsvm-postgres-zfsonlinux/a1ca5ad/ : SUCCESS in 1h 11m 16s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/22/668622/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/4a99e03/ : FAILURE in 56m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/22/668622/1/check/manila-tempest-dsvm-generic-no-share-servers/2eaf904/ : FAILURE in 35m 08s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/22/668622/1/check/manila-tempest-dsvm-scenario/f1d79c1/ : FAILURE in 1h 46m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/22/668622/1/check/manila-tempest-minimal-dsvm-cephfs-native/bf971cf/ : SUCCESS in 33m 40s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/22/668622/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/84f982b/ : FAILURE in 1h 18m 55s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/22/668622/1/check/manila-tempest-minimal-dsvm-dummy/61fff34/ : SUCCESS in 36m 51s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/22/668622/1/check/manila-tempest-minimal-dsvm-lvm/f3a3566/ : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bd23e8f35a7520d392df851883180ddcc51cf27e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-02 10:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/22/668622/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8a42d25/ : SUCCESS in 1h 10m 41s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/22/668622/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/c300704/ : SUCCESS in 1h 26m 06s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6659ac9e64d3a78caba4be98aa6b0afb3726576f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-02 12:02:23.000000000","message":"Patch Set 1: Code-Review+1\n\nFix looks right.  Is there a test run here with MOD_WSGI that used to fail and now passes as reported in the bug?","accounts_in_message":[],"_revision_number":1},{"id":"b032442b92c5ce8334e64981d4bdfac4cc623044","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-02 17:55:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac8151b55fcd1d68d0fb97a6ae03ad21b15c2b1","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-07-03 04:27:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20a87bc7bbaa5cb463b2a182feb90fd4f2753de4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-07-03 05:22:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0390bd204f3f2cc72b1d101682eecc22054f296e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:06:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/668622/2/check/openstack-tox-pep8/8a959af/ : SUCCESS in 3m 21s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/22/668622/2/check/manila-tempest-dsvm-mysql-generic/1b201da/ : FAILURE in 1h 32m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/22/668622/2/check/manila-tempest-dsvm-postgres-container/95514a5/ : SUCCESS in 55m 09s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/22/668622/2/check/manila-tempest-dsvm-postgres-zfsonlinux/92c495a/ : SUCCESS in 1h 10m 47s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/22/668622/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/a3fb35c/ : FAILURE in 1h 01m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/22/668622/2/check/manila-tempest-dsvm-generic-no-share-servers/d9d4ad9/ : FAILURE in 1h 06m 49s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/22/668622/2/check/manila-tempest-dsvm-scenario/3a1109e/ : SUCCESS in 1h 33m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/22/668622/2/check/manila-tempest-minimal-dsvm-cephfs-native/25ed4b6/ : SUCCESS in 25m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/22/668622/2/check/manila-tempest-minimal-dsvm-cephfs-nfs/150691c/ : SUCCESS in 1h 26m 15s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/22/668622/2/check/manila-tempest-minimal-dsvm-dummy/1627d3f/ : SUCCESS in 41m 25s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/22/668622/2/check/manila-tempest-minimal-dsvm-lvm/aee74e9/ : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":2},{"id":"80b3a584737c1571e921153f5bfee1c478a266e2","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-03 13:34:26.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, Liron!","accounts_in_message":[],"_revision_number":2},{"id":"3d5bacdefea4d57857dc943334aa27eb3a4280f1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-03 18:29:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM, thank you for the fix.","accounts_in_message":[],"_revision_number":2},{"id":"988e4ebebef1c5b51565e122d2ab76ac6d9e5f41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:29:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3ee3e98e64a4b5efdcbb2abb5e45eb1d6e793b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:41:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/668622/2/gate/openstack-tox-pep8/8025b95/ : SUCCESS in 3m 04s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/22/668622/2/gate/manila-tempest-minimal-dsvm-dummy/0dedeb8/ : SUCCESS in 35m 10s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/22/668622/2/gate/manila-tempest-minimal-dsvm-lvm/c3f0f8d/ : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f53eccd53e3b47b489036bcc6da56cd16e1d4140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:41:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ac95412b1b0555775b4d45f51c27af2fa84724a","revisions":{"b9801e5e61a149d40f486d33738b9fe9480a231f":{"kind":"REWORK","_number":1,"created":"2019-07-02 07:29:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/22/668622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/22/668622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/1"}}},"commit":{"parents":[{"commit":"67d4e7fb4d681b85b6bf8e380a977ff4507234c3","subject":"Merge \"Cast block count to integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/67d4e7fb4d681b85b6bf8e380a977ff4507234c3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-07-02 07:11:10.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-07-02 07:11:10.000000000","tz":180},"subject":"Look for api microversion header in \"Vary\" header data","message":"Look for api microversion header in \"Vary\" header data\n\nApache configuration adds \"Accept-Encoding\" to the \"Vary\" header\nin its responses, this causes to failures in some tests which check\nthe \"Vary\" header content of the response.\nIn order to avoid this issue we should only check that api microversion\nheader axists in \"Vary\" header data.\n\nChange-Id: I1490bf8f424181233aa3ea04a8410052b385699c\nRelated-bug: #1833766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b9801e5e61a149d40f486d33738b9fe9480a231f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b9801e5e61a149d40f486d33738b9fe9480a231f"}]},"branch":"refs/heads/master"},"0ac95412b1b0555775b4d45f51c27af2fa84724a":{"kind":"REWORK","_number":2,"created":"2019-07-03 04:27:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/22/668622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/22/668622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/22/668622/2"}}},"commit":{"parents":[{"commit":"67d4e7fb4d681b85b6bf8e380a977ff4507234c3","subject":"Merge \"Cast block count to integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/67d4e7fb4d681b85b6bf8e380a977ff4507234c3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-07-02 07:11:10.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-07-03 04:27:08.000000000","tz":180},"subject":"Look for api microversion header in \"Vary\" header data","message":"Look for api microversion header in \"Vary\" header data\n\nApache configuration adds \"Accept-Encoding\" to the \"Vary\" header\nin its responses, this causes to failures in some tests which check\nthe \"Vary\" header content of the response.\nIn order to avoid this issue we should only check that api microversion\nheader axists in \"Vary\" header data.\n\nChange-Id: I1490bf8f424181233aa3ea04a8410052b385699c\nRelated-bug: #1833766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ac95412b1b0555775b4d45f51c27af2fa84724a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ac95412b1b0555775b4d45f51c27af2fa84724a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
