)]}'
{"id":"openstack%2Fmanila-tempest-plugin~673365","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I960b6bf8f283ea928fd512811f8d88268b0945d2","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I960b6bf8f283ea928fd512811f8d88268b0945d2","subject":"Comment test tags","status":"MERGED","created":"2019-07-29 21:09:51.000000000","updated":"2019-07-31 03:15:49.000000000","submitted":"2019-07-31 03:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673365-1564542949420-bedbf1ba","meta_rev_id":"4fe05605de206b5a16297a1343191424d36b3c31","_number":673365,"virtual_id_number":673365,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2019-07-31 02:35:40.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-31 03:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2019-07-30 02:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2019-07-30 02:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2019-07-30 09:51:38.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 02:36:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-07-31 02:35:40.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 03:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53a7ca541b573277778b8a86af1c875dd4dff5ba","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-29 21:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb6d41bb2cf254d2b78d3d4bf2ba3218f453335c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-29 21:11:26.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s merge this one [1] first.\n\n[1] https://review.opendev.org/#/c/673290","accounts_in_message":[],"_revision_number":1},{"id":"b454c7837f6b09637f30f28a618a77a45b38b990","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:58:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/673365/1/check/openstack-tox-pep8/f9c1275/ : SUCCESS in 4m 09s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-mysql-generic/c662da3/ : SUCCESS in 1h 29m 23s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-container/2b3bb0c/ : SUCCESS in 54m 10s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-zfsonlinux/ac9ceaf/ : SUCCESS in 1h 15m 13s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/63b79aa/ : FAILURE in 59m 40s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-generic-no-share-servers/6b1d725/ : FAILURE in 1h 06m 36s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-scenario/4b8f545/ : SUCCESS in 1h 40m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-cephfs-native/0c7a1f4/ : SUCCESS in 32m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/64db796/ : SUCCESS in 1h 31m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-dummy/725d60a/ : SUCCESS in 39m 23s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-lvm/d28d6f0/ : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1511587c6dfc02ba1dfe4c12cd14243b9290b005","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-30 00:55:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.161.150/logs/65/673365/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b8fb433/ : SUCCESS in 1h 11m 24s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.161.150/logs/65/673365/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/39d172a/ : SUCCESS in 1h 28m 16s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9227ce169ef964446c240132dc12501cfee7b861","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-30 02:36:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, this can merge when you remove W-1","accounts_in_message":[],"_revision_number":1},{"id":"6d2c5a6b9e0694f87dbc4f057dde08e8a4f60f0b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-30 09:51:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca755787112ece0e6f46296a23b955e6486733fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:51:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec193e76a7d01ed07c4c34f2057394600d5ea9e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 10:55:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/65/673365/1/gate/openstack-tox-pep8/ee7712c/ : SUCCESS in 3m 46s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/65/673365/1/gate/manila-tempest-minimal-dsvm-dummy/e436005/ : FAILURE in 40m 53s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/65/673365/1/gate/manila-tempest-minimal-dsvm-lvm/ac7ab09/ : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e7f20ea5252dc3d56e1be9eabdb3af8cd3f12620","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-30 18:48:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"48994372831cbb8f673b343d5f633bba9f3c7124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:29:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/673365/1/check/openstack-tox-pep8/93f28aa/ : SUCCESS in 3m 42s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-mysql-generic/94edab0/ : SUCCESS in 1h 29m 39s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-container/4a62afe/ : SUCCESS in 53m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-zfsonlinux/1b94e2b/ : SUCCESS in 1h 11m 34s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/e5cc8a3/ : SUCCESS in 1h 00m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-generic-no-share-servers/86a1cfa/ : SUCCESS in 59m 27s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-scenario/16aafa8/ : SUCCESS in 1h 29m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-cephfs-native/7e1099d/ : SUCCESS in 27m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/296164e/ : SUCCESS in 1h 25m 51s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-dummy/c2c3128/ : SUCCESS in 38m 57s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-lvm/d89e50d/ : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":1},{"id":"80320bd316dd2c9c4e09c1867d648e0bc1213cab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:29:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"745a567f1d5211b9b3da59475ec781c702c75f84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:44:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/65/673365/1/gate/openstack-tox-pep8/e1c0e94/ : SUCCESS in 4m 46s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/65/673365/1/gate/manila-tempest-minimal-dsvm-dummy/c2a73ee/ : SUCCESS in 54m 55s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/65/673365/1/gate/manila-tempest-minimal-dsvm-lvm/159eea9/ : FAILURE in 55m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6382ae3aec24029508832b01b277d65fa746a623","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-30 21:56:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.161.150/logs/65/673365/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/551cb10/ : SUCCESS in 1h 10m 57s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.161.150/logs/65/673365/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1251d5c/ : SUCCESS in 1h 26m 40s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b53f737277e069858032adf4d91f00795c315131","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-31 00:30:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4dfc38bdb560427cdd7b49cd610afb77b1e3860","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 02:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/673365/1/check/openstack-tox-pep8/957b6bf/ : SUCCESS in 5m 38s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-mysql-generic/752541d/ : SUCCESS in 1h 34m 10s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-container/54d3c60/ : SUCCESS in 52m 03s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-zfsonlinux/d2e8990/ : SUCCESS in 1h 17m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/38a4ce3/ : SUCCESS in 50m 14s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-generic-no-share-servers/7dfa0bc/ : SUCCESS in 1h 06m 01s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/65/673365/1/check/manila-tempest-dsvm-scenario/2c60893/ : SUCCESS in 1h 25m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-cephfs-native/4ee0d27/ : SUCCESS in 25m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/5318a6b/ : SUCCESS in 1h 12m 34s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-dummy/7573ef0/ : SUCCESS in 42m 57s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/65/673365/1/check/manila-tempest-minimal-dsvm-lvm/a8a4159/ : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8424921fa639e9a9407aefff26a43f63725152c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 02:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f44f887aea32cc9d136fda71c12a465e015a6799","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-31 02:35:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.161.150/logs/65/673365/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/24791d0/ : POST_FAILURE in 17m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.161.150/logs/65/673365/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b35bf51/ : POST_FAILURE in 17m 13s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ed047c50af44b7513c33e11ddee82535aa52c542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 03:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4fe05605de206b5a16297a1343191424d36b3c31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 03:15:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/673365/1/gate/openstack-tox-pep8/0d5d8e7/ : SUCCESS in 2m 54s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/65/673365/1/gate/manila-tempest-minimal-dsvm-dummy/dc54248/ : SUCCESS in 41m 40s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/65/673365/1/gate/manila-tempest-minimal-dsvm-lvm/11bbdf2/ : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69f9696b0a2450b2664298be7b4af06159e68916","revisions":{"69f9696b0a2450b2664298be7b4af06159e68916":{"kind":"REWORK","_number":1,"created":"2019-07-29 21:09:51.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/65/673365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/673365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/673365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/673365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/673365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/673365/1"}}},"commit":{"parents":[{"commit":"1a96ed9f703de2080ccdf2a083aa9b30eab4d04f","subject":"Increase export IP ping count","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a96ed9f703de2080ccdf2a083aa9b30eab4d04f"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-29 21:07:03.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-29 21:07:13.000000000","tz":-240},"subject":"Comment test tags","message":"Comment test tags\n\nAdd comments to the definitions of the test tags related\nto service involvement matching the documentation on\nrunning a subset of manila tempest tests based on\nservice involvement [1].\n\n[1] https://docs.openstack.org/manila/latest/contributor/tempest_tests.html#running-a-subset-of-tests-based-on-service-involvement\nChange-Id: I960b6bf8f283ea928fd512811f8d88268b0945d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69f9696b0a2450b2664298be7b4af06159e68916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69f9696b0a2450b2664298be7b4af06159e68916"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
