)]}'
{"id":"openstack%2Fmanila-tempest-plugin~674871","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7033d726c1119c20be492d489f58d257ab9e05e4","project":"openstack/manila-tempest-plugin","branch":"master","topic":"share_quota","hashtags":[],"change_id":"I7033d726c1119c20be492d489f58d257ab9e05e4","subject":"Deleted duplicate line 366","status":"ABANDONED","created":"2019-08-06 16:46:23.000000000","updated":"2019-08-20 19:39:03.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d770823b90b97cf22a00ee9565ba4e136dca905","_number":674871,"virtual_id_number":674871,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2019-08-19 13:17:19.000000000","_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2019-08-06 18:27:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2019-08-20 00:12:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-07 03:53:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 18:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-07 03:53:14.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2019-08-20 00:12:10.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"dce2a34ee95c0f82bc19e74fce2ef95b65c40516","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-06 16:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2431b8e5b03aebf06ab9df1f5172ca43b27a05b8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-06 16:59:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThanks for your contribution. Please review the comment inline..","accounts_in_message":[],"_revision_number":1},{"id":"d111b4e948e1556ee790d87bb7813e138eed490e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/71/674871/1/check/openstack-tox-pep8/c9c8756/ : SUCCESS in 4m 29s\n- manila-tempest-dsvm-mysql-generic https://logs.opendev.org/71/674871/1/check/manila-tempest-dsvm-mysql-generic/2eb435b/ : SUCCESS in 1h 29m 37s (non-voting)\n- manila-tempest-dsvm-postgres-container https://logs.opendev.org/71/674871/1/check/manila-tempest-dsvm-postgres-container/ebaab86/ : SUCCESS in 52m 53s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://logs.opendev.org/71/674871/1/check/manila-tempest-dsvm-postgres-zfsonlinux/f97c404/ : SUCCESS in 1h 10m 12s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://logs.opendev.org/71/674871/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/46d4d8b/ : SUCCESS in 1h 02m 22s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://logs.opendev.org/71/674871/1/check/manila-tempest-dsvm-generic-no-share-servers/84c32c9/ : FAILURE in 1h 09m 54s (non-voting)\n- manila-tempest-dsvm-scenario https://logs.opendev.org/71/674871/1/check/manila-tempest-dsvm-scenario/558af8f/ : SUCCESS in 1h 31m 08s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://logs.opendev.org/71/674871/1/check/manila-tempest-minimal-dsvm-cephfs-native/428d348/ : SUCCESS in 28m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://logs.opendev.org/71/674871/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/d15cb29/ : SUCCESS in 1h 21m 06s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://logs.opendev.org/71/674871/1/check/manila-tempest-minimal-dsvm-dummy/eaf2ed9/ : SUCCESS in 37m 08s\n- manila-tempest-minimal-dsvm-lvm https://logs.opendev.org/71/674871/1/check/manila-tempest-minimal-dsvm-lvm/4cf0528/ : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f522b7f0e1adc2fe846ea1aeede41a160ee571b3","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-08-07 03:51:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d25aa4027c7dfe2f18bfc59a4b3f18d01a3e3927","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-08-07 03:53:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a0b42c5bfb38f8594f93c70034fd85874075088","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-19 13:17:19.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nMade suggested changes.","accounts_in_message":[],"_revision_number":1},{"id":"878fe04627dd6c076286009da16e94caf8eebe16","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-20 00:12:10.000000000","message":"Patch Set 1: Code-Review-1\n\nHi Vida, \n\nWith Gerrit, you can amend your git changes rather than create new commits because Gerrit maintains older commits from a change, always. So, please squash this commit, along with the commit in https://review.opendev.org/#/c/676406/1; and resubmit..","accounts_in_message":[],"_revision_number":1},{"id":"65313b28a213dde1fd5452094c66ff4746b8869e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-20 00:15:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"941354ecf751ae66aa78a458073c006a99b2c4df","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-20 00:16:45.000000000","message":"Patch Set 1:\n\nSee these instructions regarding updating changes: https://docs.openstack.org/infra/manual/developers.html#updating-a-change","accounts_in_message":[],"_revision_number":1},{"id":"5378cb1c85771487ad5623148c937c06ac52db75","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-20 19:39:03.000000000","message":"Abandoned\n\nAbandoned due to newer change","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea01a7400ba144cdbd6c88bea10e54c40885d268","revisions":{"ea01a7400ba144cdbd6c88bea10e54c40885d268":{"kind":"REWORK","_number":1,"created":"2019-08-06 16:46:23.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/71/674871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/71/674871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/674871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/674871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/674871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/674871/1"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-06 16:42:47.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-06 16:42:47.000000000","tz":-240},"subject":"Deleted duplicate line 366","message":"Deleted duplicate line 366\n\nChange-Id: I7033d726c1119c20be492d489f58d257ab9e05e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea01a7400ba144cdbd6c88bea10e54c40885d268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea01a7400ba144cdbd6c88bea10e54c40885d268"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
