)]}'
{"id":"openstack%2Fmanila-tempest-plugin~675206","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie58bc904a50da9a5541b8c956953526c45228fb3","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ie58bc904a50da9a5541b8c956953526c45228fb3","subject":"Validate and cleanup capability config","status":"ABANDONED","created":"2019-08-07 21:31:55.000000000","updated":"2020-03-24 22:37:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e20ab10f1f6c0b4cc55ae15754069f2a363c6e88","_number":675206,"virtual_id_number":675206,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2019-08-10 01:17:53.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 22:36:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"all":[{"value":1,"date":"2019-08-10 03:00:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},{"value":-1,"date":"2019-08-09 20:31:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 22:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 01:17:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 03:00:25.000000000","updated_by":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"reviewer":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"state":"REVIEWER"}],"messages":[{"id":"44bf57eaaf2a2b1a0056843217a8163ecb8003d7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-07 21:31:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a90d4b709b2d21f88e6dab734395b7ff9c5104d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 23:09:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/06/675206/1/check/openstack-tox-pep8/48b2633/ : SUCCESS in 3m 50s\n- manila-tempest-dsvm-mysql-generic https://logs.opendev.org/06/675206/1/check/manila-tempest-dsvm-mysql-generic/e468408/ : FAILURE in 1h 27m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://logs.opendev.org/06/675206/1/check/manila-tempest-dsvm-postgres-container/cb464fd/ : SUCCESS in 54m 47s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://logs.opendev.org/06/675206/1/check/manila-tempest-dsvm-postgres-zfsonlinux/c986e26/ : SUCCESS in 1h 10m 04s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://logs.opendev.org/06/675206/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/dd911e3/ : SUCCESS in 57m 24s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://logs.opendev.org/06/675206/1/check/manila-tempest-dsvm-generic-no-share-servers/3b58f57/ : FAILURE in 1h 04m 27s (non-voting)\n- manila-tempest-dsvm-scenario https://logs.opendev.org/06/675206/1/check/manila-tempest-dsvm-scenario/39c67ef/ : SUCCESS in 1h 31m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://logs.opendev.org/06/675206/1/check/manila-tempest-minimal-dsvm-cephfs-native/bbf007e/ : SUCCESS in 28m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://logs.opendev.org/06/675206/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/77d3062/ : SUCCESS in 1h 20m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://logs.opendev.org/06/675206/1/check/manila-tempest-minimal-dsvm-dummy/b41e12d/ : SUCCESS in 43m 49s\n- manila-tempest-minimal-dsvm-lvm https://logs.opendev.org/06/675206/1/check/manila-tempest-minimal-dsvm-lvm/92161c2/ : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":1},{"id":"81056bad7d0531d6a3f0de642a352973a96051dd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-08 02:35:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.161.150/logs/06/675206/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/436ab60/ : SUCCESS in 1h 22m 46s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.161.150/logs/06/675206/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9aca283/ : FAILURE in 3h 25m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c10c5056a477fedbc02c607b3f379efab0918e83","author":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"date":"2019-08-08 04:20:45.000000000","message":"Patch Set 1:\n\nHi, code looks like this:\n\ncap_snap: True\nrun_snap_test: False \n-\u003e error\n\ncap_share_from_snap: True\nrun_snap_test: False\n-\u003e error\n\nBut shouldn\u0027t we also think about this case too?\ncap_snap: False\nrun_snap_test: True\n\nIf backend has no snapshot support, running snapshot test will always failed.","accounts_in_message":[],"_revision_number":1},{"id":"292de144c4c406b57da1bafcadfa0309e0a5d868","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-09 19:24:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32caf2384bf640660ead716f62fb2a09e51c09d5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-09 19:26:00.000000000","message":"Patch Set 1:\n\n\u003e If backend has no snapshot support, running snapshot test will always failed.\n\nYes, that condition was being handled, but you\u0027re right - there\u0027s no value in supplying different values for capability_snapshot_support and run_snapshot_tests. I\u0027d be happy if we got rid of all the capability flags... but that seems like a larger effort.","accounts_in_message":[],"_revision_number":1},{"id":"01b54e118c78dd0595cd9618197c6581f8ddb7a9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-09 20:31:45.000000000","message":"Patch Set 2: Workflow-1\n\nHmmm, this doesn\u0027t make tests fail - tempest swallows any exceptions and logs it in the log file: https://github.com/openstack/tempest/blob/40fcb9ba283d36bad341d57356ed4276199d2ec4/tempest/test_discover/plugins.py#L202-L206","accounts_in_message":[],"_revision_number":2},{"id":"873c3b12e2217850e8d5c7a4890cae73d52ee4dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 22:36:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/06/675206/2/check/openstack-tox-pep8/e7096bc/ : SUCCESS in 5m 51s\n- manila-tempest-dsvm-mysql-generic https://logs.opendev.org/06/675206/2/check/manila-tempest-dsvm-mysql-generic/a9cbcf0/ : FAILURE in 1h 42m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://logs.opendev.org/06/675206/2/check/manila-tempest-dsvm-postgres-container/20e3ca0/ : SUCCESS in 53m 31s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://logs.opendev.org/06/675206/2/check/manila-tempest-dsvm-postgres-zfsonlinux/05b6135/ : SUCCESS in 1h 13m 58s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://logs.opendev.org/06/675206/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/5ca4124/ : FAILURE in 1h 03m 06s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://logs.opendev.org/06/675206/2/check/manila-tempest-dsvm-generic-no-share-servers/d4bbca7/ : SUCCESS in 1h 01m 22s (non-voting)\n- manila-tempest-dsvm-scenario https://logs.opendev.org/06/675206/2/check/manila-tempest-dsvm-scenario/8ef0558/ : SUCCESS in 1h 38m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://logs.opendev.org/06/675206/2/check/manila-tempest-minimal-dsvm-cephfs-native/31f243d/ : SUCCESS in 22m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://logs.opendev.org/06/675206/2/check/manila-tempest-minimal-dsvm-cephfs-nfs/d44c340/ : SUCCESS in 1h 25m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://logs.opendev.org/06/675206/2/check/manila-tempest-minimal-dsvm-dummy/fd77d9a/ : SUCCESS in 42m 04s\n- manila-tempest-minimal-dsvm-lvm https://logs.opendev.org/06/675206/2/check/manila-tempest-minimal-dsvm-lvm/ae53d45/ : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3a18cb2b1d1561873e6b53c09a1270f6f41b7e91","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-10 01:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.161.150/logs/06/675206/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/05e33fa/ : SUCCESS in 1h 18m 59s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.161.150/logs/06/675206/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/40addff/ : SUCCESS in 1h 33m 34s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"66fc7a874d0a30efe03712cceb63220e761da555","author":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"date":"2019-08-10 03:00:25.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e Hmmm, this doesn\u0027t make tests fail - tempest swallows any\n \u003e exceptions and logs it in the log file: https://github.com/openstack/tempest/blob/40fcb9ba283d36bad341d57356ed4276199d2ec4/tempest/test_discover/plugins.py#L202-L206\n\noh, i see. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"e20ab10f1f6c0b4cc55ae15754069f2a363c6e88","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 22:37:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2287bbc56b3a8aba8669dd1dc69dcd63278259a","revisions":{"73d6135534c2a3f53ad81a4fa27be0c328cc2fe4":{"kind":"REWORK","_number":1,"created":"2019-08-07 21:31:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/675206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/675206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/1"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-08-07 21:31:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-08-07 21:31:47.000000000","tz":-420},"subject":"Validate and cleanup capability config","message":"Validate and cleanup capability config\n\n- `capability_revert_to_snapshot_support` was deprecated\n  and unused.\n- Fail fast and early if using a mismatch of capability\n  configuration options.\n\nChange-Id: Ie58bc904a50da9a5541b8c956953526c45228fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73d6135534c2a3f53ad81a4fa27be0c328cc2fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73d6135534c2a3f53ad81a4fa27be0c328cc2fe4"}]},"branch":"refs/heads/master"},"d2287bbc56b3a8aba8669dd1dc69dcd63278259a":{"kind":"REWORK","_number":2,"created":"2019-08-09 19:24:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/675206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/675206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/675206/2"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-08-07 21:31:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-08-09 19:23:40.000000000","tz":-420},"subject":"Validate and cleanup capability config","message":"Validate and cleanup capability config\n\n- `capability_revert_to_snapshot_support` was deprecated\n  and unused.\n- Fail fast and early if using a mismatch of capability\n  configuration options.\n\nChange-Id: Ie58bc904a50da9a5541b8c956953526c45228fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2287bbc56b3a8aba8669dd1dc69dcd63278259a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2287bbc56b3a8aba8669dd1dc69dcd63278259a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
