)]}'
{"id":"openstack%2Fmanila-tempest-plugin~676406","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ifc8f949e3341618ccfa24fc21fc6bad74280f7d3","project":"openstack/manila-tempest-plugin","branch":"master","topic":"share_quota","hashtags":[],"change_id":"Ifc8f949e3341618ccfa24fc21fc6bad74280f7d3","subject":"Added code to verify the quota was updated in line 360.","status":"ABANDONED","created":"2019-08-14 12:53:51.000000000","updated":"2019-08-20 19:37:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6d66060d93d00621c24f974ec017c47a61e9c5c","_number":676406,"virtual_id_number":676406,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-20 00:10:06.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2019-08-14 16:04:33.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 14:33:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 14:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 16:04:33.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 00:10:06.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"618e11e3d393e62ffc1c0a0a0c0b085354d96a1f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-14 12:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3c50dcb799771209f40ccdc071e953987ae36d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:33:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0079ad37b546b29cd165eaf144b6b0 : SUCCESS in 3m 25s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5ad72eea609f41dd8fba6723b6aa2d3e : FAILURE in 1h 30m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1e5a21c4ca1d44e48f4aea63e41dc10b : SUCCESS in 48m 16s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/5aaa8d10e29142f3aa9c43f1073784fb : SUCCESS in 1h 12m 02s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e472762c3ff840dab32759702b109d87 : FAILURE in 55m 58s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/909be0c38cdf4bda892b612eb4bd0231 : FAILURE in 1h 14m 40s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b65ea2b629cd4f4383ce452b153a9a35 : FAILURE in 1h 33m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7da760b1ebc7412d88b0dee7fb18389e : SUCCESS in 31m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/267f0512918047f2b28ab644abd96df8 : SUCCESS in 1h 24m 26s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c45144fae54240298f432a70c6c88e89 : SUCCESS in 43m 41s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/72709c89570a493983bea6351aa3b6b7 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9aa3313929f61d86702fce8d34a20eb9f7b82cb2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-14 16:04:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/06/676406/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c56b421/ : SUCCESS in 1h 30m 55s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/06/676406/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6b14f0d/ : FAILURE in 1h 29m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"624e0fe7f0756858f62beadff3d200a9c0e089c0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-20 00:09:18.000000000","message":"Patch Set 1:\n\nPlease see comments on the parent patch: https://review.opendev.org/#/c/674871/1","accounts_in_message":[],"_revision_number":1},{"id":"4a082a908a0f36aeb314ec91edc6234a852f89fc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-20 00:10:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"946fc58a9f260d453b53d1fd6694b18cf03b6a23","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-20 19:37:51.000000000","message":"Abandoned\n\nAbandoned due to another change","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffe20f9139b68706cc0d39ce74342724a7dae487","revisions":{"ffe20f9139b68706cc0d39ce74342724a7dae487":{"kind":"REWORK","_number":1,"created":"2019-08-14 12:53:51.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/06/676406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/676406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/676406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/676406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/676406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/676406/1"}}},"commit":{"parents":[{"commit":"ea01a7400ba144cdbd6c88bea10e54c40885d268","subject":"Deleted duplicate line 366","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea01a7400ba144cdbd6c88bea10e54c40885d268"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-14 12:51:02.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-14 12:51:02.000000000","tz":-240},"subject":"Added code to verify the quota was updated in line 360.","message":"Added code to verify the quota was updated in line 360.\n\nChange-Id: Ifc8f949e3341618ccfa24fc21fc6bad74280f7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ffe20f9139b68706cc0d39ce74342724a7dae487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ffe20f9139b68706cc0d39ce74342724a7dae487"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
