)]}'
{"id":"openstack%2Fmanila-tempest-plugin~676475","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I60bdb493ad5766f28408a0c877f960922fe44ad0","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-network-multiple-subnets","hashtags":[],"change_id":"I60bdb493ad5766f28408a0c877f960922fe44ad0","subject":"Adds new functional tests for share-network-subnets","status":"MERGED","created":"2019-08-14 17:51:59.000000000","updated":"2019-09-21 01:02:58.000000000","submitted":"2019-09-21 01:02:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":76,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676475-1569027778364-953e65f2","meta_rev_id":"6a0de70da2e4d7a939e273708d371bf173ec845c","_number":676475,"virtual_id_number":676475,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2019-09-20 19:42:36.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"date":"2019-09-20 21:24:20.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-21 01:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2019-09-20 19:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2019-09-20 23:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 22:04:46.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2019-09-20 19:42:36.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2019-09-20 21:24:20.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-20 23:55:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-09-21 01:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"120e488aa0fc24bcb26266de1f3d1bb02420f040","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-08-14 17:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59c20d738c69fbe75844b48bc4bc42e0efab7507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:29:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ab6d129afe48368564f90fc13c3dd8 : SUCCESS in 3m 10s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/cdfb96e0337a474c9490872b27f3aa9b : FAILURE in 1h 22m 00s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1b03d285b27741c28c05b60c5d567ee2 : SUCCESS in 48m 16s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/6730b86857db4715b544d25a519cb87e : SUCCESS in 1h 02m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/fabed5d76bd94d178438300f39de0690 : FAILURE in 59m 50s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ea2a4b0390df4bfc9dbdfdfbfe013a32 : SUCCESS in 59m 26s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c1601f3803e4cd1bc969ec0aa5549fd : SUCCESS in 1h 33m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c692cc4d3b9649d7bffc49b789c207c0 : SUCCESS in 22m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/938caf36845d465bbb131e7ac417baf5 : SUCCESS in 1h 24m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3e68a1d2acb746e8b2d16ac0d63ab025 : SUCCESS in 41m 09s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9250f12a60934a4d991f0f15f8fe3c8c : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":1},{"id":"42d26ddb0a35cc2b7ae212eeafa37ba1534b7c84","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-14 21:09:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/74243fc/ : FAILURE in 6m 11s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/147c3af/ : SUCCESS in 1h 38m 13s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8b5350be2d53baef817f584125aa66b4f369fa6a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-08-14 21:32:19.000000000","message":"Patch Set 1:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":1},{"id":"606859024376132026fefb2223948eaa10339aea","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-14 23:31:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ae674fc/ : SUCCESS in 1h 40m 37s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/c05b147/ : SUCCESS in 1h 52m 04s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ed13403c7c18a564a30df6d4c860f15eae80fa0e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-08-20 12:02:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee18f4b9b6bb10fecf32ba23be83f10853071d4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:05:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba024aa3beed4a6f9038e56d04417be5 : SUCCESS in 4m 14s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/27cb3b6944044ba4ab5e9f29fb43ecfc : SUCCESS in 1h 19m 40s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d149dbfe1e0c43aeab792a9de35d3132 : FAILURE in 23m 03s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/56cc07c552b5417aa1f55bdb4156d06d : FAILURE in 28m 11s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b9939e42d07645928919a5c4f7e6b840 : FAILURE in 27m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4d01001b5efa4024857e26d8406c43a2 : SUCCESS in 55m 54s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be4e3add81e241ba8920bc3282007c8a : SUCCESS in 1h 35m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/4e9be67888894231b74b0a08542f7df2 : SUCCESS in 26m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/eb49c3562339431e9060e05cf58843a7 : SUCCESS in 1h 25m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1a6a6cbb32074c80a6db78f90d6e7ed5 : SUCCESS in 39m 16s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/3252954f6c9c46f593f1408c44d75fb1 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6fe916f1a85caf40267d0c43528aee50c378b00e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-20 15:58:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/db410ea/ : SUCCESS in 1h 48m 31s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/aff0eba/ : SUCCESS in 1h 28m 45s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc7abdc198bd1cce219084cc13281705f637312a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-08-27 20:22:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nHi Carlos, the patch is looking good, I\u0027m just proposing adding 3 more negative tests. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"ae29b8f93ce49045f5c1152a64ee0b11434e586f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-02 18:31:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c1edc9c9085703dd6f1b77268576e68622f15f6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-02 18:32:28.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nHey, Douglas! Thanks for reviewing this patch.\nI\u0027ve changed some things and uploaded a new patch version.","accounts_in_message":[],"_revision_number":3},{"id":"3eb2ea10fa6bed9909192b51946b09d6c34a74f5","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-02 18:52:10.000000000","message":"Patch Set 3: Code-Review+1\n\nHi Carlos, thanks for the update. LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"fa949d725191034c2befd8a89e45ab178cd48a1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 20:08:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d578ff5b180746e9bdad9c49de49fb76 : SUCCESS in 4m 07s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c19bd3540d4a461c89636ed8ef65eae5 : SUCCESS in 1h 27m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/491fdd8d0c43429e8d6e77ce49973edc : SUCCESS in 52m 49s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2af87142a4ee4518ad1e97ac1edbef9a : SUCCESS in 1h 10m 15s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e45a2a880bd44f8a9a33d48a4fb3ca06 : SUCCESS in 55m 36s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/babfe2133ab84bcab7d197c3cb3dd506 : SUCCESS in 1h 11m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9738f50edc0f48339e2d3943f9faa6e1 : SUCCESS in 1h 32m 54s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/39e4990d747148d0ba0db63ff27ede8c : SUCCESS in 25m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/074ea16387e54daaa246d066522f8087 : SUCCESS in 1h 22m 02s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f0bd9a7b56734b95971d035ed88a10ce : SUCCESS in 37m 16s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/621fb4d087ab44dc8c575c8f9d00cec3 : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d65d2cde79ee59777f81e68506186d74c6857d62","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-02 21:41:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/7611993/ : SUCCESS in 1h 22m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1396c02/ : SUCCESS in 1h 28m 02s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0050837babc9ba5da6696662b5ec6bc49d193215","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-03 16:42:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nCarlos, found out a scenario that may be a problem for some tests.","accounts_in_message":[],"_revision_number":3},{"id":"4121fb8f0dbde158d840c3b2523e32c066eb10c6","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2019-09-03 21:28:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85268e6722201b2c60b8b2f2282434c8b643616e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 23:08:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24710f8e6334620b54b829b00291807 : SUCCESS in 4m 04s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e10975791efa436fbf45d1d902a16283 : FAILURE in 1h 32m 20s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/bc4b1424d00e4edd8a841b7c97ad3ee3 : SUCCESS in 47m 54s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d55b76a0cf944f82b828d8990c977a22 : SUCCESS in 1h 17m 38s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/8e285cba47734654a1d3f88cd87979e9 : FAILURE in 1h 04m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/13b9a999a1964bce89bf49f5d95890ac : SUCCESS in 54m 54s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/363211857405488e92827456c44a4dd1 : SUCCESS in 1h 30m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2271c5da1aa74b60b68dab3891e84fd7 : SUCCESS in 31m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/698367fec0934572ac04287ed5764bdf : SUCCESS in 1h 31m 48s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/99e00c0d4a9d47acb8d4a380fcec69a3 : FAILURE in 45m 00s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/06677cdc072a4fbcba8856d9f56d4955 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":4},{"id":"c786625273543d7c2901062052a7cb1d098ac591","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-03 23:41:12.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"af81d054a4d7842cd061358e1b2d8bf1ef2c73ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:43:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b23ef9233f48e2974483e697aa79a7 : SUCCESS in 3m 38s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c64ec5420f054f8ea02a38b719a865b2 : FAILURE in 1h 29m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3897e78df8a8479eb4cc0110c8b8d5db : SUCCESS in 1h 17m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/62eeb8595b274643bc6fa0c4ee57ba93 : SUCCESS in 1h 17m 11s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/94296cc2dfe74c298daeaddad09ffab4 : SUCCESS in 1h 05m 37s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/fb4b52e1a68e44e8ad33a37042803cf8 : FAILURE in 1h 04m 02s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28a0803643e846fd98aa95676a097a0a : SUCCESS in 1h 18m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/98783b8ac6fd4fd6b796871beacd57f4 : SUCCESS in 27m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bb31a2ab63b64681a239ed7042bb83f1 : SUCCESS in 1h 28m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9fa62004aff14646b80949fa885960f0 : SUCCESS in 39m 35s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/2afd3940480c46569c2a29bb429ec224 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":4},{"id":"aace9356caa17b7086bac16512007862542b81fa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-04 07:32:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5b2a4da/ : SUCCESS in 1h 19m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e06cc20/ : SUCCESS in 1h 31m 24s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3517400d4f2ae783135bdac86d65554495717157","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2019-09-06 22:04:46.000000000","message":"Patch Set 4:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":4},{"id":"10ce490970012e22e8ea35be48fd49789712319c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-06 23:39:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3465039/ : SUCCESS in 1h 27m 25s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ed9fc07/ : SUCCESS in 1h 26m 40s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a9ea36287f167ebdb3009135fc09a12f16b0c21c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-10 23:26:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d575e06af954d688d8fdcdc2b547a87f08d21c27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 01:15:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33b2385f329489d92f0875c4cc694db : SUCCESS in 3m 19s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1b0f7ecbbc1e4f058b897d9b4a782ffa : FAILURE in 1h 37m 22s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/42cff31d3534409ea8da857bdb733ddf : FAILURE in 45m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8535a88e215945fc8ae3810c5a2f5c5d : FAILURE in 1h 11m 33s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/4980e216c9ee46758b0f9cfa7f3769ab : SUCCESS in 1h 02m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3323413c3f694b6ca4c45fed92eef514 : SUCCESS in 1h 06m 24s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef97a8bfc3074fed9ff498a2f453a24a : FAILURE in 1h 28m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/40faa11dcb59475fac7b2a3dee9e2213 : FAILURE in 32m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/93cd863e2c414862a3799971756aefb5 : FAILURE in 1h 37m 05s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a49e1460a50f437dade35fc5f764ad4c : FAILURE in 39m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c22049bddd8a4972a0b86504d4d4393c : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":5},{"id":"b2050f21319f4820c24c4c0e887cc3bff6460221","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-11 03:32:48.000000000","message":"Patch Set 5:\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bb1ec0b0a14e74d842bac3c1d17eca023369a27e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-12 13:42:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7f9cec04c902241d956e8b1f5f1c555122cec886","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-12 13:49:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"49bdc8b6cd8da932871cba0f54894a4db0e0451d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-12 14:25:13.000000000","message":"Patch Set 7:\n\n(8 comments)\n\nHey, Goutham! Thanks for reviewing this patch :)\nI\u0027ve addressed the answer to the comments and some changes.","accounts_in_message":[],"_revision_number":7},{"id":"18b6c517d951c18ffdc17030ee996eec05cbe73f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 15:31:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dfa9511b4914d00b4f8d49beade1120 : SUCCESS in 3m 58s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/82a5efc4eb5b4fb3bb4ada5b12e305a1 : SUCCESS in 1h 37m 40s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/19a3cff6af2f48f3919b781b045ac1ec : FAILURE in 53m 42s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b5747ff0349b485e97e97db075555397 : FAILURE in 1h 19m 14s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/59ddd5a9e8234d9fa51dab814ffea3cd : SUCCESS in 1h 03m 18s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/cd50b3aa497145e598a2b73002169195 : FAILURE in 1h 08m 56s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbaf657ce40e4f33951fa55e3fa98ab3 : FAILURE in 1h 37m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/50bf17183bb74c6d8d2d30f31d960a79 : FAILURE in 29m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/efab1c3ab2cf45bf8e16b804ea3312a9 : FAILURE in 1h 24m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e4ccd5a2150a45efa1d548bade970ac1 : FAILURE in 43m 04s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4d14eed427034501a33ad0e24f4fcd01 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":7},{"id":"a376d95ab7a35ff3bdb8edb3b8b5812f0721822a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-13 00:37:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6186f16be6aa7530e644bb5141bd6be58a09466d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 02:41:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56eabe6a8ab343e381d534855dc979db : SUCCESS in 4m 25s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/14633f1d7f6441f996d3e53767604347 : FAILURE in 1h 48m 11s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e69fbb78693f4e47bc1a6bc39f360872 : SUCCESS in 59m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b198bbe1b4b5481f8b48ea982bb5edcf : SUCCESS in 1h 14m 05s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e6871ff24e0d4322a600b3f33d1ee58a : SUCCESS in 1h 05m 15s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/6f0a0c06c27f4c2abb4483753ced9927 : SUCCESS in 1h 04m 19s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ceb8001e1bd4b119e25ce66d02575bd : FAILURE in 1h 47m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/fc0324893da844d48b296c85ea193f9e : SUCCESS in 34m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b66393c284254c6fb10bd5d629c6d0ab : SUCCESS in 1h 35m 02s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ec78f0953bbd4337a95fe71994c646a1 : SUCCESS in 46m 11s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/0eebea3d63474c229759158f89bb00d1 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":8},{"id":"a7000c748a88cf172d1cec37272abf6cac9dcf24","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-13 04:00:41.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"94e458882f40650ea7b2b2f3e43c4cfc7a848b59","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-13 04:14:04.000000000","message":"Patch Set 8:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":8},{"id":"28842b7b87e1b1f9c7400cd84aec81f9ae2ffe94","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-13 05:40:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/470a79b/ : FAILURE in 1h 22m 38s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b98fbe4/ : SUCCESS in 1h 23m 34s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"7d5cb4e856ebbe604e066d8a679e89fa7841603c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-13 15:48:17.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"db3b192c8a8d03c04dfca5551959717c7f49da1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 17:46:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75e94a30b4947b383d7a17fea435241 : SUCCESS in 4m 21s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/82cad5569f31429e8b8f9bab2e482be8 : SUCCESS in 1h 36m 14s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d02cff1c9c354928bf268d304c5b813b : SUCCESS in 55m 32s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/18fa2aac9cfb473b80a701aea6ac6742 : FAILURE in 1h 12m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5c66c910653440eca0b4817d0109f132 : SUCCESS in 1h 04m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/b6d59130f2c249e38b334163677ab3ca : SUCCESS in 1h 02m 31s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66d2ef989d064b4bb91c4c294bfd49fb : FAILURE in 1h 52m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/9e5e29983e0b47c2bd6c9ff0e0c24816 : SUCCESS in 32m 40s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9fc7df8262bd4fcf8259364e443a4a61 : SUCCESS in 1h 36m 16s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/405a86bab4ed45bd906b767b61e7c05f : SUCCESS in 50m 50s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/cef640cc1a4c49569d399e6879bbbf38 : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":8},{"id":"3ee4bd73c30048e39a431974f67e45f73968ce15","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-13 19:19:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/adb9ac7/ : FAILURE in 2m 50s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4835630/ : SUCCESS in 1h 33m 03s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"b62c7e81a30be18574e94519b25b5935739fdc3a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-13 19:22:31.000000000","message":"Patch Set 8:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":8},{"id":"8e2ffaa0f129e4b2215d567f5a2c2f1e8f52e70f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-13 23:23:02.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/afe8928/ : SUCCESS in 1h 25m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/22687b9/ : SUCCESS in 1h 37m 51s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"4ab1c8e735dc352c0257007aecac30ab814f6039","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-14 12:09:29.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"948f5910ecf9086a5e9d389dc7a16a159e8dea08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 14:16:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e937c2750e8463f936d90b567b0c37a : SUCCESS in 3m 24s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/186b431db8e1468f8b774e6505c5e284 : SUCCESS in 1h 33m 15s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/bd82b32fdee246cfb81c1d61da192911 : SUCCESS in 58m 08s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9fde046d0a754179b0518eb3bfa1b4a9 : SUCCESS in 1h 11m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f87f4150fd9b4def84cd9261f55bf251 : SUCCESS in 1h 01m 37s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/c84de37593fb48ee88cd5e74833d8563 : FAILURE in 1h 12m 20s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f570523d4cc4c828b513bf50a098ca9 : FAILURE in 1h 25m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/cf04cb32290e448faba0f807c17aac99 : SUCCESS in 27m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4c301a1eff6b465d82daa8a57adb097b : FAILURE in 1h 53m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3a97a66381c74c0594357ae08efb1084 : SUCCESS in 50m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/bc10b718f1114a1985120c13395306ab : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":9},{"id":"deced7c1282f11a1ca6b93320db85100235694b2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-14 15:54:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b1e2cc4/ : SUCCESS in 1h 24m 48s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0ef33bc/ : SUCCESS in 1h 37m 02s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"6c6eb0e7f7b8302975fa80ad449e7c2187c86b66","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-17 18:46:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a18314d1bedd333c3ff2394b2265a1003cf4e6a9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-17 19:20:42.000000000","message":"Patch Set 9:\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"08427891e6c9f9ba2ba3d0fa62f678099cf98e4b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-17 19:21:16.000000000","message":"Patch Set 10: Code-Review-1\n\nPlease see comments on PS#9","accounts_in_message":[],"_revision_number":10},{"id":"0856582aa9db965a1904591fc269e278853f4053","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:36:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be6caa80c944ae4a97f46a2989d88a1 : SUCCESS in 3m 38s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e3910b11bcb0480a9552c52b444bea3b : SUCCESS in 1h 43m 53s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8b47cac8beaa435f8cf9bc7ad19943fa : SUCCESS in 1h 13m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/dfd9a3df522e418a9664818e0f30a68b : SUCCESS in 1h 26m 46s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/cfa432dd46ab4f658909bf4d7979f595 : SUCCESS in 1h 17m 37s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2fbde7ecec6045d7864517bfefd36541 : FAILURE in 1h 28m 30s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61a8130c7c344253839ff97bdb0d96c1 : SUCCESS in 1h 40m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/484f1e01e2eb4fcaabfc241b0d55a08d : SUCCESS in 34m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5760cb68633b48399a889c5b55ce2550 : SUCCESS in 1h 43m 59s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8a946664cf8f4051a2ea9f5064f08e5c : SUCCESS in 42m 14s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/66cd2d8a4a204eff873d11ef1217ce40 : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":10},{"id":"7c314ce46035a23f24ca659da8e4b4903fe7b26d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-17 23:20:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/9bd2055/ : SUCCESS in 1h 43m 57s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/cf1808a/ : SUCCESS in 1h 23m 54s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"9d6a0745c7c290dd1544c84a7499cb1226a96936","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-18 21:06:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b83350ac6fab32e61263933d09e741d4f895257f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-18 21:10:33.000000000","message":"Patch Set 11:\n\n(15 comments)\n\nHello Goutham, thanks for reviewing this patch. Just upload a new PS regarding your comments.\nThanks","accounts_in_message":[],"_revision_number":11},{"id":"2bdba9247ca121736248d0c998dab7f80bd7627e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:06:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4090b6c79ff4a519215062b719409e3 : SUCCESS in 5m 23s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/316f7f212f53420ab5fb98d571ef6ecb : SUCCESS in 1h 29m 31s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8bd9f6a0df4f41ba903df9c12194114c : FAILURE in 1h 00m 26s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/0f9aeed66d4044d89cd3bc8ea882ab8c : SUCCESS in 1h 20m 27s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/15c6b06eba694b53b3cd74ab1526fb05 : FAILURE in 58m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ffab9b25559f45cf86aeeef508a1dc33 : SUCCESS in 1h 00m 37s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58fb036955dc453482e320d21499b5da : FAILURE in 1h 52m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7d3cd0706881490289e5e87980f430ff : SUCCESS in 36m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9b177080f51f4e20a7b74b69c7f9baaa : SUCCESS in 1h 40m 42s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d46fb77746d44a6299d8d2c95c4f8320 : FAILURE in 48m 08s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/99ed22a5c35d4d3f86abb7b735abd753 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":11},{"id":"8ef5afe26133dfd0ef1890f59e31e9ede5656db7","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-19 13:16:58.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"256288313124101e1c0cdddc5de682ebaef6cc8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 15:46:25.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c8e0322cbb4dd3a08ab5697568c1bb : SUCCESS in 2m 49s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d7c71069c5e94ca195e0000eb1206316 : FAILURE in 12m 41s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/9f2df7d3321c4e5db4be8d6056c0f68d : FAILURE in 59m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b6a8f4f7d604453e8797e137297006eb : FAILURE in 1h 11m 55s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5f851ef5dfb04e0fa395601ac899bcaa : SUCCESS in 1h 10m 40s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d755f94f1a6a4f6eaf0f0223057e6c48 : SUCCESS in 1h 07m 54s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/faa3ecb978344023abb22b2370ccb9f7 : FAILURE in 1h 26m 48s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/1432983a0a334e738445c6846568191f : FAILURE in 30m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/58eb22ca15a042dfa512255a327d1e64 : SUCCESS in 1h 37m 12s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0c0d80b3d2f74f1d835de654df97d5cc : FAILURE in 45m 50s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/6e1c281df77b459380c9336e730d33d1 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":11},{"id":"ae5eb62d0f2820afb145dcdef122e3c3c59db839","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-19 18:01:40.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b629326929d56aa8816e0338183dd272bca01803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:42:09.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed08b76459b40078193b83d60b428ca : SUCCESS in 2m 47s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/078a372f8fe64b6eb26914cd925dabfd : FAILURE in 1h 22m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/a0161ff627a341d788ad97fa1970a9e9 : FAILURE in 53m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f7fb57e3e379437c9363f08a3f55b08b : SUCCESS in 1h 01m 01s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/22c8fd36a1954cbaae42a72623935762 : SUCCESS in 59m 00s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/20422ef2465546398ddae28db7f7ac5b : SUCCESS in 1h 02m 27s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0fce91b47244a919ee63efecbf123b0 : SUCCESS in 1h 32m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d958b5cbecec44ffa51c2cc587187ec7 : SUCCESS in 33m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bffc36b3ce534c79abe032652b0834d9 : SUCCESS in 1h 33m 19s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/97c079beb2754612a079b2a4fbfe93b3 : FAILURE in 39m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/f91be27b9d7b4c0e948c0c57a61e6484 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":12},{"id":"a03cbc3f145f7202b49299352d7d1c064baa9721","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-19 20:10:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"49eb6c5a7ac4425166143f00484bf422e68e7d14","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-19 21:08:13.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d38c0ace3f5e2135be8813f5a5bb6dc85bd71833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 23:03:56.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ed0c752cc8478585e0bd4e3d1928da : SUCCESS in 3m 37s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/63d020e3b5614734b92df731a3fe114c : FAILURE in 1h 35m 22s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/44ddf31938aa419b959e7085bc4b5b86 : SUCCESS in 47m 12s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/82c1bb81dedd4e45b5d883f88baffc89 : SUCCESS in 1h 01m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ccbea8a851df465494e9ca5272e25f19 : SUCCESS in 1h 05m 26s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/144302f3a7ad48fd92546ded9d0a044a : FAILURE in 1h 07m 28s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c75ff154b8345779dbd69bb7284a3f1 : SUCCESS in 1h 31m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ee2eed4b197c45d9bbc4121ac9b17ef5 : SUCCESS in 30m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9990e5f11fb74aa0b3967396dfb82efd : SUCCESS in 1h 41m 01s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9bde3dc4cfd04d60a896b000cf35dd91 : SUCCESS in 43m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/3d879c269f734b80837fe542688ca9c6 : SUCCESS in 1h 07m 17s","accounts_in_message":[],"_revision_number":14},{"id":"c1ceb6019e079529941fd4d8c807fd34686df64f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-20 01:33:29.000000000","message":"Patch Set 14:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":14},{"id":"f1e1cdd708087d2fef0d0e7dc61edc9f037753fe","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-20 02:58:46.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/14/upstream-check/manila-cDOT-manila-tempest-plugin-ss/0f1127e/ : SUCCESS in 1h 17m 59s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/14/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f9c364f/ : SUCCESS in 1h 24m 49s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"f6c024951ba14f39f6b0e52c0bb9c53bff53c6c4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-20 06:00:57.000000000","message":"Patch Set 14: Code-Review-1\n\n(11 comments)\n\nHi Douglas, thank you for the changes... some comments inline","accounts_in_message":[],"_revision_number":14},{"id":"a2d72117eacf9ec4a2e1d7544f28e2dc0edab2a9","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-20 16:33:30.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"fb698d97358130d33acca52d8f93d386f631a138","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-20 17:00:42.000000000","message":"Patch Set 14:\n\n(11 comments)\n\nHi Goutham, thanks for reviewing it again. Added a comment regarding share\u0027s creation without providing an AZ.\nThanks.","accounts_in_message":[],"_revision_number":14},{"id":"f5ea8bb405b0034cbb7000b3933f6724d239ad18","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-20 17:34:03.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)\n\nThanks, changes LGTM - Waiting on CI\n\nPlease address the comment inline if you need to make another PS.","accounts_in_message":[],"_revision_number":15},{"id":"2193e4d087ef485c0bc2e414959998ddb7006095","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-20 18:51:27.000000000","message":"Patch Set 15:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":15},{"id":"1bf97efd7b324ac6613348b597ad7dda7483bfbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 18:57:05.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6650ee6967d4196973f78fb46f9914e : SUCCESS in 3m 54s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/29ee007d679f4189b64b138c325b7a71 : FAILURE in 1h 32m 45s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0cba0d2a3600419bac4e24ddb5ef4256 : SUCCESS in 59m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/5226f01465f248e398de6ec10cc73b95 : FAILURE in 1h 13m 31s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/03d88a1ef06747988458b923a052e0bb : SUCCESS in 1h 12m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e5adfbafa3784e51b9d06b04697b4147 : SUCCESS in 1h 17m 59s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c3bcb889753485796c8cdf641d44c95 : FAILURE in 1h 54m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0ae8e05e88ee48cea2f3ca4e89d940fa : SUCCESS in 35m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ef27e88145ab417491ae7085ff4d9d65 : SUCCESS in 1h 24m 43s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5e8b533837474654a85f41ad383f8909 : SUCCESS in 1h 02m 10s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/0cc6f8f0c1de4859b21bc75656b05545 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":15},{"id":"da63f740eafccec2bf65508e79c0f351ff326a4b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-20 19:26:23.000000000","message":"Patch Set 15: Code-Review+2\n\nrecheck\n\n\nfor sanity, let\u0027s recheck - shouldn\u0027t affect the NetApp CI currently running","accounts_in_message":[],"_revision_number":15},{"id":"73ad4d007cc5786b319cf4d576752a04fbea5227","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-20 19:39:52.000000000","message":"Patch Set 15:\n\nrun-netapp-ci\n\n\nThe last one failed on devstack","accounts_in_message":[],"_revision_number":15},{"id":"4d6899e0b9364e153c8ccda04ee51e3bf03f6fa0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-20 19:42:36.000000000","message":"Patch Set 15:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":15},{"id":"b4d52af06f443a70885594c118c1ecfff47053a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:15:24.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d365afe1aeb14922a62b2c5623e8d666 : SUCCESS in 4m 42s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a6184ff2708845e7abe674c028d4291c : FAILURE in 1h 20m 05s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/417ddc8a5bf54868a26d83d8b724fdc3 : SUCCESS in 1h 05m 14s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/de47cf8f28fb43b19792fb7c748b58b6 : SUCCESS in 1h 06m 18s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/39771fa174ae41e5bab9153a406f5384 : SUCCESS in 1h 08m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/222387aebdc84f468649035d5fe2b025 : SUCCESS in 1h 15m 42s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05efef438cce49a0b76285773bd4d522 : FAILURE in 1h 44m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/52a21c6bbcee4f628fc3bc29943989fb : SUCCESS in 29m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4d76779ecc154070a166e7858cd84681 : SUCCESS in 1h 24m 07s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4910f853d0b94f87a9d6eec2b6109afd : SUCCESS in 46m 22s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e4e4a94aafb448c3b46d256ad2a89c14 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":15},{"id":"ad3ec60efaacaf54782807eda64a53be02a87456","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-20 21:24:20.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/75/676475/15/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ac3b21f/ : SUCCESS in 1h 29m 18s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/75/676475/15/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/af946da/ : SUCCESS in 1h 41m 24s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"1ceed2d13c554e69eb42125231f33074332b1f7c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-20 23:55:03.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"076784e3507f980ed15c10953fd4ee3ac7d25dff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 23:55:16.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"dc10f4ffe4d9cf9133f879ddbe9fbebd65217861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:02:58.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5260d928c4fb4bf0a3d822ddb87c999d : SUCCESS in 3m 49s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/db69c56f53bd4417a12152b33e93ed6f : SUCCESS in 46m 24s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/bd60fb37c4274fcf91fcf62f45ebbc33 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":15},{"id":"e3c2011c05a9d640f96f3e25e118abb6a1686e66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:02:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"b7e27e7719a2319e8df9a170456d605619af7008","revisions":{"d110d139f391ce63c2d851c2df589b57f30ba22a":{"kind":"REWORK","_number":1,"created":"2019-08-14 17:51:59.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/75/676475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/1"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-08-14 10:57:40.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.50\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d110d139f391ce63c2d851c2df589b57f30ba22a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d110d139f391ce63c2d851c2df589b57f30ba22a"}]},"branch":"refs/heads/master"},"e35adc33b1b98e3863b6aa47cf81cece841cc1c0":{"kind":"REWORK","_number":2,"created":"2019-08-20 12:02:29.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/75/676475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/2"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-08-20 12:01:18.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.50\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e35adc33b1b98e3863b6aa47cf81cece841cc1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e35adc33b1b98e3863b6aa47cf81cece841cc1c0"}]},"branch":"refs/heads/master"},"ffdf7b0f7365865407692bc4369b91335c26c88f":{"kind":"REWORK","_number":3,"created":"2019-09-02 18:31:31.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/75/676475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/3"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-09-02 18:31:03.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.50\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ffdf7b0f7365865407692bc4369b91335c26c88f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ffdf7b0f7365865407692bc4369b91335c26c88f"}]},"branch":"refs/heads/master"},"a80f77244507a00c08f6681292892b019c705733":{"kind":"REWORK","_number":4,"created":"2019-09-03 21:28:25.000000000","uploader":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"ref":"refs/changes/75/676475/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/4"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2019-09-03 21:27:40.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.50\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a80f77244507a00c08f6681292892b019c705733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a80f77244507a00c08f6681292892b019c705733"}]},"branch":"refs/heads/master"},"34396e69c40846411fbcd7be29398d8ef5a5ec43":{"kind":"REWORK","_number":5,"created":"2019-09-10 23:26:39.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/75/676475/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/5"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-09-10 23:26:12.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/34396e69c40846411fbcd7be29398d8ef5a5ec43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/34396e69c40846411fbcd7be29398d8ef5a5ec43"}]},"branch":"refs/heads/master"},"25c4ff90f8e37be5fc43d4ec7b542e6dde9daad6":{"kind":"REWORK","_number":6,"created":"2019-09-12 13:42:10.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/75/676475/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/6"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-09-12 13:41:54.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/25c4ff90f8e37be5fc43d4ec7b542e6dde9daad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/25c4ff90f8e37be5fc43d4ec7b542e6dde9daad6"}]},"branch":"refs/heads/master"},"bbda193a3dfe374086534f11373c7dc1d73db3e5":{"kind":"REWORK","_number":7,"created":"2019-09-12 13:49:45.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/75/676475/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/7"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-09-12 13:49:33.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbda193a3dfe374086534f11373c7dc1d73db3e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbda193a3dfe374086534f11373c7dc1d73db3e5"}]},"branch":"refs/heads/master"},"8d965981bd068556254e78f13a55f91041fd898d":{"kind":"REWORK","_number":8,"created":"2019-09-13 00:37:37.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/8"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-13 00:36:14.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\nDepends-On: Id8814a8b26c9b9dcb1fe71d0d7e9b79e8b8a9210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d965981bd068556254e78f13a55f91041fd898d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d965981bd068556254e78f13a55f91041fd898d"}]},"branch":"refs/heads/master"},"9141c23902f75881d2916e29fa8d63a06ee70794":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-09-14 12:09:29.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/9"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-14 12:09:14.000000000","tz":0},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9141c23902f75881d2916e29fa8d63a06ee70794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9141c23902f75881d2916e29fa8d63a06ee70794"}]},"branch":"refs/heads/master"},"da0708b0e8da5e6663ea6560877780e1351c140b":{"kind":"REWORK","_number":10,"created":"2019-09-17 18:46:07.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/10"}}},"commit":{"parents":[{"commit":"0d1d29f5618cfbdef31c2acdf537799b01aee232","subject":"Add update share type functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d1d29f5618cfbdef31c2acdf537799b01aee232"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-17 18:45:01.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da0708b0e8da5e6663ea6560877780e1351c140b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da0708b0e8da5e6663ea6560877780e1351c140b"}]},"branch":"refs/heads/master"},"c796af414116fd59c4e594597462c142ed7cd039":{"kind":"REWORK","_number":11,"created":"2019-09-18 21:06:00.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/11"}}},"commit":{"parents":[{"commit":"0d1d29f5618cfbdef31c2acdf537799b01aee232","subject":"Add update share type functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d1d29f5618cfbdef31c2acdf537799b01aee232"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-18 21:04:46.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c796af414116fd59c4e594597462c142ed7cd039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c796af414116fd59c4e594597462c142ed7cd039"}]},"branch":"refs/heads/master"},"83cc93d5dc2496d2db6e8baf1a05ee2082f24112":{"kind":"REWORK","_number":12,"created":"2019-09-19 18:01:40.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/12"}}},"commit":{"parents":[{"commit":"0d1d29f5618cfbdef31c2acdf537799b01aee232","subject":"Add update share type functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d1d29f5618cfbdef31c2acdf537799b01aee232"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-19 18:01:29.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/83cc93d5dc2496d2db6e8baf1a05ee2082f24112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/83cc93d5dc2496d2db6e8baf1a05ee2082f24112"}]},"branch":"refs/heads/master"},"f3783ff36313560f4ad915cf2681ce23cd86bd72":{"kind":"REWORK","_number":13,"created":"2019-09-19 20:10:59.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/13"}}},"commit":{"parents":[{"commit":"0d1d29f5618cfbdef31c2acdf537799b01aee232","subject":"Add update share type functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d1d29f5618cfbdef31c2acdf537799b01aee232"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-19 20:05:32.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3783ff36313560f4ad915cf2681ce23cd86bd72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3783ff36313560f4ad915cf2681ce23cd86bd72"}]},"branch":"refs/heads/master"},"d6e79c4b0c4592769f821aaed35cdb0ed6c53e63":{"kind":"REWORK","_number":14,"created":"2019-09-19 21:08:13.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/14"}}},"commit":{"parents":[{"commit":"0d1d29f5618cfbdef31c2acdf537799b01aee232","subject":"Add update share type functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d1d29f5618cfbdef31c2acdf537799b01aee232"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-19 21:05:26.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6e79c4b0c4592769f821aaed35cdb0ed6c53e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6e79c4b0c4592769f821aaed35cdb0ed6c53e63"}]},"branch":"refs/heads/master"},"b7e27e7719a2319e8df9a170456d605619af7008":{"kind":"REWORK","_number":15,"created":"2019-09-20 16:33:30.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/75/676475/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/676475/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/676475/15"}}},"commit":{"parents":[{"commit":"0d1d29f5618cfbdef31c2acdf537799b01aee232","subject":"Add update share type functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0d1d29f5618cfbdef31c2acdf537799b01aee232"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-06 22:40:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-20 16:33:12.000000000","tz":-180},"subject":"Adds new functional tests for share-network-subnets","message":"Adds new functional tests for share-network-subnets\n\nThis patch adds functional test for share-network-subnets and\nvalidates the new API version 2.51\n\nChange-Id: I60bdb493ad5766f28408a0c877f960922fe44ad0\nPartially-implements: bp share-network-multiple-subnets\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b7e27e7719a2319e8df9a170456d605619af7008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b7e27e7719a2319e8df9a170456d605619af7008"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
