)]}'
{"id":"openstack%2Fmanila-tempest-plugin~677557","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Idd3396834ca204b51671e4422a6223754342e21f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"share_quota","hashtags":[],"change_id":"Idd3396834ca204b51671e4422a6223754342e21f","subject":"Validate share quota after updating","status":"MERGED","created":"2019-08-20 19:35:24.000000000","updated":"2019-08-22 21:05:51.000000000","submitted":"2019-08-22 20:42:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677557-1566506532534-456e8807","meta_rev_id":"1f3faa04d5e0c5e22c96f7a51f0b1311b0c68eba","_number":677557,"virtual_id_number":677557,"owner":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2019-08-22 21:05:51.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-22 20:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2019-08-22 17:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2019-08-22 17:54:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 07:24:28.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2019-08-22 17:54:37.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-08-22 20:42:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 21:05:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40dc79ac61213e5338b152d57958463af5ebf6c4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-20 19:35:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0850667ab3e8033b712d55c23ce2bda7add2888f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-20 19:50:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"30f07182e73e15a28ee6972c2f66c4227ac9d1de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-20 19:57:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nThanks for the changes Vida, and for abandoning https://review.opendev.org/#/c/674871/ and https://review.opendev.org/#/c/676406/ - we\u0027ll continue on this change.. please see comments inline..","accounts_in_message":[],"_revision_number":2},{"id":"633e5489744e3a5dbf35599f2b7fbcab52aa09a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 21:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8896d6f00a4258838ca5bb5dc165ff : SUCCESS in 3m 34s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/57d33b1a27464b579aa177c9e49ae5dc : FAILURE in 1h 31m 52s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0d46dd515b474388bd68cf1a44cc7daa : FAILURE in 30m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9a21a352d5464c239056a4d74fe9c836 : FAILURE in 26m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/30d6e16c879c4198850a9777ad11a5b6 : FAILURE in 27m 21s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/7e8e342611934230823033a2bf2851ba : FAILURE in 1h 05m 00s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c2972edcab642d88564e05586d9339c : SUCCESS in 1h 31m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0b3165ea9a5040a696529ff1af6794c2 : SUCCESS in 28m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9c2f717abd9549719ee5c75634e40109 : SUCCESS in 1h 30m 05s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3d37fea4702a4c08b8cc32df29d6488e : SUCCESS in 46m 35s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8e5bb6ebfc0144adae0f0f40965812c6 : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5a785bb7826b20712cdbc3e90379a2c96a7e4f7f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-21 00:03:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/57/677557/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5162fbd/ : SUCCESS in 1h 18m 44s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/57/677557/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/be41953/ : SUCCESS in 1h 31m 11s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2bafd38bc44c4806a4028027438d722fc0c9dfb","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-21 03:18:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe40d477351174eb75611e9363862ea3ba1a1e24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 04:58:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252e88b8e3424db4bdd4ba24a4740b16 : SUCCESS in 3m 24s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2bf4c80c0eb248ba97ff8500100028ad : FAILURE in 1h 30m 08s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/254bad9a88da40408b63e3dd7e81c918 : FAILURE in 30m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/cf579924bcd5421e82532d40120b6b85 : FAILURE in 25m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ac2b20ed3b8b45a58430fc16bfb31e1b : FAILURE in 27m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d1a5e27204914766969bb81506deab9d : SUCCESS in 1h 00m 39s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06e11215ff6e401fbd851b0674274f0d : SUCCESS in 1h 34m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c609be87458140a191b5d3d884ef989f : SUCCESS in 24m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6181865a3b1945b5bdf5f05c01b19b0b : SUCCESS in 1h 20m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8dbbb681044842d5aa10317314cfd3c8 : SUCCESS in 38m 50s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8db82073c9354225bd3e69677685de52 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":3},{"id":"caf9ef68ccbabed90c5c162cc59bd4788f6c71c0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-21 06:45:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/57/677557/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b2072a0/ : SUCCESS in 1h 24m 18s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/57/677557/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/70a64e5/ : SUCCESS in 1h 46m 59s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"58b700c9fc734433a87f92576bf5e669115ea35c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-08-21 07:24:28.000000000","message":"Patch Set 3: Code-Review+1\n\nVida, thanks for your work","accounts_in_message":[],"_revision_number":3},{"id":"d9c4320e1558d46b8d4dc9267e1c73a012d49a74","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-21 13:39:40.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nThanks for the changes, and your patience. Please see comment inline for an improvement.","accounts_in_message":[],"_revision_number":3},{"id":"16a3ed6190451d55020feb923e4091ddd2c4fc38","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-22 11:46:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\n\u003e (1 comment)\n \u003e \n \u003e Thanks for the changes, and your patience. Please see comment\n \u003e inline for an improvement.\n\nHi Gouthamr,\nThanks for your suggestion, I moved the line below cleanup.\nVida","accounts_in_message":[],"_revision_number":3},{"id":"4a6cc62cf4d6aa685625fb21f5ae9c78fe32b3c5","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-22 11:59:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8909749363c08170c60d723c6c06f40f3c3fc336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:52:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548cd90865004cf4a8c7dee67e653766 : SUCCESS in 4m 56s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1e62520950704dd4b8c65c5bd83d300d : SUCCESS in 1h 32m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/29972b140de1468f88a93ab8967bac58 : FAILURE in 26m 54s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3c6f31f3dc1046978dbebc9d2286cb4a : FAILURE in 31m 23s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/9a7ee254e14042a2a66b03a4089ee668 : FAILURE in 22m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/402d675c43de4a0aa30263d7fa7e685b : SUCCESS in 1h 02m 47s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a378495e79504daea18e648d4f9b4d20 : FAILURE in 1h 50m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/29d1995fb372455ca522f9868fa2d51c : SUCCESS in 27m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0eba04aeed5e403eb992ea2be3bf303e : SUCCESS in 1h 22m 01s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/340ed5ed7b624c608c4e4c66766d4682 : SUCCESS in 44m 37s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/59c4c568299142b4b105cde8684ec6e7 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":4},{"id":"6c2d69c0b0b90dd02aa30ebbe4ddb64b419dea03","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-22 15:21:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/57/677557/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/62c28a7/ : SUCCESS in 1h 18m 19s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/57/677557/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9fade43/ : SUCCESS in 1h 28m 26s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"82b3319973e03077eb2fd23d570f7070319577fa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-22 16:56:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nThanks for moving the cleanup - your commit message needs a fix..","accounts_in_message":[],"_revision_number":4},{"id":"e8638c6b994d8f5907b7183daa3038a11f607bb7","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-08-22 17:42:55.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"48570ddab33ee90e14c42c6b0b4824a5b0a03b7c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-08-22 17:54:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nAwesome, ty for the changes!","accounts_in_message":[],"_revision_number":5},{"id":"cce53f2929030fa404743fa895b03b5b32782b2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 19:37:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547404ffbbec4662ba94d134adc24d86 : SUCCESS in 4m 10s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/cc33285a9ad74bc1bfac2ff3096689de : FAILURE in 1h 21m 37s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ee00bf715f7c4a0faff5640bcc1ef1d1 : FAILURE in 27m 24s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/405c46d1a60e442b90388014645eb4b3 : FAILURE in 29m 30s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7c219ce810af4b1bb31260ff1f76b5b3 : FAILURE in 27m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/a5c982f736074a9ab9a6d0bb677824df : SUCCESS in 1h 09m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/489f924b1efe4ae8a9f73c992548f050 : FAILURE in 1h 49m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/81d5c3041b724937aadced389665fa53 : SUCCESS in 28m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7406cb60b0104722b294e0f67cbd0478 : SUCCESS in 1h 24m 38s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/42b2887656594003805958c758ce568d : SUCCESS in 42m 19s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/cc73bc291a6f433d93bdedf0fcc19122 : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":5},{"id":"a8194fa237f2273c9a02650bd2098fb6fd97152c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 19:37:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b83609c078f91a913b098b90a0a9b8cd41ff5de4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:42:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"cd155ef869f2432a6cb4e5d7cc8e6fb53d10566e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:42:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b0cbf6fdbc4f0aad43b7cb5828e6e9 : SUCCESS in 4m 02s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/fe23cc34e5ab49a481aa51201b77852f : SUCCESS in 41m 48s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/edbf75ba9e9e47eca3d81d8e292d4eaa : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":5},{"id":"0e14a1c44b1d9e2b6164d564b9d66fb6e0150e59","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-22 21:05:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/57/677557/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e4dccfb/ : SUCCESS in 1h 27m 58s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/57/677557/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4c7b38d/ : SUCCESS in 1h 27m 47s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","revisions":{"ad12be2ab1aa737a91c8d844ccc4c2474f3d7e1e":{"kind":"REWORK","_number":1,"created":"2019-08-20 19:35:24.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/57/677557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/677557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/1"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:34:40.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:34:40.000000000","tz":-240},"subject":"Added code, line 362-362, to verify share quota.","message":"Added code, line 362-362, to verify share quota.\n\nChange-Id: Idd3396834ca204b51671e4422a6223754342e21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad12be2ab1aa737a91c8d844ccc4c2474f3d7e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad12be2ab1aa737a91c8d844ccc4c2474f3d7e1e"}]},"branch":"refs/heads/master"},"21f80562900c0155fcf466f7407b008894144637":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-20 19:50:31.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/57/677557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/677557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/2"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:34:40.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:47:52.000000000","tz":0},"subject":"Added code, line 362-362, to verify shares quota after updating it.","message":"Added code, line 362-362, to verify shares quota after updating it.\n\nChange-Id: Idd3396834ca204b51671e4422a6223754342e21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21f80562900c0155fcf466f7407b008894144637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21f80562900c0155fcf466f7407b008894144637"}]},"branch":"refs/heads/master"},"fb12e1eea441a7306d72b6297f79eeab34bbc6cf":{"kind":"REWORK","_number":3,"created":"2019-08-21 03:18:10.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/57/677557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/677557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/3"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:34:40.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-21 03:08:55.000000000","tz":-240},"subject":"Validate share quota after updating","message":"Validate share quota after updating\n\nChange-Id: Idd3396834ca204b51671e4422a6223754342e21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fb12e1eea441a7306d72b6297f79eeab34bbc6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fb12e1eea441a7306d72b6297f79eeab34bbc6cf"}]},"branch":"refs/heads/master"},"1dda0dd2b55c27e6529937308f4bbdcf4d156ca5":{"kind":"REWORK","_number":4,"created":"2019-08-22 11:59:34.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/57/677557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/677557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/4"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:34:40.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-22 11:57:42.000000000","tz":-240},"subject":"Moved resource cleanup after changing share quota","message":"Moved resource cleanup after changing share quota\n\nChange-Id: Idd3396834ca204b51671e4422a6223754342e21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dda0dd2b55c27e6529937308f4bbdcf4d156ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dda0dd2b55c27e6529937308f4bbdcf4d156ca5"}]},"branch":"refs/heads/master"},"493a2dfe6fa92cb81e9c9a20d062076c9a265405":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-22 17:42:55.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/57/677557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/677557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/677557/5"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-20 19:34:40.000000000","tz":-240},"committer":{"name":"vhari","email":"vhariria@redhat.com","date":"2019-08-22 17:41:47.000000000","tz":-240},"subject":"Validate share quota after updating","message":"Validate share quota after updating\n\nChange-Id: Idd3396834ca204b51671e4422a6223754342e21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
