)]}'
{"id":"openstack%2Fmanila-tempest-plugin~677573","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Icb93fc346b71edc7bb3ac989e9751899bb80fbe5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"replication_dhss_true","hashtags":[],"change_id":"Icb93fc346b71edc7bb3ac989e9751899bb80fbe5","subject":"Enable share-replica tests in multitenancy enviroments","status":"MERGED","created":"2019-08-20 21:51:28.000000000","updated":"2019-09-21 01:02:59.000000000","submitted":"2019-09-21 01:02:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677573-1569027779127-c07f50fd","meta_rev_id":"2d1f035ed192e03c16b8a7baf16e9daa53c01af3","_number":677573,"virtual_id_number":677573,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2019-09-14 18:05:46.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-09-21 01:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2019-09-17 19:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-19 15:34:38.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 11:03:12.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2019-09-14 18:05:46.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-17 19:34:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-09-19 15:34:38.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-09-21 01:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a94d3a03226ba49d462368f91b787110752528e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-08-20 21:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b6b241f2bcba9152db8d9bdab756958c36fe4f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 23:41:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2bcddbfcc914265979ccad59e264413 : SUCCESS in 4m 40s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/fb3ae823ddbf496685842a17ad2d2328 : FAILURE in 1h 33m 38s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ac9a01cdb75541e197592876ed121c1c : FAILURE in 29m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7af81a5aa9a44389e8d472d674a8325 : FAILURE in 31m 58s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5f1a882227d840dcae984f707bc6c350 : FAILURE in 32m 25s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f84d0492671b46d2a35da2c9268c6fb3 : FAILURE in 1h 16m 32s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb746f66dc3944e8946456a10e735231 : SUCCESS in 1h 32m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/3cd29aa379dc4f948acd702d8e334540 : SUCCESS in 31m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f4bee7fcb0e54ccdbfc7d29f9cb7be54 : SUCCESS in 1h 34m 14s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0dcd14a7661c4cbb8001d16a8df57e49 : SUCCESS in 44m 38s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/b08b118ed50f45de95832345753f56d2 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a6f208977350002bd7b68eccc5af4e465f7a3d68","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-08-21 00:40:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e1c2261dfee8e3591c1520b7855f0cb5c580ba05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 02:59:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1edf8a543e49758f43e580994a18d0 : SUCCESS in 3m 46s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/866255d9b1524be3a78613c9ee25fc4f : SUCCESS in 1h 33m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/66706f994b9c4706a1cc44a7e3c8f87d : FAILURE in 34m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/24461fe4e1cf446fb6aca40e36d19d2d : FAILURE in 34m 12s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5c1d92eab1114ed3aaaee6dce831a5d5 : FAILURE in 35m 56s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4fd3a95627b34e5188179229f2c6eaef : SUCCESS in 1h 09m 40s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53586818663746adad8a64a3d2aa5618 : SUCCESS in 1h 39m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0b8d1bd59f4547f38b9d692414eaebca : SUCCESS in 33m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57f24d1036c64b6781c6a5db030108eb : SUCCESS in 1h 25m 04s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7feb44ea92114e149328671f591b5cc4 : SUCCESS in 44m 45s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8059f210e0e44aa2a8f1dd42d9ea63c0 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":2},{"id":"aeead9733fbaa617170178e4ed4f42fca56a1872","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-08-21 08:01:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/73/677573/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c99108f/ : SUCCESS in 1h 49m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/73/677573/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6ee12b5/ : SUCCESS in 1h 24m 23s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"9fff4494b52f7a8a575c0d384fe526c010488ffb","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-08-21 13:16:49.000000000","message":"Topic changed from adds_replication_dhss_true to replication_dhss_true","accounts_in_message":[],"_revision_number":2},{"id":"7d254aaf8f9f615597972ca500a4b1783cee5c68","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-08-27 11:20:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nHey, Douglas.\nGreat job on this patch.\nLiked the way you handle with multitenancy and share network subnets support and how you minimized its impact in the existent classes.\nAlso, I\u0027ve missed two tests for replication with share network subnets and addressed two comments for them.\nThanks! :)","accounts_in_message":[],"_revision_number":2},{"id":"6b1ec5da6e007e2f18b4420affe90b7ba2233b3a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-03 16:18:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9065a8406ee1601681165ae007b836ef52718546","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-03 16:25:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23e79e2a93a996fcbecc53c8c8b528fb0292e0fa","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-03 16:33:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fead4a535ec3e808771a1ddb4c87e8f75c33b421","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-03 17:47:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c3c124c7c40252b76e3d30d42c53ef340eed879b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:38:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e906d04791744dbd98ec96ac3311a696 : SUCCESS in 4m 19s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3fc963f440784a89bd0d509994d18ed1 : FAILURE in 1h 32m 59s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1b2bfe0a1dac4044a30b461b55355161 : SUCCESS in 57m 02s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3d62bb7a42254fbcb503957288bd5f9d : SUCCESS in 1h 14m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ea53affdc0654a0388903b17dba2bb76 : SUCCESS in 1h 14m 52s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8f6533a8aad940bb89a99c10432b3bd1 : FAILURE in 1h 03m 32s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/472fa47fc6984041be4adfc6a22a68e5 : FAILURE in 1h 44m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6f3787474b644a57b17bbc66dcf41533 : SUCCESS in 37m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ae079f981f25429d82554439573f30f4 : SUCCESS in 1h 32m 44s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ee700e3db3be408a9546b8763f2ea8fb : SUCCESS in 45m 45s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/0e8864d940fd46f680078e1b5c588497 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":6},{"id":"4d0a9d6494e8d56889a04fceaf9328b251c95f92","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-03 21:28:40.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"c21e4645eae6a608e286fc5699bcbafcbd96a25c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-09-05 11:03:12.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks for answering to my comments, Douglas!\nLGTM :)","accounts_in_message":[],"_revision_number":6},{"id":"208fcf15dc364f6b55a2149e53c2ed0417edc8c8","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-13 22:25:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9a34931e7b1e77808caddb3fbf31a954a0e88a2b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-13 23:21:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fa469b80e2093185edcb6368752b362e871da03e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:22:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae903ef5a1bf4df29bc8fa6d456ddb50 : SUCCESS in 3m 02s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/23f5eda305b341a295cf46cc6035b544 : FAILURE in 1h 26m 01s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/dac83fcab1f54c49921a736c3bfa49b5 : SUCCESS in 52m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e914c81900ae45fb83140bff9d7dfec8 : FAILURE in 1h 08m 42s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/97125be1ca364cbbb1e86d1913f9fd89 : SUCCESS in 59m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e934439b4dbe4204a5d260b70c94b398 : FAILURE in 1h 11m 05s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2659508f3dcf4b31942b638e23694726 : SUCCESS in 1h 32m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/8c4444d4d1dd441b9c8b376c54f4a414 : SUCCESS in 34m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/082632cbbed4439ca4e372f6e2da0f99 : SUCCESS in 1h 38m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8ae77b45ba564fdcaacbd31283c10c27 : FAILURE in 48m 07s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c12ad74a8b4d4de99b31e26a9dd32042 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":8},{"id":"2eb76dc2506c192d0f313600891c7bd100c97351","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-14 13:59:30.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"688c100484a8c807071fe1faa47d97166918e854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 15:40:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc73106e09148bd8d2770fe8defcda3 : SUCCESS in 4m 12s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/30171b75b9ce4b88b3dd9894a538df93 : FAILURE in 1h 21m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/eb0bf20ca4a5447089d471a55093572c : SUCCESS in 54m 11s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9dd96a9d7a874ff4a33de15aced561e5 : SUCCESS in 1h 13m 05s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3d82922c01b947e887e8168b46a88222 : FAILURE in 1h 04m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/769c68ef82b3421cb40ed863dc816c0c : FAILURE in 1h 11m 01s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff06bd27629147ffb528718e680724cf : SUCCESS in 1h 32m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/fb7fca19504d4a5ba5840d70dcfe333c : SUCCESS in 33m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ff76e122e4bb42c19b2ea9185fa90339 : SUCCESS in 1h 28m 53s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3846a7f7c3f844b9948f2e4da25e6340 : SUCCESS in 41m 09s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8bbddeceafa04a6fae6d232c80135d6a : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":9},{"id":"705983c93bb25d655bbee2e4dc81a075d784fb39","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-14 18:05:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/73/677573/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2e36056/ : SUCCESS in 1h 30m 36s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/73/677573/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/40772df/ : SUCCESS in 1h 38m 39s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"b2341b97e6a5b541cafed4de4acf85fc4dc423c5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-17 19:34:33.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)\n\nThanks, this is good test coverage.","accounts_in_message":[],"_revision_number":9},{"id":"56136a41f973ab00d15488e82982409f7ee9eb6c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-09-19 15:34:38.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1e7bf15d5ba58d392795c2fea89207285b486e71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 23:55:17.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d0a46691393fb20cfc0b5fb73d71206ea3b6cc2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:02:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1de9372111d4c719ab704b4c2570851 : SUCCESS in 3m 18s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/cd4756ce05854943ae89a9c9ad9b7e86 : SUCCESS in 45m 38s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/75c795f0da744f0ebb42696e28233c5d : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":9},{"id":"2d1f035ed192e03c16b8a7baf16e9daa53c01af3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:02:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"bd4e78c6222834db5bb0130e656df1b6d66bf4ca","revisions":{"a8df90e18bf94d1af4d44bb708b82879d1a6b97e":{"kind":"REWORK","_number":1,"created":"2019-08-20 21:51:28.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/1"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-14 20:54:12.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-20 16:37:52.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and \nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a8df90e18bf94d1af4d44bb708b82879d1a6b97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a8df90e18bf94d1af4d44bb708b82879d1a6b97e"}]},"branch":"refs/heads/master"},"b0e7b9bbc1232f564def4d0ecea69568ce67027c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-21 00:40:24.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/2"}}},"commit":{"parents":[{"commit":"267dd04fc7518a274622a9ff341db9ffe479f5ee","subject":"Merge \"Comment test tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/267dd04fc7518a274622a9ff341db9ffe479f5ee"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-14 20:54:12.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-08-21 00:40:22.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0e7b9bbc1232f564def4d0ecea69568ce67027c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0e7b9bbc1232f564def4d0ecea69568ce67027c"}]},"branch":"refs/heads/master"},"8d75a9200a65e4deb05a92a76991297fbdc699e1":{"kind":"REWORK","_number":3,"created":"2019-09-03 16:18:48.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/3"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-03 16:17:55.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d75a9200a65e4deb05a92a76991297fbdc699e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d75a9200a65e4deb05a92a76991297fbdc699e1"}]},"branch":"refs/heads/master"},"b1db31787f1e5813f387f891e1809a91478ca51b":{"kind":"REWORK","_number":4,"created":"2019-09-03 16:25:44.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/4"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-03 16:25:33.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b1db31787f1e5813f387f891e1809a91478ca51b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b1db31787f1e5813f387f891e1809a91478ca51b"}]},"branch":"refs/heads/master"},"8d5c659541a83164648c11c7c34c49af018373d0":{"kind":"REWORK","_number":5,"created":"2019-09-03 16:33:51.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/5"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-03 16:32:31.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d5c659541a83164648c11c7c34c49af018373d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d5c659541a83164648c11c7c34c49af018373d0"}]},"branch":"refs/heads/master"},"cf749f00a362b75a2a0102e27919cf9975fa13c4":{"kind":"REWORK","_number":6,"created":"2019-09-03 17:47:16.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/6"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-03 17:36:27.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf749f00a362b75a2a0102e27919cf9975fa13c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf749f00a362b75a2a0102e27919cf9975fa13c4"}]},"branch":"refs/heads/master"},"00cbcfa88fead91ae1751287af8ccad5dd04bce6":{"kind":"REWORK","_number":7,"created":"2019-09-13 22:25:37.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/7"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-13 22:25:21.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00cbcfa88fead91ae1751287af8ccad5dd04bce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00cbcfa88fead91ae1751287af8ccad5dd04bce6"}]},"branch":"refs/heads/master"},"7356fede9ac26feadb183e7d1e08ac79bd800473":{"kind":"REWORK","_number":8,"created":"2019-09-13 23:21:04.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/8"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-13 23:20:52.000000000","tz":-180},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.50.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7356fede9ac26feadb183e7d1e08ac79bd800473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7356fede9ac26feadb183e7d1e08ac79bd800473"}]},"branch":"refs/heads/master"},"bd4e78c6222834db5bb0130e656df1b6d66bf4ca":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-09-14 13:59:30.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/73/677573/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/677573/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/677573/9"}}},"commit":{"parents":[{"commit":"493a2dfe6fa92cb81e9c9a20d062076c9a265405","subject":"Validate share quota after updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/493a2dfe6fa92cb81e9c9a20d062076c9a265405"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-02 20:16:30.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-14 13:59:19.000000000","tz":0},"subject":"Enable share-replica tests in multitenancy enviroments","message":"Enable share-replica tests in multitenancy enviroments\n\nThis patch enables share-replica tests to be executed in a multitenancy\nenvironment, providing a share-network to be used by both shares and\nshare-replicas.\n\nThe multitenancy test for share-replicas will be enabled only for API\nversion equal or greater than 2.51.\n\nDepends-On: I60bdb493ad5766f28408a0c877f960922fe44ad0\nChange-Id: Icb93fc346b71edc7bb3ac989e9751899bb80fbe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd4e78c6222834db5bb0130e656df1b6d66bf4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd4e78c6222834db5bb0130e656df1b6d66bf4ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
