)]}'
{"id":"openstack%2Fmanila-tempest-plugin~684803","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I58e67c93daaf6ac8573b8b7b54ba1f2b5958a931","project":"openstack/manila-tempest-plugin","branch":"master","topic":"skip_update_share_type_test","hashtags":[],"change_id":"I58e67c93daaf6ac8573b8b7b54ba1f2b5958a931","subject":"Skip update share type tests for microversion lower than 2.50","status":"MERGED","created":"2019-09-25 16:11:23.000000000","updated":"2019-09-26 18:49:58.000000000","submitted":"2019-09-26 18:49:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684803-1569523798323-e03b07bf","meta_rev_id":"29c01dcb38820e85bdd03f738e56d251f0c5bc5b","_number":684803,"virtual_id_number":684803,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-25 18:37:01.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2019-09-25 20:16:01.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 18:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2019-09-25 17:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2019-09-26 17:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 20:16:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-26 17:12:36.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-09-26 18:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74b23cdcb2036326657ce057ff4e2a41b40cef2e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-25 16:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e3196919e83970e125afcffd8ac2be3c8e10bff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-25 17:25:06.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"f9b2b5cf6194896df2fd19e23d47fc00a9e1a182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:04:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234b67dd53cd4302aa0eef2780615509 : SUCCESS in 2m 46s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2233fe5dc5934370a33cc4a47b0a9e7c : SUCCESS in 1h 25m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e9c927b6c8464403b40b6485b39b2b1a : SUCCESS in 1h 00m 06s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f32833af831b4463ab5adb11109ff20a : SUCCESS in 1h 23m 03s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0bf658b884fc4d9c9fa931bcfefab66f : SUCCESS in 1h 22m 24s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/14078e5b756c4d948bcb30ac79f4265a : FAILURE in 1h 16m 48s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88ac21fe1499446d9ede4a50444009e1 : SUCCESS in 1h 46m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ca06a604cfc548b8999e177aa957842f : SUCCESS in 41m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dc3a758987ac40dca591f61fce2a34ae : SUCCESS in 1h 28m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/01d1efd1fbac4d628d8ecc46894a491a : SUCCESS in 53m 36s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ca94d06dd28d442d806577c6bf8ea20f : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c5fc41d062a35c60b669dd8b35bc1530843b5398","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-09-25 18:37:01.000000000","message":"Patch Set 1:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":1},{"id":"7895b95f4571493da35242128adfe70ff3894a49","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-25 20:16:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://54.183.250.52/logs/03/684803/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1943e58/ : SUCCESS in 1h 21m 31s\n- manila-cDOT-manila-tempest-plugin-no-ss http://54.183.250.52/logs/03/684803/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d68d04e/ : SUCCESS in 1h 32m 38s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ae27747e8cfe592acceb57723053770d001c9b9d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-26 17:12:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f54a1da37f75ce928e01e8b6a4dca329131bc53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 17:12:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"104497a91009d419adfabf6c6b2ecaa8db46e293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 18:49:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d04aacdaf298491d869e1bce94158b63 : SUCCESS in 3m 01s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4759bbe547314e8b809e9d9e6a8ec62e : SUCCESS in 1h 34m 21s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/009533e771764370ac205a0cc494ee49 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b27758d9f533527bb6991971a7a524f195981ceb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 18:49:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92925822a3f63e8ead84be7736416829f4ce69f1","revisions":{"92925822a3f63e8ead84be7736416829f4ce69f1":{"kind":"REWORK","_number":1,"created":"2019-09-25 16:11:23.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/03/684803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/03/684803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/03/684803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/03/684803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/03/684803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/03/684803/1"}}},"commit":{"parents":[{"commit":"e2eb90b3f4465a188351677ccc4396c265d9efe4","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2eb90b3f4465a188351677ccc4396c265d9efe4"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-25 14:49:52.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-09-25 16:08:51.000000000","tz":-180},"subject":"Skip update share type tests for microversion lower than 2.50","message":"Skip update share type tests for microversion lower than 2.50\n\nChange-Id: I58e67c93daaf6ac8573b8b7b54ba1f2b5958a931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/92925822a3f63e8ead84be7736416829f4ce69f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/92925822a3f63e8ead84be7736416829f4ce69f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
