)]}'
{"id":"openstack%2Fmanila-tempest-plugin~688829","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9","subject":"DNM - debugging ....","status":"ABANDONED","created":"2019-10-15 22:38:24.000000000","updated":"2020-03-29 20:30:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9755c8473560029fea1b5d3fbb355aad195200f4","_number":688829,"virtual_id_number":688829,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-26 13:35:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 01:52:57.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-26 13:35:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"643d3ed77c0e7484522d1bd35e34a3cc455d255c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-15 22:38:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdcdf68b2d3e1352c1faa0f09d3adb07cbc45265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 00:10:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949ed9169c5944d09a21ba0fe7028ecb : SUCCESS in 2m 53s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/63a903c260da4c2f80025441f2a0ef79 : SUCCESS in 1h 28m 42s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b58cc1b8d70444d180ebc97678b4e23f : SUCCESS in 48m 52s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/32331d0cd9e24cd5842e579327808673 : FAILURE in 1h 03m 28s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/53a7d4c91bb344ea85488b5886c26e68 : FAILURE in 48m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/0fc5d83cf03047e384288fac37e5942a : SUCCESS in 1h 12m 00s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f1b2bb209454d0888fcf00890d70325 : SUCCESS in 1h 29m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/63a2de86721e4441947af4a9d6df053e : FAILURE in 26m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/710d7252c37a4c8199de5aabf19b6048 : FAILURE in 1h 13m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d60591570eb847e4a0e7557043e486e6 : SUCCESS in 39m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e8c15b69f373432aa174d0f84b111f83 : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":1},{"id":"41a02b1b88e5e243b1274172c8427aaa9df5b29b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-16 01:13:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d957e4f0eaa9a90f06ca277fcb8b32807016125f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 04:40:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f3e8ff2697b48db85540ff9876ac49c : SUCCESS in 4m 32s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/094355688f814196b199e18bd722a7f1 : SUCCESS in 1h 28m 48s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f8164da334f24ee4b7b5b8aeaa099c1c : SUCCESS in 56m 44s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c3d73639cdf14ee48127839c492e8ff6 : FAILURE in 1h 14m 13s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/8e8efcc34efa45cf89b907edb33f491f : SUCCESS in 58m 52s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/963d2cd74d064490b2abeab24afcc350 : FAILURE in 1h 04m 47s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e16bae3da51b437d92f984a42e5919a7 : FAILURE in 1h 28m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/61d098eb1ded4490bb3087d5666197fa : SUCCESS in 31m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/720a9c1c93624cc3b0f93d2cb0f3696e : FAILURE in 1h 23m 07s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/067ba520609d46e9857eeb49e8c4a8a3 : FAILURE in 45m 05s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c10f26fc41e644b2ac359b06465a5685 : FAILURE in 56m 26s","accounts_in_message":[],"_revision_number":2},{"id":"43a247394249518b1c02559ce5ec1c866b78ffa6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-16 10:19:48.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"02d722f81d3b41f65f088b6e7488a52328f75d9e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-16 14:16:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1fbe3fc3c9f5a54d278efb14f526dd81ecb1b434","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-16 15:25:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d887b44cca963a35b8ce31d92f387612ffec83f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:10:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f6a929dcd44e8e9be57dc8f7447cd3 : FAILURE in 3m 30s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/caf5a3af48034ab79f22210d154aa0e2 : FAILURE in 1h 18m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/20a3f37aaf6a4c56ab73336576827ce4 : SUCCESS in 55m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/06ae434d3dce4b679830d8d7789af30d : SUCCESS in 1h 12m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ec9dd15db5e14c119fec5ccaf733f999 : SUCCESS in 1h 13m 27s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/690c5f591acd448da50f6277446f0528 : FAILURE in 1h 10m 02s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6fef59405fb405b802400eedc598f19 : FAILURE in 1h 38m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a3c66cdab2b24a8f83b9daa5a79a153e : SUCCESS in 29m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d83194c15d9d4929bd44632ccf616076 : FAILURE in 1h 20m 58s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3bf3f5fea07c4741b1103f643b4f7806 : SUCCESS in 48m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/1416d96e5bcb4def902382b2662cb465 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":4},{"id":"9afc700f750349633ffe6c796c90e6b47a268a79","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-16 17:47:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2494f65c4b18d36eb04fe440310d2f6410d9233a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 21:58:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b714005e2ae411b98b9be7016efb613 : SUCCESS in 6m 09s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/4014685731ba43a0a6d9089f24785d46 : SUCCESS in 1h 46m 19s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e6f5407c18ea4f518d93f11e53fd90b4 : SUCCESS in 54m 40s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/870986a08263471895eea91d83249f2b : SUCCESS in 1h 18m 30s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/fa3cf6f86c32441084d2d22866dcdf55 : SUCCESS in 1h 07m 14s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/569c65e544c840948b6cbd0600f58740 : FAILURE in 1h 10m 37s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4874bfe97d344884bc3103c158c3fa2a : FAILURE in 1h 47m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7e6c0c53053e4241960efd1b1147c221 : SUCCESS in 35m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7d696bbbbc4b4f9887e54e64afb8d1ca : FAILURE in 30m 37s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6790d92a333a4582925eee69995076a9 : SUCCESS in 44m 53s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c0397f63f9874ea49e50ad0ebd88a494 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":5},{"id":"1fedd58f97ae351421b5a72191474eebdf3a3810","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-17 01:52:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/29/688829/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/413f1ae/ : SUCCESS in 1h 26m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/29/688829/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9404a97/ : SUCCESS in 1h 33m 25s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"93a45964a6f3e34e58736688ea74b619380f57e5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-25 18:16:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"183f720b1abd816abfb93297dc8274e7ad30ac47","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-25 19:56:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2aeafa9ca236efc855311a32eb62d4ee4e1b05c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 21:47:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd86ea64ba394f9ea2dfda808cb0768a : SUCCESS in 2m 46s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1c7df1b80d08402e9864cae1d27d1045 : FAILURE in 1h 33m 55s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/962ee525a62140a1a49db4b2a3628b62 : SUCCESS in 1h 18m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/874b9bec51ca4ab0a03c6e19d2833e44 : FAILURE in 1h 21m 04s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/88885ade256049749211cde7a0e25398 : SUCCESS in 57m 50s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d948fb9eca434de2aa90af92d1544a0e : FAILURE in 1h 17m 57s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f66e8e21a22c45e99b694e995ba48772 : FAILURE in 1h 45m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/17bd56a004344fe69e6fccbd4e32e2e2 : SUCCESS in 37m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ca1f8a42569b407293fa6d1891c47de8 : FAILURE in 1h 46m 39s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2b8d765a18e349f0b13094d1275def64 : SUCCESS in 39m 34s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/89290b0da14f405c97ee2dfd306648a5 : FAILURE in 1h 17m 04s","accounts_in_message":[],"_revision_number":7},{"id":"bdc69d29edfd05d584c56127e14d757f68f55061","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-26 10:08:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8b7f55e7ea289f24483218e0fdef52105860a5ad","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-26 11:05:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8ba6b90b4882e0da6d663af7b09e761a1635b4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 13:35:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a170d0d3e34beca88db64d47a86970 : SUCCESS in 4m 03s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/6ab0704669af4d6fa413f2643aa15b3e : FAILURE in 1h 19m 07s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6fb3150f3c2547ff8483b837f894e21a : SUCCESS in 42m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/7e2e481cf9be4bf3ab8203c58306fa42 : FAILURE in 47m 44s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/597abb34b10e41ae916ba878d43ff4d9 : FAILURE in 1h 00m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e2b72ca0ca554c488db758afeb122d1a : SUCCESS in 54m 11s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66d0b2467a9a44638004e75668b5444f : FAILURE in 49m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e9ba2f74d22345a88b8766f01f94a4d7 : SUCCESS in 25m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ed5d44978a8f4282a58c6bc64451b054 : FAILURE in 1h 57m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/94d6c3ebdfa4473981e51115043d9050 : SUCCESS in 43m 49s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/161af97c11684155a5898fc193c59874 : FAILURE in 59m 24s","accounts_in_message":[],"_revision_number":9},{"id":"9755c8473560029fea1b5d3fbb355aad195200f4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-29 20:30:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"ebeae9159a0f2a68c835a50a006866cf5869a681","revisions":{"adecff63792a4307ee2989a01245ead8261ee8d5":{"kind":"REWORK","_number":1,"created":"2019-10-15 22:38:24.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/adecff63792a4307ee2989a01245ead8261ee8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/adecff63792a4307ee2989a01245ead8261ee8d5"}]},"branch":"refs/heads/master"},"911617e546f5c1d11d435577368903aeddba7e65":{"kind":"REWORK","_number":2,"created":"2019-10-16 01:13:48.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/2"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-16 01:13:43.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/911617e546f5c1d11d435577368903aeddba7e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/911617e546f5c1d11d435577368903aeddba7e65"}]},"branch":"refs/heads/master"},"5c772ad2ef5f309648deb46b26d0273a31301985":{"kind":"REWORK","_number":3,"created":"2019-10-16 14:16:47.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/3"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-16 14:15:36.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c772ad2ef5f309648deb46b26d0273a31301985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c772ad2ef5f309648deb46b26d0273a31301985"}]},"branch":"refs/heads/master"},"7ca7678c387184d2509b3854aae975eaeee9bdf7":{"kind":"REWORK","_number":4,"created":"2019-10-16 15:25:49.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/4"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-16 15:24:39.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ca7678c387184d2509b3854aae975eaeee9bdf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ca7678c387184d2509b3854aae975eaeee9bdf7"}]},"branch":"refs/heads/master"},"1dfb6113142560230834a68f7072460587b9cfda":{"kind":"REWORK","_number":5,"created":"2019-10-16 17:47:29.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/5"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-16 17:46:09.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dfb6113142560230834a68f7072460587b9cfda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1dfb6113142560230834a68f7072460587b9cfda"}]},"branch":"refs/heads/master"},"c033d9ef2d87f15073b742405fd6b045ed57d867":{"kind":"REWORK","_number":6,"created":"2019-10-25 18:16:36.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/6"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-25 18:15:51.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c033d9ef2d87f15073b742405fd6b045ed57d867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c033d9ef2d87f15073b742405fd6b045ed57d867"}]},"branch":"refs/heads/master"},"62b1de75308fe2cf8d341a5b1b4bf84cc772e55d":{"kind":"REWORK","_number":7,"created":"2019-10-25 19:56:28.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/7"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-25 19:56:19.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62b1de75308fe2cf8d341a5b1b4bf84cc772e55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62b1de75308fe2cf8d341a5b1b4bf84cc772e55d"}]},"branch":"refs/heads/master"},"d12bc24324277bcbc42859bc83e549dad15770ee":{"kind":"REWORK","_number":8,"created":"2019-10-26 10:08:01.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/8"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-26 10:06:42.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d12bc24324277bcbc42859bc83e549dad15770ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d12bc24324277bcbc42859bc83e549dad15770ee"}]},"branch":"refs/heads/master"},"ebeae9159a0f2a68c835a50a006866cf5869a681":{"kind":"REWORK","_number":9,"created":"2019-10-26 11:05:45.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/29/688829/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/688829/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/688829/9"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 22:38:15.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-26 11:04:27.000000000","tz":-240},"subject":"DNM - debugging ....","message":"DNM - debugging ....\n\nChange-Id: Ib5d85d6532e2bbd14868f7730188ed6d7ccfcad9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ebeae9159a0f2a68c835a50a006866cf5869a681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ebeae9159a0f2a68c835a50a006866cf5869a681"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
