)]}'
{"id":"openstack%2Fmanila-tempest-plugin~689569","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8d8063d46284076499d868a5f9b3d13a2a78b5b0","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/scenario-tests","hashtags":[],"change_id":"I8d8063d46284076499d868a5f9b3d13a2a78b5b0","subject":"Add knowledge of storage network to scenario tests","status":"MERGED","created":"2019-10-18 19:56:47.000000000","updated":"2020-03-24 15:46:01.000000000","submitted":"2020-03-24 15:46:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689569-1585064761112-81340386","meta_rev_id":"4f59682bc5c89e21690cc2ce0c2a63637cd14d7e","_number":689569,"virtual_id_number":689569,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-03-24 14:23:22.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-03-21 03:46:11.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 15:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"value":0,"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2020-03-24 14:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-24 14:42:10.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"value":0,"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-24 14:42:10.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-24 01:45:52.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"date":"2020-03-24 01:45:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"value":0,"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 19:57:54.000000000","updated_by":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"reviewer":{"_account_id":29455,"name":"Jason Grosso","email":"jgrosso@redhat.com","username":"jgrosso"},"state":"REVIEWER"},{"updated":"2019-11-20 18:33:03.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-03-21 03:46:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 01:45:38.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2020-03-24 01:45:52.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-03-24 14:38:52.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-03-24 14:42:10.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-03-24 15:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cb41ce512b7417152d859e9700f8b866573feef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-18 19:56:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b56cb79ac8d0a27b05fe92ff8e887abac9cda03d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-18 21:13:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"140fbf83e17a4fc7fbbd52028b7ab4dffd7540a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 22:46:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1514f07e0e44b88dc739f899cb774a : SUCCESS in 3m 47s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2eef8d27cf4c44c09ca2dd1381f8100e : SUCCESS in 1h 25m 31s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ae09af001acd4f5dbcf6f2fe67e0630f : SUCCESS in 53m 32s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/579bbb2422ee430aacaf4b1ca02e02e6 : SUCCESS in 1h 05m 01s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3ebb30f5e7eb43518b770afa45331d21 : FAILURE in 56m 32s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/b113e6710e8146d7adce1b4d5ff58603 : FAILURE in 1h 02m 36s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96529e4c9404495d82e3ce3aff16d9ab : FAILURE in 1h 26m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/beed5684e76a4669b5c8ec3ee1f1fa16 : SUCCESS in 26m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bcb632fe9cd7495fb31fd0237536e2ef : FAILURE in 1h 19m 56s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/048ee9e3fc2542599412c512f7cf28d1 : SUCCESS in 45m 01s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/33415304734446a4bc39eb4d9d6fce09 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6d5e21a57c6b46257febc51f5f5f64d8b1e18337","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-19 00:19:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/69/689569/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/fb093e1/ : SUCCESS in 1h 20m 15s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/69/689569/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a3837a9/ : SUCCESS in 1h 32m 24s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"68b8764853e2364f384de1748bd4c4b30d7e4097","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2019-11-20 18:33:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2b9d9bcdcc281a79f160ee08dfec3b6c3656ac0f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-20 23:45:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1e6873022d07b636a2664cf67c32a40452f5569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 01:38:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d13af1f54448e5ba26fedbc667c168 : SUCCESS in 3m 26s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d43ab94f5812468da115ab751a4a6e85 : FAILURE in 1h 19m 24s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1d1a97ea8b8c4167baf76790190ed8bb : SUCCESS in 57m 06s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/023d0b6de79b42f9911fc033a724a152 : FAILURE in 1h 24m 26s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f7e5aeee8d8f40118d5a448355594d23 : SUCCESS in 52m 15s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/791304f86fc94eea823205efba960d77 : SUCCESS in 1h 01m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd6381daf27f4d3fbf861228f8e5e1b0 : FAILURE in 1h 51m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/39a93c823706486590adf66d59210dfb : SUCCESS in 26m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3013002ce5f74ed78417de979b51b0ed : SUCCESS in 1h 17m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c611abec847f4087946c0da80832463f : SUCCESS in 50m 46s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/48974087142243c58600f20d6a5e65eb : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":3},{"id":"e62939ac77e22a716fa7be9e3f7c37587639ad5e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-21 03:46:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/69/689569/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c8e6e5f/ : SUCCESS in 2h 06m 53s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/69/689569/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1e54788/ : SUCCESS in 1h 24m 21s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"642ceeefd191b5250f8256c6d0eb771b999efec1","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-24 14:05:37.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Goutham, this improvement looks good and very useful.\nIs there any zuul job that can test this change?","accounts_in_message":[],"_revision_number":3},{"id":"6bc1eadae75471798d3e3c1a0f3c4691943995dd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 14:23:22.000000000","message":"Patch Set 3:\n\n\u003e Thanks Goutham, this improvement looks good and very useful.\n\u003e Is there any zuul job that can test this change?\n\nHey Douglas - not yet. We\u0027re testing this with CephFS-NFS on TripleO, and we\u0027ll likely get that into a TripleO job here soon...","accounts_in_message":[],"_revision_number":3},{"id":"a02a950d4b81dd58aeeb380cc3a16e46998f72c1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 14:23:33.000000000","message":"Topic changed from scenario-testing-patch2 to bp/scenario-tests","accounts_in_message":[],"_revision_number":3},{"id":"e6eda1ae6466268415994d6f0f9c1ca11898161c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-24 14:38:52.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for sharing. LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"a3c5aba90331ff6f40797710bff084c1a3d0593d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-24 14:42:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"27f806ea2469bee3539bd0bae7abe27c0017d9cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:42:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bc4526dda832cf051ac081f911586bec426660d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 15:46:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175854f0a8de40b3a34b222e7b3b9b16 : SUCCESS in 6m 01s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0bd909feb3cf4a55b6d6081873463b6b : SUCCESS in 43m 36s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/d96457c3ae28407ebd8831272eeb1535 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4f59682bc5c89e21690cc2ce0c2a63637cd14d7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 15:46:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"37ee6779e5ba4ab960e170a50bc6b27f1f4c6691","revisions":{"e24a3cc33ed0e564ec753ec027a9d7104518465c":{"kind":"REWORK","_number":1,"created":"2019-10-18 19:56:47.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/689569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/689569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-18 19:53:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-18 19:53:22.000000000","tz":-420},"subject":"WIP: Add knowledge of storage network to scenario tests","message":"WIP: Add knowledge of storage network to scenario tests\n\nChange-Id: I8d8063d46284076499d868a5f9b3d13a2a78b5b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e24a3cc33ed0e564ec753ec027a9d7104518465c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e24a3cc33ed0e564ec753ec027a9d7104518465c"}]},"branch":"refs/heads/master"},"dda4b998af6535c441278d96887e12918fb6a08f":{"kind":"REWORK","_number":2,"created":"2019-10-18 21:13:47.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/689569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/689569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/2"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-18 19:53:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-18 21:13:28.000000000","tz":-420},"subject":"WIP: Add knowledge of storage network to scenario tests","message":"WIP: Add knowledge of storage network to scenario tests\n\nChange-Id: I8d8063d46284076499d868a5f9b3d13a2a78b5b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dda4b998af6535c441278d96887e12918fb6a08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dda4b998af6535c441278d96887e12918fb6a08f"}]},"branch":"refs/heads/master"},"37ee6779e5ba4ab960e170a50bc6b27f1f4c6691":{"kind":"REWORK","_number":3,"created":"2020-03-20 23:45:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/689569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/69/689569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/69/689569/3"}}},"commit":{"parents":[{"commit":"39ef895d112f947564d7ca52a73a5b2afb82f741","subject":"Fix link to bug tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39ef895d112f947564d7ca52a73a5b2afb82f741"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-18 19:53:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 23:45:43.000000000","tz":-420},"subject":"Add knowledge of storage network to scenario tests","message":"Add knowledge of storage network to scenario tests\n\nIn many deployments, shared file systems are exported\nover a shared storage network. Typically, this shared\nstorage network is not reachable from within\nprivate tenant networks. Our tests currently\nhave no cognizance of this topology, and expect that\nshares are always exported on a network routable or\ndirectly reachable through private tenant networks.\n\nAdd a configuration option to allow configuring the\nname or UUID of a shared storage network. If specified\nthis network will be plugged into VMs created by the\nscenario tests and access control will be on the\nstorage network IP, if NFS.\n\nChange-Id: I8d8063d46284076499d868a5f9b3d13a2a78b5b0\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37ee6779e5ba4ab960e170a50bc6b27f1f4c6691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37ee6779e5ba4ab960e170a50bc6b27f1f4c6691"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
