)]}'
{"id":"openstack%2Fmanila-tempest-plugin~691186","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8c10d169302c6f1c41e238ae9f787455a2b15a4d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I8c10d169302c6f1c41e238ae9f787455a2b15a4d","subject":"[ussuri][goal] Drop python2.7 support","status":"MERGED","created":"2019-10-25 05:22:42.000000000","updated":"2020-01-06 17:28:36.000000000","submitted":"2020-01-06 17:28:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691186-1578331716197-17819b5d","meta_rev_id":"ee50ef344ad7036a993527f438b38f9aeff92061","_number":691186,"virtual_id_number":691186,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"date":"2019-12-11 04:59:18.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 17:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2019-12-13 18:28:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 16:08:53.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2019-12-13 14:02:48.000000000","_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"date":"2019-12-13 14:02:44.000000000","_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 16:08:53.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 04:59:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-13 14:02:44.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2019-12-13 14:02:48.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2019-12-13 18:28:19.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-01-06 16:08:53.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-01-06 17:28:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d08ff056fdb7ebc3248ec72d144cf04061875ea","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-25 05:22:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ddbe3d1f4035c4deac4fb0b94dd7a1c09c99952","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-25 06:23:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"15c7577b67539820ee9e46b596ccac724af41539","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-25 06:41:16.000000000","message":"Patch Set 2: Workflow-1\n\nCan be workflowed only on or after Ussuri Milestone 1 (Dec 12). See [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html","accounts_in_message":[],"_revision_number":2},{"id":"c0f33e5547d1280a05588ff469dddfef5c301357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 08:59:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e31d0b71540443669a531bf8794c0bee : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44d70945a054bd0a50009e5fc41f12b : SUCCESS in 5m 29s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c666ac49940143fa87389190b7a3914f : FAILURE in 1h 22m 36s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/bc1b68ffa0e44a9ea368786a335bf459 : SUCCESS in 47m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/0321ab988a9544a788ed20852b5298ff : SUCCESS in 1h 11m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f9b0210753cc4e1cbc81b3f74c170488 : FAILURE in 58m 55s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/78c48a0db2294382806d691d48776a56 : FAILURE in 1h 02m 18s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76a164392d3a471db4c7688d8d15d986 : FAILURE in 1h 42m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/1d1a78cda6084ffd9fc6d50e233c65e6 : SUCCESS in 30m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/51211783e9f443aaa259f05a75660471 : SUCCESS in 1h 21m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/baa03168b2054324a2c3c186bbf37347 : SUCCESS in 44m 01s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e364aabe0c534d26a4f16c2b8a3c423b : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f68bb511bd9b61ea3b58b2b411ea437d683e625c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-10-25 10:37:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/86/691186/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d3ee0d1/ : SUCCESS in 1h 33m 08s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/86/691186/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/da7258a/ : SUCCESS in 1h 37m 12s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b8717f8bbc0a6b2a859fa885b431b34bbdb99d85","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 22:30:15.000000000","message":"Topic changed from drop-py2 to drop-py27-support","accounts_in_message":[],"_revision_number":2},{"id":"6b8da0fd1a7429332e57eeaa5318fee72a185d4b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-12-11 01:43:16.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"97b038458f200ac7b98f3b9f457bba8efdee87b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 03:29:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23766a3e9ccd41eea5f444496e57f37f : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30adaeb13a4a45228b6a3495c5efd159 : SUCCESS in 5m 56s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/74d207079c2144d3af151ecf380a55fc : FAILURE in 1h 31m 50s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e032b38e95d64e9392eac74b893e331c : SUCCESS in 55m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/13f4a5a861e5498e9e45e7b1f4b78169 : FAILURE in 1h 10m 23s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/534ad97bbf0d466180870c67faeb9d24 : SUCCESS in 1h 00m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f328e06617584e34bfd0478a577b6948 : FAILURE in 1h 15m 16s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d6903d8a737436dacd72bdf1cc6ce48 : FAILURE in 1h 34m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2bdcb30cc24948c68cb391dfea71752b : SUCCESS in 21m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/09f88855e3ca4805b6650c4925431daa : SUCCESS in 1h 31m 53s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/70ad4c4cc2eb4724a801452a795bf8b9 : SUCCESS in 41m 43s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4ba07f07c6434028ad551e39e9925f28 : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":3},{"id":"e793c5edea7abdb3dcc9c095a4cd321c6b3fcadf","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-11 04:59:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/86/691186/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/7d6141c/ : SUCCESS in 1h 21m 53s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/86/691186/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/cbee316/ : SUCCESS in 1h 29m 22s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"ec936ee2369e040210e5ec0d6c16a92cea3f1bcd","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-12-13 18:28:19.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks Goutham, didn\u0027t see anything left to do, LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"8fca4a0fa212cc7dc66c7d32d7edd0a30d4bec9e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-01-06 16:08:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"812f390b8574e33278058476240294db17ef2989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:09:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95d7fdeeecd6122dc1505e39706d56237121180e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:28:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cfd9ba36c76741168795de20a65de4bb : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c674a53eeeb94d2e985850b460eba7f8 : SUCCESS in 7m 40s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1a1ab309018d43858e0df0e2b9c94367 : SUCCESS in 56m 47s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/92a2ba5584944f23a8078a86d256eb15 : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ee50ef344ad7036a993527f438b38f9aeff92061","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:28:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eff4f9b87f0d36e0cfa4b1d861125f456f341af9","revisions":{"349d3a72f860146e564aa6c49faec27ae2f94fd7":{"kind":"REWORK","_number":1,"created":"2019-10-25 05:22:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/691186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/691186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-25 05:22:28.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-25 05:22:28.000000000","tz":-420},"subject":"Drop python2.7 support","message":"Drop python2.7 support\n\nAs per the official python support timeline\nset forth by the OpenStack TC [1], OpenStack\nTrain is the last release that will support\npython2.7. manila-tempest-plugin is a \u0027branchless\u0027\ntempest plugin. However, after this change,\nusers may no longer expect to install this package\nwith python2.7.\n\nDepends-On: https://review.opendev.org/#/c/688997/\n[1] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html\n\nChange-Id: I8c10d169302c6f1c41e238ae9f787455a2b15a4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/349d3a72f860146e564aa6c49faec27ae2f94fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/349d3a72f860146e564aa6c49faec27ae2f94fd7"}]},"branch":"refs/heads/master"},"9fa25734232db136754b5845c6b6a00484562ddd":{"kind":"REWORK","_number":2,"created":"2019-10-25 06:23:06.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/691186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/691186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/2"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-25 05:22:28.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-25 06:23:04.000000000","tz":0},"subject":"Drop python2.7 support","message":"Drop python2.7 support\n\nAs per the official python support timeline\nset forth by the OpenStack TC [1], OpenStack\nTrain is the last release that will support\npython2.7. manila-tempest-plugin is a \u0027branchless\u0027\ntempest plugin. However, after this change,\nusers may no longer expect to install this package\nwith python2.7.\n\nDepends-On: https://review.opendev.org/#/c/688997/\n[1] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html\n\nChange-Id: I8c10d169302c6f1c41e238ae9f787455a2b15a4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fa25734232db136754b5845c6b6a00484562ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fa25734232db136754b5845c6b6a00484562ddd"}]},"branch":"refs/heads/master"},"eff4f9b87f0d36e0cfa4b1d861125f456f341af9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-11 01:43:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/691186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/691186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/691186/3"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-25 05:22:28.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-12-11 01:42:55.000000000","tz":0},"subject":"[ussuri][goal] Drop python2.7 support","message":"[ussuri][goal] Drop python2.7 support\n\nAs per the official python support timeline\nset forth by the OpenStack TC [1][2], OpenStack\nTrain is the last release that will support\npython2.7. manila-tempest-plugin is a \u0027branchless\u0027\ntempest plugin. However, after this change,\nusers may no longer expect to install this package\nwith python2.7.\n\nDepends-On: https://review.opendev.org/#/c/688997/\n[1] https://governance.openstack.org/tc/resolutions/20180529-python2-deprecation-timeline.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I8c10d169302c6f1c41e238ae9f787455a2b15a4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eff4f9b87f0d36e0cfa4b1d861125f456f341af9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eff4f9b87f0d36e0cfa4b1d861125f456f341af9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
