)]}'
{"id":"openstack%2Fmanila-tempest-plugin~691511","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I99b47b4278be1881b4012f1b695621bb42b0e532","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I99b47b4278be1881b4012f1b695621bb42b0e532","subject":"remove ssh bug workaround","status":"NEW","created":"2019-10-27 10:22:56.000000000","updated":"2019-10-28 00:05:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d7910f5884128c3a8398def03704afb10482f3b","_number":691511,"virtual_id_number":691511,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-27 12:20:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":-1,"date":"2019-10-28 00:05:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-27 12:20:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3dc88a98711352a5b589cb70019d327dd60dd49b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-27 10:22:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70048620e6461352ea3adc4a898fe0cbd788bf58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 12:20:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/137a1b101cd64c0488facff4757be1cb : SUCCESS in 3m 57s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/fdc1de8b1dec4ae590fa10b7d68f1aec : FAILURE in 1h 37m 12s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/99cb1ca5abad4d78b475082ff6ed11b6 : SUCCESS in 1h 08m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9385b3ef882e448d828b14be9fd44b6e : SUCCESS in 1h 11m 33s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/119f32d85feb4471abd7a6508c391581 : SUCCESS in 1h 17m 32s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/12d47142982148c0adcbad9d347efd99 : SUCCESS in 1h 10m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4950d9c4c1d4158a4351adcc936f519 : FAILURE in 1h 53m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ee69d2734f3f4b8dbc2c5d6032da4781 : FAILURE in 34m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/59a38139849b4851a71ec06a5c936ac7 : FAILURE in 1h 48m 12s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ccd1541709ec44c6bd16359907b48722 : SUCCESS in 50m 46s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5ced55bb7b164b7696174c7ece692157 : FAILURE in 1h 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b6b211ff4b66216d040492262b1030432a617e04","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-27 12:30:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"91a6906d29d3932299b46b63f91fd35b38899fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 14:14:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563e046bdeec4eee8c8ab9cab7454196 : SUCCESS in 3m 32s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d1918429dcdf49f8b9b2d6d62b5114ca : FAILURE in 1h 27m 45s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f0671d0867344b57aba7b085f6eb397d : SUCCESS in 43m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a779bee56cd84aeaaaceedb107cb1510 : SUCCESS in 1h 07m 10s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d0225f4511ea441ba1ee922ad790e07c : SUCCESS in 1h 08m 09s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e5c6256a52034358b2def8ecf7b1777d : FAILURE in 1h 14m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84f8133eff3245beadfb98127ad573fb : SUCCESS in 1h 42m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a622b54e484a41869c6c2c03dd8d92c3 : FAILURE in 35m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/098bbcafc88e42cfb3c2f6e8f447ac6f : FAILURE in 1h 40m 05s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/89e166cbbd144047b7fef0a00894737c : SUCCESS in 42m 02s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/47d5922bd4024026bc69794ca34b176d : FAILURE in 1h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0d7910f5884128c3a8398def03704afb10482f3b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-28 00:05:01.000000000","message":"Patch Set 1: Workflow-1\n\nIPv6 tests fail when the workaround is removed, need to investigate more.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16bca37c43a9769ccf6031b1af54898239f66d4b","revisions":{"16bca37c43a9769ccf6031b1af54898239f66d4b":{"kind":"REWORK","_number":1,"created":"2019-10-27 10:22:56.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/11/691511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/11/691511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/691511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/691511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/691511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/11/691511/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-27 10:19:45.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-27 10:19:51.000000000","tz":-240},"subject":"remove ssh bug workaround","message":"remove ssh bug workaround\n\nA workaround for bug 1465682 was introduced to our code\nbase in mitaka.  The bug has long been fixed, so remove\nthe workaround.\n\nChange-Id: I99b47b4278be1881b4012f1b695621bb42b0e532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16bca37c43a9769ccf6031b1af54898239f66d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/16bca37c43a9769ccf6031b1af54898239f66d4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
