)]}'
{"id":"openstack%2Fmanila-tempest-plugin~693592","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I51fa672ca5ede786c9d7b79dc8b4bf9d7c853ced","project":"openstack/manila-tempest-plugin","branch":"master","topic":"debug_plugin","hashtags":[],"change_id":"I51fa672ca5ede786c9d7b79dc8b4bf9d7c853ced","subject":"Testing manila plugin on master branch","status":"ABANDONED","created":"2019-11-08 21:44:19.000000000","updated":"2020-04-18 00:48:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3a5713fcec63eb22481d8f019679f6a4b1f266e","_number":693592,"virtual_id_number":693592,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-08 23:20:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":-1,"date":"2019-11-08 21:49:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 23:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9308fa4171419b7226a224a37763c2c9928ccd7","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-11-08 21:44:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"997883ec4379e909e21ee0bbb53e49a772075d47","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-11-08 21:49:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"41c1243b6953ab2ac6801f4089b96f8c73779bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 23:20:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27aafda5c594c25b083dbe50c59fb66 : SUCCESS in 2m 56s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0b6e7b82ee1e4889bb95d24681d0b7fe : SUCCESS in 1h 34m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/2ad9c95ee37b410191c2ce610d54d449 : SUCCESS in 52m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/ccf8c458358c4fb19fbbab99f5c68d58 : SUCCESS in 1h 20m 28s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/970198fbbc654de0916b888d1d81f2cf : SUCCESS in 57m 31s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e8f39687dfd84d978bd0a0454486b78c : FAILURE in 1h 00m 22s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4ccab2d1bce4c01aa304fa88880c471 : FAILURE in 55m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/3858340b7e3f47b8b6b4284b7392abe1 : SUCCESS in 44m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/84820ee3b84d4af29918d3f65036889f : FAILURE in 1h 17m 20s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7e5d76bded2f4b26b9754276aeba8d59 : FAILURE in 57m 28s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/2924090f202b48868c10a4740a4830a6 : FAILURE in 1h 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d3d8aa0a54eaeba09f63572fa59b079799bbfa93","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-18 00:48:23.000000000","message":"Abandoned\n\nDNM patch, can restore when necessary","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"530395b54f2a4302e393b648af6ed263b567d172","revisions":{"530395b54f2a4302e393b648af6ed263b567d172":{"kind":"REWORK","_number":1,"created":"2019-11-08 21:44:19.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/92/693592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/693592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/693592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/693592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/693592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/693592/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-11-08 21:43:42.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2019-11-08 21:44:03.000000000","tz":-180},"subject":"Testing manila plugin on master branch","message":"Testing manila plugin on master branch\n\nChange-Id: I51fa672ca5ede786c9d7b79dc8b4bf9d7c853ced\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/530395b54f2a4302e393b648af6ed263b567d172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/530395b54f2a4302e393b648af6ed263b567d172"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
