)]}'
{"id":"openstack%2Fmanila-tempest-plugin~696138","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I34832d395afc0e009a19ecaec26a3f586f45db38","project":"openstack/manila-tempest-plugin","branch":"master","topic":"limit-share-replicas-per-share","hashtags":[],"change_id":"I34832d395afc0e009a19ecaec26a3f586f45db38","subject":"[WIP] Add new test for share types replica limit","status":"NEW","created":"2019-11-26 17:16:13.000000000","updated":"2020-01-29 13:10:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3fb2e68d232d15de84d88eae4946437e45bb3f1c","_number":696138,"virtual_id_number":696138,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-09 11:38:31.000000000","_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-26 19:38:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 19:38:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99b7df1a559c8951fb9455515c3b2a30adbc67bc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-11-26 17:16:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b181fc2da478452b3256ea4b77a03f390536bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:38:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6cf3213be74275ba74baf49785690b : SUCCESS in 3m 23s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/7ce19c911aba428aafc998be1e7fb469 : FAILURE in 7m 09s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8aef5f3058b342a5bb1c47c7dc78a3db : SUCCESS in 58m 24s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7bd29c1dec948e68ab3827a3c05c858 : FAILURE in 1h 05m 16s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/284e01b1659a4cc3943b939f72cb6c93 : SUCCESS in 55m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/63d16d9c93fc4f91a7b1c0aeb90cb562 : SUCCESS in 1h 00m 48s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82e0437fb3cd427e989871f91e0a4ff5 : FAILURE in 1h 38m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f276b035888c492f9a6a6433fed7a6cf : SUCCESS in 25m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b782608664024e97ad493412b9292947 : SUCCESS in 1h 28m 14s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/28a59de8245c4c60bfd4915b54ce5b14 : FAILURE in 8m 18s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/dca073c82ddf429fbe4ff72fd399b539 : FAILURE in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"75b5f52449818e61ed165fe6be9519499185fe35","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-09 11:38:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7481d64e8d0cc3dcc1955adb2ac3718f504db3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:22:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6756eb074f543dcad5e8827283564e1 : SUCCESS in 4m 19s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b61cac71f3fb48749fec24d493f0f266 : FAILURE in 26m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/14ffb84d31cb4de199c67285dfd329ee : FAILURE in 32m 25s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/5aeb5cf0b7634c919566ac74266d0258 : FAILURE in 30m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5c3f4a72012143d6a0630c629471e298 : FAILURE in 35m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/85111b8bdc8049748c8a2fa3290f791d : FAILURE in 36m 45s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9992728b731e401bbe470088399a7917 : FAILURE in 36m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d73824c0c0b048c7a6918052f23e3de3 : FAILURE in 26m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/23428a78a7234d998c0182047bda8032 : FAILURE in 31m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f47fbd0015a94a8d8f9c0bd5b2283a3a : FAILURE in 25m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/318ad1e31bc343d0a4b2439f10f4a327 : FAILURE in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3fb2e68d232d15de84d88eae4946437e45bb3f1c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-01-29 13:10:54.000000000","message":"Topic set to limit-share-replicas-per-share","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f330a85e9e0108616eb91a5732ebb8c46139461","revisions":{"7f330a85e9e0108616eb91a5732ebb8c46139461":{"kind":"REWORK","_number":1,"created":"2019-11-26 17:16:13.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/38/696138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/696138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/696138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/696138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/696138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/696138/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-11-21 18:02:05.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2019-11-26 17:13:59.000000000","tz":-120},"subject":"[WIP] Add new test for share types replica limit","message":"[WIP] Add new test for share types replica limit\n\nThis patch adds a new test in order to validate the provided fix\nfor creating unlimited share replicas.\n\nChange-Id: I34832d395afc0e009a19ecaec26a3f586f45db38\nDepends-On: I4bd02918ff08711dfadbd129b70b79d843ef7634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f330a85e9e0108616eb91a5732ebb8c46139461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f330a85e9e0108616eb91a5732ebb8c46139461"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
