)]}'
{"id":"openstack%2Fmanila-tempest-plugin~700833","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic685d1b9574daf7e6c9b90d5636f1b614a70b0da","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1858024","hashtags":[],"change_id":"Ic685d1b9574daf7e6c9b90d5636f1b614a70b0da","subject":"Use a single keypair for multiple instances","status":"MERGED","created":"2020-01-01 13:58:36.000000000","updated":"2020-01-21 02:56:26.000000000","submitted":"2020-01-21 02:56:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700833-1579575386375-641ca554","meta_rev_id":"63aa4dcf63fd3317a9763bbf2fb164b4f8d280ae","_number":700833,"virtual_id_number":700833,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-01-09 12:45:28.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-21 02:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"date":"2020-01-09 09:40:19.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-01-21 01:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-01-09 16:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":2,"date":"2020-01-09 13:06:49.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-09 19:27:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-01-21 01:53:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 12:45:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-09 13:06:49.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-01-09 16:56:24.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2020-01-09 19:27:38.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-01-21 01:53:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-01-21 02:56:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb94ace0cbe11ac57d1ca333c8b0a6927d0f169f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-01 13:58:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83e78ab8291cd496a69f5e91386e97ac2f48d079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 15:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c646661095846cb8eefe93d77a0eb33 : SUCCESS in 3m 15s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/558b85cde7c0463c88019362b437d60f : FAILURE in 1h 20m 16s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/16d1a3a588944a97ba7368ac0ca8b098 : SUCCESS in 49m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/32904cee8c594650bbdd9973d43ab918 : SUCCESS in 1h 09m 24s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/763f506df0044ad597feadb93cc96c23 : FAILURE in 1h 00m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2ec952483a854216baebbd9b6cbde3b2 : SUCCESS in 57m 33s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e55dd3ede644aa3a638270171b2faaa : FAILURE in 1h 32m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f89bb27e1aa740bc9ae7217193eff07d : SUCCESS in 23m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/190d1113834647789fa01616aad38e55 : SUCCESS in 1h 13m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/72afd124f9a44f29a05bcb42c8a9b642 : SUCCESS in 39m 55s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/373f0b7c1532457380c2c47632092728 : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4a27b249bac0b46ff198b02ab345c1c48d0e542b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-01-01 22:00:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"010be9b0cc8c21502f0a2122ca7c316292b8b74a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 23:30:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0755f9a169f14e4b9ef467352cbcc239 : SUCCESS in 3m 22s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b48de6a113b8420cb880dd46217da63a : SUCCESS in 1h 26m 10s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/694c16870b6b436ba63801542894ddb0 : SUCCESS in 47m 09s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/1c3a571f6403490586920b097b275352 : SUCCESS in 53m 25s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/a90c78ef160c46b58e16c8293311208c : SUCCESS in 42m 16s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/b1ee736160c0462fb45c7ce782e8ede1 : SUCCESS in 53m 50s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6eb0a2aac3e47598ecbbb20d1ee8809 : FAILURE in 1h 29m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e4985c7621b147a9bd3620bc4ad07c42 : SUCCESS in 21m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a195d0bb1b404ff49d23aa8cda0769f5 : SUCCESS in 1h 20m 48s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/aeaf1fbd265b4d8ea42330127b28ff12 : FAILURE in 35m 28s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/f147fd92cd514e35bed486ec92dc2e3a : RETRY_LIMIT in 10m 10s","accounts_in_message":[],"_revision_number":1},{"id":"98cd380419ede64b9941e37e19d6355f56b54c28","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-02 06:11:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef7e8c9e788360ac4e1424f79a316215f8fb9280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 08:16:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6788d36ab141420fbe822c705ec9ec0f : SUCCESS in 5m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/90b80d904a5649719032ae3b3492ea27 : SUCCESS in 1h 34m 17s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/57d15882eee7431f8d21bad0cc53321e : SUCCESS in 1h 13m 23s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/fa33ac2053f44fceb46b980dd54b95ea : SUCCESS in 1h 17m 40s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3e227bd9ffe3467daa3a1d365cdb7f02 : FAILURE in 1h 14m 37s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/54d194587b6844db8baf72607ec345a6 : FAILURE in 1h 07m 37s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1094ded6ab24a65b0e84f603c1f8f84 : FAILURE in 1h 54m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/15231e3ac5fe41dc9813132af4c33232 : SUCCESS in 43m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3edb25f2cc7943e7bbc9615f3c8bba6a : SUCCESS in 1h 12m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d362b366a8d4457b9d63d0fd145e6fa4 : SUCCESS in 38m 37s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/acb6c76d34624189ab65dc762a9bd6ab : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1},{"id":"516eab714c93cbca60c70b0cc8451a01254d55ac","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-01-06 10:35:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23a3a7dc30f998981dbdf01410ff8056ac654b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:32:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1055693c90d4f8794541123c4f18125 : SUCCESS in 3m 33s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/658da813941943c1bc220f81253dacab : SUCCESS in 1h 36m 48s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/fccf38d26b054ce1ae52392d9073324f : SUCCESS in 1h 07m 38s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d7803d559f1a40ac8269d77b27641ed4 : SUCCESS in 1h 09m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b0d5402a3a9b49d78f159ed99005ea23 : SUCCESS in 54m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/686476a0f0b843cfa6a7b593d6a3298b : FAILURE in 1h 01m 41s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33b18e401ee64b46a3ef2d898c44001d : FAILURE in 1h 54m 48s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/dc098592f54c4b3b9ca004876b461a08 : SUCCESS in 28m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8ee6395b92884f95affcefdb802f40f0 : SUCCESS in 1h 16m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/25570a909af04c9ba487795f34ab612a : SUCCESS in 41m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c2adbb3da7c541328ea5077cd63e924b : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ead48bc8eb75c5d0c99566ba1d4b88819dc69dd3","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-01-08 16:37:51.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks!\n\nThis looks reasonable but we need a bug-fix releasenote.\n\nAlso, the scenario job is timing out.  It\u0027s also timing out in a baseline run [1].  If we can see tests passing here with your fix that fail in the baseline run please point to their logs so that reviewers can see that something has been fixed and that we should merge even though there may be other issues that need to be addressed separately.\n\n[1] https://review.opendev.org/#/c/700840/","accounts_in_message":[],"_revision_number":1},{"id":"3f3606c08dcfccb9f97f4b057b5bdc130a1c5418","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-09 09:06:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1aa88f0b2c6cd77d02e6d2d9b943c05a50f8654f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-09 09:40:19.000000000","message":"Patch Set 2:\n\n\u003e Thanks!\n \u003e \n \u003e This looks reasonable but we need a bug-fix releasenote.\nDone\n \u003e \n \u003e Also, the scenario job is timing out.  It\u0027s also timing out in a\n \u003e baseline run [1].  If we can see tests passing here with your fix\n \u003e that fail in the baseline run please point to their logs so that\n \u003e reviewers can see that something has been fixed and that we should\n \u003e merge even though there may be other issues that need to be\n \u003e addressed separately.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/700840/\n\nThere is no ssh time out exception in [1].\nIf you dive into the logs, you can see that in some cases the login procedure \nis done by password and in other cases by key authentication.\nIf \"image_password\" parameter in tempest.conf will be remove, the issue I \ndescribed will be happened.","accounts_in_message":[],"_revision_number":2},{"id":"ddeb27a27092518e1595dd898e9d11140514df97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:08:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a0f7910e474cf8b3a0508d23f007df : SUCCESS in 4m 45s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/90af671239804e5da23230109918a45c : SUCCESS in 1h 25m 36s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/741636a11f754c5c9ff6da9baf5b8100 : SUCCESS in 52m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/db1f9727ca8140bb9f5e41be84259c28 : FAILURE in 1h 11m 45s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/670df269d6df4b7899d24f0e860dfceb : SUCCESS in 51m 36s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/36d9fc4f314849edaa4d8a9277c7da56 : FAILURE in 50m 43s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/416646ad299447a5b70bbdea1fe1f2f2 : FAILURE in 1h 54m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f14167f13be24bf08d88c565c6d9dd82 : SUCCESS in 28m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/36b3216d7f434f39abbf99f9aa1b4c84 : SUCCESS in 1h 20m 41s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/68c41408de9c4884b51097abd6586b9f : SUCCESS in 45m 08s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/1bc6a8ea4ff14316a0d2d2fca5d1da6e : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":2},{"id":"28febd906f2389f5b20b70af0416ec8dbae99413","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-01-09 12:45:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/33/700833/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2feeb74/ : SUCCESS in 1h 27m 59s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/33/700833/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2197666/ : SUCCESS in 1h 37m 00s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"6d5d9122f620546d58621f63272a8bcfa0c21673","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-01-09 13:06:49.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, Liron!","accounts_in_message":[],"_revision_number":2},{"id":"0df56bf82f3bd71249784bffb9d12d0811542e66","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2020-01-09 16:56:24.000000000","message":"Patch Set 2: Code-Review+1\n\nthis change is good and is green to merge, however, other issues are blocking the CI tests so we cannot validate it at this moment","accounts_in_message":[],"_revision_number":2},{"id":"c6381bc7c15d8615ce87c7e281685aca81939228","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2020-01-09 16:57:35.000000000","message":"Patch Set 2:\n\nAlso, one thing that crossed my mind, is that if our tests are unable to ssh using password, then perhaps it could use an improvement to be able to. The generic driver can ssh with password authentication.","accounts_in_message":[],"_revision_number":2},{"id":"5f53e514d440415eecacd82f2562890421e8dd62","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-01-09 19:27:38.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Liron, looks good to me!","accounts_in_message":[],"_revision_number":2},{"id":"9ebb71fb6cfbb6217f24f42d948f21f9351fd7fd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-01-21 01:53:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\n\u003e this change is good and is green to merge, however, other issues are blocking the CI tests so we cannot validate it at this moment\n\nGanso: I see that the HTML version of the testr results isn\u0027t getting rendered properly for any of the jobs - but the rest of the logs seem fine, with some intermittent failures. The voting job\u0027s running this test for instance - what am I missing?\n\nApproving this, let\u0027s circle back in case it breaks something...","accounts_in_message":[],"_revision_number":2},{"id":"a497be41b2880136f71af807ddc91d49ea79549e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 01:53:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0eb05bbfc6bb9a20c00be3de0d27543ce459f84d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 02:56:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f225542e614217a3af307075615a1e : SUCCESS in 3m 27s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3a0e80e27d224511bee2d2de9157d3d4 : SUCCESS in 40m 17s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/21ba8e8def634dada2317d150af28506 : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4a1f205bdb6fa434f9b53862cbdc10852197a281","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 02:56:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cd6c999d4985e10be2566c631312720efe8413c","revisions":{"edb86b76c18544382b3af0d8090ba8d279f90feb":{"kind":"REWORK","_number":1,"created":"2020-01-01 13:58:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/33/700833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/33/700833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/1"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-01 13:56:52.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-01 13:56:52.000000000","tz":120},"subject":"Use a single keypair for multiple instances","message":"Use a single keypair for multiple instances\n\nWhen I ran manila scenario tests, some tests failed due to\nssh key authentication. I noticed that tests with single\ninstance pass, however, tests with more than one instance\n(e.g. \"test_read_write_two_vms\" test) fail.\nI figured out that there is a mismatch between both instance keys.\nI tried to use a single key for both instances and the test passed.\n\nChange-Id: Ic685d1b9574daf7e6c9b90d5636f1b614a70b0da\nRelated-bug: #1858024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/edb86b76c18544382b3af0d8090ba8d279f90feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/edb86b76c18544382b3af0d8090ba8d279f90feb"}]},"branch":"refs/heads/master"},"0cd6c999d4985e10be2566c631312720efe8413c":{"kind":"REWORK","_number":2,"created":"2020-01-09 09:06:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/33/700833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/33/700833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/700833/2"}}},"commit":{"parents":[{"commit":"4d7b4949906a8935f2fe5a404c234e392b6e4cde","subject":"Merge \"Skip subnets test when version 2.51 is not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d7b4949906a8935f2fe5a404c234e392b6e4cde"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-01 13:56:52.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-09 09:02:53.000000000","tz":120},"subject":"Use a single keypair for multiple instances","message":"Use a single keypair for multiple instances\n\nWhen I ran manila scenario tests, some tests failed due to\nssh key authentication. I noticed that tests with single\ninstance pass, however, tests with more than one instance\n(e.g. \"test_read_write_two_vms\" test) fail.\nI figured out that every instance initialization creates a\nunique keypair.\nAn unnecessary resource duplication, when a single user boots\nmore than one instance. Depending on test flow this may also\ncause keypair mismatch.\n\nChange-Id: Ic685d1b9574daf7e6c9b90d5636f1b614a70b0da\nRelated-bug: #1858024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cd6c999d4985e10be2566c631312720efe8413c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cd6c999d4985e10be2566c631312720efe8413c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
