)]}'
{"id":"openstack%2Fmanila-tempest-plugin~704246","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia2a0fefc34085e21e521acb7ccfdeecef7fda186","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ia2a0fefc34085e21e521acb7ccfdeecef7fda186","subject":"Fix export location used for extend/shrink tests","status":"MERGED","created":"2020-01-26 12:00:31.000000000","updated":"2020-04-23 00:32:02.000000000","submitted":"2020-04-23 00:32:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704246-1587601922389-1d1e633c","meta_rev_id":"50d3d8ef74e6549fcfd66dbc6238059865c92ea0","_number":704246,"virtual_id_number":704246,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 00:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-04-22 17:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-04-22 16:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-04-22 21:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"date":"2020-04-21 14:18:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 13:19:18.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-02-05 17:59:10.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-13 15:22:40.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-02-13 15:22:43.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-04-21 14:18:24.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2020-04-22 17:13:42.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-22 21:05:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-23 00:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a165325f4b2b7a71efcac265070b8c713d85e175","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-01-26 12:00:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77af2f44a4a4161132e8121237f9c69830fb94f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 13:41:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac565fef0034216a31cf128374ef335 : SUCCESS in 2m 59s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f9665ed374374d3e83f7761ae63b98bf : FAILURE in 1h 11m 33s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/acad5088346545e9ae396543187119b1 : SUCCESS in 55m 02s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d090935bfffa4633827bab0fe7a0b93d : FAILURE in 1h 09m 25s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/dfe380def6a0464f9f1b3e3cd778db66 : FAILURE in 1h 01m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/641832db3b1f45258d07117a9eb57c67 : SUCCESS in 59m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2446d6fa0d014c02bbef34ca8283933f : FAILURE in 1h 35m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/58306cb8ddcf42d98c27e620e3431075 : SUCCESS in 28m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0dcb598210434d25a37cec313abde3ae : FAILURE in 1h 20m 18s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c238a440395c4245858f430b1305c366 : SUCCESS in 41m 30s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/3480b181718c4997973a58e69eadc704 : FAILURE in 1h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"615c9305ec6150bc2d1c2e98a571547b53131883","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-01-27 13:38:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5855d676e77205a04a7cb667a117af1196ba8c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:40:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab7d2b835f84985838c5d5802fd26ae : SUCCESS in 3m 00s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/505c5b8fc696485db766f57f6d64c82f : SUCCESS in 1h 30m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3c29b81d13ea4a638b7c2c43e6e100a1 : SUCCESS in 45m 23s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/717a53e94ac149aea4d5b1ea3fff1464 : FAILURE in 1h 27m 59s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b2b5ef9e61c2432a9857980e2b86400a : SUCCESS in 1h 41m 55s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/a39b0c62004049439bbdccaae1f0c9d1 : SUCCESS in 1h 17m 56s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a699ca76b8104810906c65b8abc9e600 : FAILURE in 1h 58m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6bee41b2e0b8499398a2dd7295305204 : SUCCESS in 34m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/65f5bc47e46a4dd5a1bf5b78110de2b4 : FAILURE in 1h 55m 59s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/32815369cc3744deafb16c5e47b1327f : SUCCESS in 1h 01m 23s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/bb19c7376af64b70a676bad932b2f1c6 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c60c5113247c7028685b845cdf13082362fa46e3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-01-27 18:07:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/46/704246/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c6b7396/ : SUCCESS in 1h 23m 48s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/46/704246/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1a2fde3/ : SUCCESS in 2h 26m 31s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cbb51ec720e1df9fcfab247da0fbb6cd070848f2","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-02-05 11:50:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2987f516ada7c153239a771163d2d35e9899a68d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:38:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3240376c850a41688c3cbffc9dc7cf39 : SUCCESS in 4m 49s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/737fe2b97231482b8025cb0227382cce : SUCCESS in 1h 24m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/077d4e65fe0b4331a420f0bc9712e3ff : SUCCESS in 1h 09m 46s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e9658eea68a4421483ce788e453779fe : SUCCESS in 1h 25m 42s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ecb6a8137de04cc5b252b72879954e35 : FAILURE in 53m 29s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/042a8f5cc2d54da2914acc35b280fd56 : SUCCESS in 1h 16m 26s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec5e17b5be744d74a6f41c7e5af55ed0 : FAILURE in 1h 21m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/612980c50af948918133e29ca6704c85 : SUCCESS in 26m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b90f032534b048a58e512aa912654c82 : SUCCESS in 1h 40m 34s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/651053cc90124f31aaeb46a9345c0976 : SUCCESS in 40m 43s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/7511f7aeae6b464abcdd8e963744446e : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"00f41b1a3f08171fa17f826d751aa23da892e9e8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-05 17:59:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/46/704246/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/97cdfca/ : SUCCESS in 1h 27m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/46/704246/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4bdac8b/ : FAILURE in 2h 44m 49s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8a32ef93ff537102b1ec608f45ca2b0e5b690083","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-02-14 03:10:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nThanks Tom, comments inline..","accounts_in_message":[],"_revision_number":2},{"id":"e96bf1de5d36e6f304d8791f24f871902ff35354","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-18 00:44:32.000000000","message":"Patch Set 2:\n\nping for update","accounts_in_message":[],"_revision_number":2},{"id":"a71354ba197d4b2c9ab2e05814647258c9bbd21f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-21 10:25:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91f5cef412ef8121b618101f6e6b29eeac63633f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 11:58:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b77bb30a6072458897940ec285300cc5 : SUCCESS in 3m 48s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/eaebf3af4936452b8b1d83efd5e313a5 : SUCCESS in 1h 27m 02s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4d2a52df996344548760374be882f758 : SUCCESS in 43m 16s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c2af738ca1134b029cba7380e44b47a4 : SUCCESS in 49m 10s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/16c62a4b106846c486926d05d1360205 : SUCCESS in 56m 36s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/1d67b2c1db4b479cbb678208d2dc6848 : FAILURE in 46m 37s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e244b2e84c994cda91af27d34bce0dfa : SUCCESS in 40m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0d0be3c1421249e99ac24ea134f63c47 : SUCCESS in 29m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2c65582534594f8f854b880db6ded6be : FAILURE in 1h 17m 03s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c4cc97db10df486dba86171cee518fbc : SUCCESS in 47m 25s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/903e412e1e9e47a88ba225e1cfdf2233 : FAILURE in 1h 03m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c90f7468c2898e9e1d2e27a3ec420b5ad891ecf4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-21 12:28:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad42c9f66476ef50220950896177707695e2ecc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 14:05:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b0293c57c44077a207e0993047eb13 : SUCCESS in 4m 43s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/82122b693dd8458e9acad8dc272a1ddf : SUCCESS in 1h 30m 05s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/33dca3cde67d40ba9a6c3059b2ba3202 : SUCCESS in 1h 02m 52s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/0052dc3e96a3406eb2b32bff04e19c3e : SUCCESS in 39m 41s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/243f59dc5ca74ce8afa409acce1b2921 : FAILURE in 1h 18m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/1db710738a2e40d4807eb95b14b623a2 : FAILURE in 45m 10s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a01f655be2e46a08dce8504652fffaf : SUCCESS in 40m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/8d36970988c74f198b6eb86c84be9a91 : SUCCESS in 26m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/991a849470064630bdea7d6a7ae2d90e : SUCCESS in 1h 16m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0ec74870ffbd426187b2bdd993be41ca : SUCCESS in 45m 04s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/dc1cf9b0fcd8474d8997decf84a685bc : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":4},{"id":"29c11dd04e1c74e66fdc310f651c95c334a2f0de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-22 16:50:20.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks Tom!","accounts_in_message":[],"_revision_number":4},{"id":"6fcbe10228ce88c046a6e922d9a2fb091774fe0a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-22 17:13:42.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM. Thanks, Tom.","accounts_in_message":[],"_revision_number":4},{"id":"588bafa8bf06d0004fe3e5c161a187e26c4a78b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-22 21:05:53.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"838be31bd3492a53dbc5d9177831484c771553d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 21:06:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dd958d316e7b3fa07530ce93b6baa1c7df56d84e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 00:32:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf592aa3250e44adbac0d3c4c2c4a946 : SUCCESS in 3m 16s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7546a79490c54e439165b4181a01d837 : SUCCESS in 46m 13s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/98963234f11a48498ab66cb7ea818eda : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":4},{"id":"50d3d8ef74e6549fcfd66dbc6238059865c92ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 00:32:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"645e4eb092a6cd7ba946ec949cec684112629eed","revisions":{"7d2dbb30cca4fd967983816f8b8555b6f8ebc9b5":{"kind":"REWORK","_number":1,"created":"2020-01-26 12:00:31.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/46/704246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/46/704246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/1"}}},"commit":{"parents":[{"commit":"ca1340cc44d1dec1ed3351d9ea650a179b5d695a","subject":"Merge \"Use a single keypair for multiple instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca1340cc44d1dec1ed3351d9ea650a179b5d695a"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 21:26:41.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-01-26 11:58:23.000000000","tz":-300},"subject":"Fix export location used for extend/shrink tests","message":"Fix export location used for extend/shrink tests\n\nInstead of using the first export location, use the first export\nlocation matching the address family (IPv4 or IPv6) to be used\nfor the mount.\n\nCloses-Bug: #1848278\nChange-Id: Ia2a0fefc34085e21e521acb7ccfdeecef7fda186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7d2dbb30cca4fd967983816f8b8555b6f8ebc9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7d2dbb30cca4fd967983816f8b8555b6f8ebc9b5"}]},"branch":"refs/heads/master"},"cb8113de58cd06cbd6325f3cf4e583103d4e3ad3":{"kind":"REWORK","_number":2,"created":"2020-01-27 13:38:09.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/46/704246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/46/704246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/2"}}},"commit":{"parents":[{"commit":"ca1340cc44d1dec1ed3351d9ea650a179b5d695a","subject":"Merge \"Use a single keypair for multiple instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca1340cc44d1dec1ed3351d9ea650a179b5d695a"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 21:26:41.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-01-27 13:37:57.000000000","tz":-300},"subject":"Fix export location used for extend/shrink tests","message":"Fix export location used for extend/shrink tests\n\nInstead of using the first export location, use the first export\nlocation matching the address family (IPv4 or IPv6) to be used\nfor the mount.\n\nCloses-Bug: #1848278\nChange-Id: Ia2a0fefc34085e21e521acb7ccfdeecef7fda186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cb8113de58cd06cbd6325f3cf4e583103d4e3ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cb8113de58cd06cbd6325f3cf4e583103d4e3ad3"}]},"branch":"refs/heads/master"},"8f0b3eb79fe7553b6cb28f91f01d2c40f7b74dd1":{"kind":"REWORK","_number":3,"created":"2020-04-21 10:25:45.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/46/704246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/46/704246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/3"}}},"commit":{"parents":[{"commit":"a096815be929abd4b91290dff40e90315d2395ce","subject":"Merge \"Refactoring skip conditions process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a096815be929abd4b91290dff40e90315d2395ce"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 21:26:41.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-04-21 10:24:49.000000000","tz":-240},"subject":"Fix export location used for extend/shrink tests","message":"Fix export location used for extend/shrink tests\n\nInstead of using the first export location, use the first export\nlocation matching the address family (IPv4 or IPv6) to be used\nfor the mount.\n\nCloses-Bug: #1848278\nChange-Id: Ia2a0fefc34085e21e521acb7ccfdeecef7fda186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f0b3eb79fe7553b6cb28f91f01d2c40f7b74dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f0b3eb79fe7553b6cb28f91f01d2c40f7b74dd1"}]},"branch":"refs/heads/master"},"645e4eb092a6cd7ba946ec949cec684112629eed":{"kind":"REWORK","_number":4,"created":"2020-04-21 12:28:47.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/46/704246/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/46/704246/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/46/704246/4"}}},"commit":{"parents":[{"commit":"a096815be929abd4b91290dff40e90315d2395ce","subject":"Merge \"Refactoring skip conditions process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a096815be929abd4b91290dff40e90315d2395ce"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-10-15 21:26:41.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-04-21 12:28:39.000000000","tz":-240},"subject":"Fix export location used for extend/shrink tests","message":"Fix export location used for extend/shrink tests\n\nInstead of using the first export location, use the first export\nlocation matching the address family (IPv4 or IPv6) to be used\nfor the mount.\n\nCloses-Bug: #1848278\nChange-Id: Ia2a0fefc34085e21e521acb7ccfdeecef7fda186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/645e4eb092a6cd7ba946ec949cec684112629eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/645e4eb092a6cd7ba946ec949cec684112629eed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
