)]}'
{"id":"openstack%2Fmanila-tempest-plugin~705917","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1ba4ae3dcc274bf5e6dee5320e52117f440ebb5d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1862001","hashtags":[],"change_id":"I1ba4ae3dcc274bf5e6dee5320e52117f440ebb5d","subject":"Fix resources cleanup","status":"MERGED","created":"2020-02-05 10:41:15.000000000","updated":"2020-02-05 23:41:06.000000000","submitted":"2020-02-05 23:41:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705917-1580946066329-b4850ae3","meta_rev_id":"6355fcdb91be335508ff1e8df8af335c59a8711e","_number":705917,"virtual_id_number":705917,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-02-05 16:51:38.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 23:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"date":"2020-02-05 12:43:45.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-02-05 20:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-05 22:17:57.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-02-05 15:16:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-05 22:17:57.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-05 15:16:28.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 15:16:28.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-02-05 15:16:36.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-02-05 16:51:38.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-05 20:02:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-02-05 22:17:57.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-02-05 23:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66c02ffe61c7253d3e26e9c00bccbf8ec315e017","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-02-05 10:41:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41146f7c4ec0f05a9cb3065a753e0c6c1cf6a4da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 12:39:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d5e2a2e9e74e14af00839ea0269228 : SUCCESS in 3m 47s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/67aefd8d81dd4848838883eeaafa39c4 : SUCCESS in 1h 47m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/183840b95c674c35877db7e59808f5fd : SUCCESS in 1h 01m 32s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/58393c73a58147bc852335444edf1d1a : FAILURE in 1h 19m 22s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b3a39ffe4c4e4eb687da000218a0530f : FAILURE in 52m 32s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/7685abbf98954245886654166c78d8a4 : SUCCESS in 1h 14m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf6c964d09c64053b2c6d1f275ae7917 : FAILURE in 1h 51m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/42b5bb1beb1c4b62a9eb198410718a52 : SUCCESS in 27m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/961875474b5343399af771854e08fd3f : SUCCESS in 1h 30m 03s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1d37ea383dc5415eb89b9e3c58c3205e : FAILURE in 45m 51s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5a35307d97af4c3ba57ebf9d74a07af1 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e1487033c0c9b1891397e850bbaf17ac5f4dcb4d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-02-05 12:43:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3ce135ae9461bf7e97db33d7d2c7cc05b0a851c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 14:32:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463b097ff0c648999f840740219ae180 : SUCCESS in 4m 08s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/53e3cdd9c6bf4de694aa2cf2265d9b40 : FAILURE in 1h 40m 09s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/cd4c7ea726884e65bce579c9f03425f7 : SUCCESS in 1h 05m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/66a5a5dde06440088ccaa39e135b2a55 : SUCCESS in 1h 05m 17s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/fe39b3c121494814bf037dfc285967cc : SUCCESS in 49m 46s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/42236d145c2e4728aba657df97a84a4a : FAILURE in 1h 01m 30s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/132f6919938d482ba62dfb5b20ab29c7 : FAILURE in 1h 42m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/58b394b5fc9f48fbbf6cce7839feaa36 : SUCCESS in 29m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5c942d4ac70f4fd0b497df48178452b2 : SUCCESS in 1h 29m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b1e52a553bf145f2bb8b1dcfc3017a2a : SUCCESS in 50m 49s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/d467fb238e124d73a353d34110338231 : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"000ff82d3c8e2a8da9917f7884b920ca574eb189","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-02-05 16:51:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/17/705917/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c4313a7/ : SUCCESS in 1h 24m 33s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/17/705917/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8b3acd9/ : SUCCESS in 1h 32m 17s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"468a87cb71810dd29a74c0cb9c8e489f0ab27fec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-02-05 20:02:47.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"bbf6b3847ec0fad1dd2a42a35a6b0807f4c0bb80","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-02-05 22:17:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, Liron!","accounts_in_message":[],"_revision_number":1},{"id":"0bb13636092a395a2fd8030def063ee3861b346e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 22:18:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"805971214151924e57d25b7da85d0bfda2cba2dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:41:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5a912621282cfe153c6d5afc840df2924433b6ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:41:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c58513fa8154dc7a8771011d8bcf914 : SUCCESS in 5m 43s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/09749a607b6f49cea14a3000f7651c98 : SUCCESS in 1h 02m 25s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8dd7bd2a858a4c8aab183650186988cd : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"820064b6d7c4a39a39400c11a3c138d6bf4d9074","revisions":{"820064b6d7c4a39a39400c11a3c138d6bf4d9074":{"kind":"REWORK","_number":1,"created":"2020-02-05 10:41:15.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/17/705917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/705917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/705917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/705917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/705917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/705917/1"}}},"commit":{"parents":[{"commit":"ca1340cc44d1dec1ed3351d9ea650a179b5d695a","subject":"Merge \"Use a single keypair for multiple instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca1340cc44d1dec1ed3351d9ea650a179b5d695a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-02-05 10:07:52.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-02-05 10:40:47.000000000","tz":120},"subject":"Fix resources cleanup","message":"Fix resources cleanup\n\nWhen running \"test_create_manage_and_write\" test with\ndefault share type (i.e. set default_share_type_name\n\u003d default in tempest.conf) it passed, however, when\nthere is no default share type it failed with\n\"Target share type is still in use.\"\nThis is because the failed managed share is not deleted\nbefore the share type it is based on.\nThis patch makes sure the manila share is deleted before\nits share type.\n\nChange-Id: I1ba4ae3dcc274bf5e6dee5320e52117f440ebb5d\nRelated-bug: #1862001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/820064b6d7c4a39a39400c11a3c138d6bf4d9074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/820064b6d7c4a39a39400c11a3c138d6bf4d9074"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
