)]}'
{"id":"openstack%2Fmanila-tempest-plugin~710661","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1654598","hashtags":[],"change_id":"Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4","subject":"Fix export location negative tests","status":"MERGED","created":"2020-03-01 20:42:47.000000000","updated":"2020-04-06 17:36:16.000000000","submitted":"2020-04-06 17:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710661-1586194576460-6ca8d248","meta_rev_id":"12ab06b6c7d0dddccdefbcd08d602449a7ee60cb","_number":710661,"virtual_id_number":710661,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-04-04 04:05:03.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 17:36:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-04-04 12:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-04-04 12:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-04-04 00:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-04-06 16:10:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-04 11:56:22.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-04 04:05:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-04 11:56:22.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-04-04 12:09:42.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-04-04 12:32:06.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-06 16:10:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-06 17:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d78d9af57fffe1471e59370e07e0d76ddd5ebe61","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-01 20:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f1b97577a1448def0291b04ac90d8dcd813df7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 20:50:25.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"9de9bd2d94a35ebe6b1711cf8fc377a42f4bce91","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-02 06:38:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"93b08f1df98b749bf66ac74169390e060675e502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 07:34:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016d84598a2e4f98ac13c20e08e6f9c0 : FAILURE in 7m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3a4bd9c0be55422e8e04e56452f68629 : FAILURE in 43m 23s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ff803aa27f494e42a10b1d4ec3cd1141 : FAILURE in 44m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/4174a54829714fd79e3d3cf0c63642d9 : FAILURE in 41m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/9eae62676d754686b6143bcf98ff12df : FAILURE in 48m 35s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/caae348215e04b9eb66cd948038e53ef : FAILURE in 44m 28s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4237989c6cf486e829a38196e4b502e : FAILURE in 46m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/80c9085f073a4de993fe18dbcb69424a : FAILURE in 23m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57febfcea97244eaad917198c0e32697 : FAILURE in 44m 06s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/289e239a0a40492dae5d96791b8fbe5f : FAILURE in 26m 57s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/cb9a3ab0820e4bbbb5419a330c802bac : FAILURE in 41m 45s","accounts_in_message":[],"_revision_number":2},{"id":"988099256f74f1effe0b3c91a8735e107d1322df","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-02 10:17:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ab96e39236bb11487445555f7c52202b302575ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:23:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a006522fae4a1dad79d16324fd6aac : SUCCESS in 4m 34s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5c3aa3c0ee474d25be570c1702b5d0e9 : FAILURE in 41m 42s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6537c333d046421cb1cbbdc2da3db3e3 : FAILURE in 46m 32s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a9c2817bd62e43e48a7249dbc462ea98 : FAILURE in 46m 58s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/1b22f6cd0cfe4d18937ea978aef76444 : FAILURE in 51m 59s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/20fc0aee1df1450fa9d8af4f8b7106e5 : FAILURE in 53m 47s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa68a36c3cd04048be08977f3de33c82 : FAILURE in 41m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/bdb9f6531a8e433795e295d654f95223 : FAILURE in 26m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4f20ea82964042c4bb2c354a0f6eda84 : FAILURE in 36m 04s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/17790b8c45514d669c65d3326a2a3f98 : FAILURE in 27m 43s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ab3e2ddbfdc84c67b133c64591f19f1d : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":3},{"id":"772e4b63314de39524b3eeb8a25639a664989553","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-02 15:14:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"eb42ff3a70f512550571b4178ca6c55366e8a9db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:07:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445bf40cba1e46708071ac01e6e378af : SUCCESS in 5m 51s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a63a83c960fc417aaa7f6412267fa16f : FAILURE in 1h 29m 57s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4c7b4ada1c624fa69378050f7ab97d5f : SUCCESS in 51m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f9c4443152e64527aca6da4739cfa99e : SUCCESS in 1h 12m 26s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f381fae3aaa24f97ac813e9c529a81eb : FAILURE in 58m 25s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/db770ec8fbe24d4b87d875be0a5bceeb : SUCCESS in 1h 09m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/035e78c2c54d44ccb7759e1cea7cfdd3 : FAILURE in 1h 45m 28s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/edd8ad92ab6a4a29bc49bce84f5fbcb9 : SUCCESS in 33m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/421e180f161341349e7fdb5d9173d7a1 : SUCCESS in 1h 16m 35s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/337f111b2be94566a7a25f1fe9fa6607 : FAILURE in 42m 33s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/63855c9dc7c24c2bb441f81a8b164cda : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":4},{"id":"4d147a7ca5ff6cff240531fc9d5c333a09411881","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-02 18:12:26.000000000","message":"Patch Set 4:\n\nrecheck\n\nunrelated intermittent replication test failure","accounts_in_message":[],"_revision_number":4},{"id":"dcd09fc76d6eee44c53ccbec48a58daca130dd4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 19:55:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e6cbb3b4064879a062a32b1474a081 : SUCCESS in 5m 19s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/afa651f668ae4a9a98f5d60841f24084 : FAILURE in 1h 25m 08s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/43747ebb5eb342048e14957b6f156f5e : SUCCESS in 54m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f66e729bb63847569d6f75ece2f334de : SUCCESS in 1h 07m 15s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ba1d4fbb948346eb80347c5a19b6b1d7 : SUCCESS in 55m 48s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/0e42c42c471a4d359e325586eff05e51 : SUCCESS in 1h 07m 33s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6d2ca2ed5e14d1b906fb3fb5da99746 : FAILURE in 1h 36m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e1209e136b02400588e4666cca34886d : SUCCESS in 30m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a9b20a3a031c45318bd9cb80f6b420ec : SUCCESS in 1h 32m 00s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d8cbdf3d043644f892f2a44248881698 : SUCCESS in 49m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c20b06a25d5340a99ba452e9aec1ef68 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":4},{"id":"2ff6dd5a4366f96e4966e734761ec8c7a12c8cab","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-02 23:35:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/61/710661/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/375d32e/ : SUCCESS in 2h 02m 30s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/61/710661/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9daf7b4/ : SUCCESS in 1h 31m 46s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"7417a102ff077243fb4c4ecd169e3fe87b9aa1e5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 07:15:13.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f957e0adf2a0a4f9c77d2acb69f4d140a536e4b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 09:33:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd30bd10a3f44609fa74a46e61e5032 : SUCCESS in 6m 25s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/83a5292ee69f4a58b62c22e571a37da9 : FAILURE in 1h 26m 28s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ddd4159e93784634bffc5850a25f11d1 : SUCCESS in 57m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd20cb64696b43b8a6656f3d30166238 : FAILURE in 1h 17m 02s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/423cbcb8355b4954a9a61a5cc7dfb182 : FAILURE in 1h 00m 13s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/6a99af6ee74d4852bc97a3948fd3021a : FAILURE in 1h 08m 50s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caebb989135749a4894f9d0363a81d91 : FAILURE in 1h 26m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c66a37e228ae4d1dbfd5ad191a89be18 : SUCCESS in 31m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/092afd2090ed41d1aa54572aaa8c07c1 : SUCCESS in 1h 26m 30s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6698f36da8174feba3c368f95f40bf14 : SUCCESS in 46m 38s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/7097157bc08947089af1418221960b3a : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b6051be8b530074a13413b0b24c62fad42c0adc2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-03 09:52:53.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"0993a36f91b4a375a478ae85ca8f27b734749648","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 16:42:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0f86b973cb422ce36be841813afe47311e85a72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 18:46:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a486b8a98342ab97422a59b2965916 : SUCCESS in 6m 24s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/111b5c158b6b44c9b1e1e01017a4ead5 : FAILURE in 1h 27m 02s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/497387e4fd7441c5a6055851915132cd : SUCCESS in 48m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/44a2fd640cd44549b3e32eaf44f3fcec : FAILURE in 1h 19m 49s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/4ba3b645239748ab860c60b796d474d2 : SUCCESS in 57m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/1498c82158984506a6664dac8df4973e : SUCCESS in 1h 04m 52s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5a44e2ca31f431c8ed0b4a73ebdee10 : FAILURE in 1h 47m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/154f56255af3498599dc8c7055a1cac0 : SUCCESS in 31m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1d8e69e8ee854011814bfea211ff567a : SUCCESS in 1h 37m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9bb1cd5a91154094baa487ff23b2ada2 : SUCCESS in 44m 11s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/76edbb014568434aa3d9124527681ff3 : FAILURE in 1h 06m 34s","accounts_in_message":[],"_revision_number":5},{"id":"b37d8083027e586df8c6ceaa47091353275252a2","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-04 09:32:13.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3c2f96f100349cc4a037c9a573fcf2beca8e4ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:16:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36776c2cfe5f44b59ba660656003bc76 : SUCCESS in 4m 27s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/dc0188bf7069442abe75d9d1d7fe7d05 : FAILURE in 1h 37m 20s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/88a9444c7c03407081b4e347f821cbab : SUCCESS in 56m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8c2a3a695cfd4b33997134d4866d5a9c : SUCCESS in 1h 10m 01s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7288d92eb40e46159eb23e1b39c4d596 : FAILURE in 55m 40s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d668a92a4ee54e06bde0011ce91133e4 : FAILURE in 1h 05m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a69a2280c4b4cc88a980464429fa356 : FAILURE in 1h 39m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d62eb252e1134d9486570074483027fe : SUCCESS in 34m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4a8d3d0491d44a8f942d975f77c74b3c : SUCCESS in 1h 22m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/086f1d1cd4e041049d1ec0102c338f88 : SUCCESS in 50m 23s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c4f3396b1c33438d82f9fdf47fd481dc : SUCCESS in 1h 22m 36s","accounts_in_message":[],"_revision_number":5},{"id":"26734e47f903041bc794999938eb29d2ac2fdea2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-04 14:01:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://logs.openstack.netapp.com/logs/61/710661/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1a807dd/ : SUCCESS in 1h 51m 55s\n- manila-cDOT-manila-tempest-plugin-no-ss http://logs.openstack.netapp.com/logs/61/710661/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/de93b6e/ : FAILURE in 2m 29s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"7cab032e9c26a565f87bd734f309894ee020b7b9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-24 11:51:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"55da7b1bb86d4a0b8371908fcfe39f7749a86cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 13:54:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de657ee725de4277b7b577b90055ac93 : SUCCESS in 3m 37s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d32063a8e99a4031aa7e6acfc6af2128 : FAILURE in 1h 18m 14s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b470d2c3351f4e9ab9015f8c19c6902f : SUCCESS in 1h 02m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/63eeaa0a247f459b82d704165eb76be0 : FAILURE in 1h 12m 13s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5149a8cb63de473a80df1c48370b057b : SUCCESS in 52m 15s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/72961fca7510430bbeb2156d665212c7 : SUCCESS in 1h 07m 02s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72b174b354a449bc93a1ce3f351b79ea : FAILURE in 1h 37m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/bdb009371836476796a0c12a706e9bdb : SUCCESS in 25m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d4232cbb384c4cb79c595f33a0553370 : SUCCESS in 1h 26m 45s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2fad782fd5fb464eb55b3e909f21b263 : SUCCESS in 46m 53s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/0bdeceadaa0e45278daee2d7f9ac4510 : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":6},{"id":"1ec78526ea148a0e0ae4e470ef57edc54ef15bde","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-24 15:23:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/61/710661/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/4437f46/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/61/710661/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/32954ab/ : SUCCESS in 1h 28m 40s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"ed6085f32b891696debf080b86673a41a53cbdd4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-24 15:31:51.000000000","message":"Patch Set 6:\n\nI see I renamed the wrong tests in PS#6.","accounts_in_message":[],"_revision_number":6},{"id":"a334d76abe7b04e9834771e5697f94b6020a401b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 16:36:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7a582daa6f6602c03a068081fa307039ef478d16","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-24 18:14:04.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"6c04d4b1ed3ff3cf5cfac8dd72846fed970d6e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:33:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d568a3e4b1347f9b9bf64084af348bb : SUCCESS in 7m 04s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/277e5e26504d4dc89aae6dd7ce133636 : FAILURE in 1h 32m 20s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1dfd79bbe6894b028855034ab0a395b9 : FAILURE in 1h 06m 42s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/4700acbb3cb54f8dba038d2ad921c5af : SUCCESS in 1h 12m 34s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/53511cc7e50446c0b33e434fadace062 : FAILURE in 56m 00s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4ecd5fd08cbe4567ac91e5b0bc89d2df : SUCCESS in 58m 26s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8e832a925cb4c7cb6322a15b012ebb2 : FAILURE in 1h 52m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7fa95f8226f84f64851847cd05fec268 : FAILURE in 28m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/192207c8e0f843929229af994861ddaa : FAILURE in 1h 23m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/640f4fb3159d45c3b19f938a7f390a72 : FAILURE in 42m 20s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a1db420164e6466ab101e2da156a57c6 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":7},{"id":"d579f4f37c9a67397cfef7743018eaa89e0125e9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 18:40:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3584cb884a3d27224c7e11d0a2db5e4d5f34911f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 20:59:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e679c54c5ae644b49adf0a97109dbf89 : SUCCESS in 5m 13s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/914a7183037e4bcd9ac34be4ee313212 : FAILURE in 1h 36m 58s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/77d6a71c96d0433bba61534cfaf71889 : SUCCESS in 1h 05m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/28bf4f0d61ef4dfa87098e5bfd07ba3f : SUCCESS in 1h 09m 25s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f813c9ab0c0d470da2ad9455b65fe8ac : SUCCESS in 58m 24s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/dc426a5ae54f4f4981bcca376c63d93d : FAILURE in 1h 03m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd310e54ebec418b858be7cd2948b84e : FAILURE in 1h 35m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/bb14c15a5af3415dbf0524491339a8d9 : FAILURE in 28m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/010d6c28c2dd444e988a88449ebaeb58 : SUCCESS in 1h 25m 35s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/48d63b4b6105488a839c2c30ee63aa0d : SUCCESS in 45m 17s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9288aa7359fb48fe9509dd2a104c4f13 : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":8},{"id":"37a7bc578d72d80cbdb6b847d08df05f7e898b18","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-24 21:06:43.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"60415f05805fc63a98718bf6f4169dbdc1265ec3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-24 22:27:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/61/710661/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/310606d/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/61/710661/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d61e3be/ : SUCCESS in 1h 27m 15s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f5493acdbb4eb06fa5236fa591ad7f598ae92b04","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-04 00:04:57.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"00c84ef55dea18a600cb2dec51d6f2fa22a7db47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 02:04:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0deb62296e83408ab79dec9a650b1a0b : SUCCESS in 4m 37s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/002ca364738245b3a9702a79ebf1adf8 : FAILURE in 1h 56m 42s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/7bdf10a065cd4693a9bae9700a3f9086 : SUCCESS in 1h 26m 02s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/7702efc8aa8e427192c4005dad3cc9b0 : FAILURE in 1h 37m 12s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/185a50e885fc40acb076734a284d2977 : SUCCESS in 1h 23m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5e6c8972d5cb47228b98a8fed2e3ebcf : FAILURE in 50m 52s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd9177bd44b44529a752a5d9f6e4f2a7 : FAILURE in 1h 56m 54s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/cfb33fd4dfee4f819e9fb5e00edbde7c : SUCCESS in 42m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d9c163c56baa4645a56ac7106b99335d : FAILURE in 1h 56m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6b63a7e316e94744aad6b598a4c72299 : SUCCESS in 1h 00m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/d4561b84134f4a289b4cb83937222e3b : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":9},{"id":"017e1d6bbf078f2e6d2f97908c9ea4cb615b179e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-04 04:05:03.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/61/710661/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8d6a67d/ : SUCCESS in 1h 59m 54s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/61/710661/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/846ecfd/ : SUCCESS in 1h 28m 16s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"299cfc76cd43c248dfe1a9071b6d25356242cf25","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-04 12:09:42.000000000","message":"Patch Set 9: Code-Review+2\n\nTks, LGTM.","accounts_in_message":[],"_revision_number":9},{"id":"0dc6ac5b83f7bdead2dde7a3c886ed38a45ccadb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-04 12:32:06.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM :)","accounts_in_message":[],"_revision_number":9},{"id":"497e8fa72f4e32e66a5541db36999112bd4df59e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-06 16:10:17.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"2ebe26b150c381999ed00f0a864e6eb801f25ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:10:31.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ed847ad79dbf4b75e06f715e2ba951873cc701db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:36:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"60534f9d286cc82d5b5e461ae9e5f4b7f2887f63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:36:16.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61aedc22ce14f5bb87a2748efa1ec7a : SUCCESS in 4m 46s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/16e026e177104ffcac2e9661d2a0d9e2 : SUCCESS in 46m 58s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/861431dbbfa84bd987f6c4ba5dd57b8d : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","revisions":{"486d9b82864a4ef39a5af0e6621a932435ca2d6b":{"kind":"REWORK","_number":1,"created":"2020-03-01 20:42:47.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/61/710661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/1"}}},"commit":{"parents":[{"commit":"8521aef448154285702f671a9b7c6a2de12e0a02","subject":"Skip invalid member share-export tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8521aef448154285702f671a9b7c6a2de12e0a02"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:41:46.000000000","tz":60},"subject":"Test for policy violation getting export locations","message":"Test for policy violation getting export locations\n\nWhen running as a regular user, attempts to get share export\nlocations belonging to another user should be forbidden.\n\nRelated-bug: #1655427\nDepends-on: https://review.opendev.org/#/c/710639/\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/486d9b82864a4ef39a5af0e6621a932435ca2d6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/486d9b82864a4ef39a5af0e6621a932435ca2d6b"}]},"branch":"refs/heads/master"},"419a935aa30d786394fbb0f86a4a37ee912e2ce1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-02 06:38:16.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/61/710661/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/2"}}},"commit":{"parents":[{"commit":"8521aef448154285702f671a9b7c6a2de12e0a02","subject":"Skip invalid member share-export tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8521aef448154285702f671a9b7c6a2de12e0a02"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-02 06:38:10.000000000","tz":0},"subject":"Test for policy violation getting export locations","message":"Test for policy violation getting export locations\n\nWhen running as a regular user, attempts to get share export\nlocations belonging to another user should be forbidden.\n\nRelated-bug: #1655427\nDepends-on: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/419a935aa30d786394fbb0f86a4a37ee912e2ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/419a935aa30d786394fbb0f86a4a37ee912e2ce1"}]},"branch":"refs/heads/master"},"9682587cac4ee951c188d1034a0c30b280980398":{"kind":"REWORK","_number":3,"created":"2020-03-02 10:17:24.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/61/710661/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/3"}}},"commit":{"parents":[{"commit":"8521aef448154285702f671a9b7c6a2de12e0a02","subject":"Skip invalid member share-export tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8521aef448154285702f671a9b7c6a2de12e0a02"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-02 10:17:19.000000000","tz":0},"subject":"Test for policy violation getting export locations","message":"Test for policy violation getting export locations\n\nWhen running as a regular user, attempts to get share export\nlocations belonging to another user should be forbidden.\n\nRelated-bug: #1655427\nDepends-on: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9682587cac4ee951c188d1034a0c30b280980398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9682587cac4ee951c188d1034a0c30b280980398"}]},"branch":"refs/heads/master"},"98d26145dd20c7b3fa6e935196f83e138ebe268c":{"kind":"REWORK","_number":4,"created":"2020-03-02 15:14:16.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/61/710661/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/4"}}},"commit":{"parents":[{"commit":"8521aef448154285702f671a9b7c6a2de12e0a02","subject":"Skip invalid member share-export tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8521aef448154285702f671a9b7c6a2de12e0a02"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-02 15:14:10.000000000","tz":0},"subject":"Test for policy violation getting export locations","message":"Test for policy violation getting export locations\n\nWhen running as a regular user, attempts to get share export\nlocations belonging to another user should be forbidden.\n\nRelated-bug: #1655427\nDepends-on: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98d26145dd20c7b3fa6e935196f83e138ebe268c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98d26145dd20c7b3fa6e935196f83e138ebe268c"}]},"branch":"refs/heads/master"},"e4e6d204cde147abf3fd40c1340dcdc272f5185e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-03-03 07:15:13.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/61/710661/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/5"}}},"commit":{"parents":[{"commit":"8521aef448154285702f671a9b7c6a2de12e0a02","subject":"Skip invalid member share-export tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8521aef448154285702f671a9b7c6a2de12e0a02"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-03 07:14:09.000000000","tz":0},"subject":"Test for policy violation getting export locations","message":"Test for policy violation getting export locations\n\nWhen running as a regular user, attempts to get share export\nlocations belonging to another user should be forbidden.\n\nCloses-bug: #1655427\nDepends-on: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4e6d204cde147abf3fd40c1340dcdc272f5185e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4e6d204cde147abf3fd40c1340dcdc272f5185e"}]},"branch":"refs/heads/master"},"35df7c9ddd6353d270a610775b220cb870bf5cab":{"kind":"REWORK","_number":6,"created":"2020-03-24 11:51:47.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/61/710661/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/6"}}},"commit":{"parents":[{"commit":"c678e21b5e0d0aec2d014a40ceadcd573bac08f9","subject":"Fix export locations tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c678e21b5e0d0aec2d014a40ceadcd573bac08f9"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-24 11:48:34.000000000","tz":-240},"subject":"Test for policy violation getting export locations","message":"Test for policy violation getting export locations\n\nWhen running as a regular user, attempts to get share export\nlocations for a share belonging to another user should be\nforbidden.\n\nCloses-bug: #1655427\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35df7c9ddd6353d270a610775b220cb870bf5cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35df7c9ddd6353d270a610775b220cb870bf5cab"}]},"branch":"refs/heads/master"},"8ec1ca34db46056cad95dc3c9cefad14add28496":{"kind":"REWORK","_number":7,"created":"2020-03-24 16:36:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/710661/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/7"}}},"commit":{"parents":[{"commit":"c678e21b5e0d0aec2d014a40ceadcd573bac08f9","subject":"Fix export locations tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c678e21b5e0d0aec2d014a40ceadcd573bac08f9"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-24 16:31:15.000000000","tz":-420},"subject":"Fix export location negative tests","message":"Fix export location negative tests\n\nWhen running as a regular user, attempts to get share export\nlocations for a share belonging to another user should be\nforbidden.\n\nShare instance export locations are not available to regular\nusers by virtue of default policy.\n\nRelated-bug: #1654598\nCloses-bug: #1655427\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ec1ca34db46056cad95dc3c9cefad14add28496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ec1ca34db46056cad95dc3c9cefad14add28496"}]},"branch":"refs/heads/master"},"6c0b250bcf5679c4f6f518df0ccbca9a622d8448":{"kind":"REWORK","_number":8,"created":"2020-03-24 18:40:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/710661/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/8"}}},"commit":{"parents":[{"commit":"c678e21b5e0d0aec2d014a40ceadcd573bac08f9","subject":"Fix export locations tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c678e21b5e0d0aec2d014a40ceadcd573bac08f9"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-24 18:37:54.000000000","tz":-420},"subject":"Fix export location negative tests","message":"Fix export location negative tests\n\nWhen running as a regular user, attempts to get share export\nlocations for a share belonging to another user should be\nforbidden.\n\nShare instance export locations are not available to regular\nusers by virtue of default policy.\n\nRelated-bug: #1654598\nCloses-bug: #1655427\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c0b250bcf5679c4f6f518df0ccbca9a622d8448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c0b250bcf5679c4f6f518df0ccbca9a622d8448"}]},"branch":"refs/heads/master"},"b5ed5dfaa5ed4989bdefc30abb00902a46052951":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-04-04 00:04:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/710661/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/61/710661/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/61/710661/9"}}},"commit":{"parents":[{"commit":"ad6f7de944c78168f7cb567e6999ea40068fafeb","subject":"Fix hacking test requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ad6f7de944c78168f7cb567e6999ea40068fafeb"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:27:26.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-04 00:04:57.000000000","tz":0},"subject":"Fix export location negative tests","message":"Fix export location negative tests\n\nWhen running as a regular user, attempts to get share export\nlocations for a share belonging to another user should be\nforbidden.\n\nShare instance export locations are not available to regular\nusers by virtue of default policy.\n\nRelated-bug: #1654598\nCloses-bug: #1655427\nChange-Id: Iabe7fb68facd0ddffec738ab4e98d1de3a704ee4\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
