)]}'
{"id":"openstack%2Fmanila-tempest-plugin~714181","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I12a1df36e8d928c54c03ed644dd60557f349ddb3","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1654598","hashtags":[],"change_id":"I12a1df36e8d928c54c03ed644dd60557f349ddb3","subject":"Fix export locations tests","status":"MERGED","created":"2020-03-20 18:13:56.000000000","updated":"2020-03-23 17:25:26.000000000","submitted":"2020-03-23 17:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714181-1584984326585-81340386","meta_rev_id":"cf7a9c84389d53e08c922510219f8fa5afa66042","_number":714181,"virtual_id_number":714181,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-03-21 03:25:01.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 17:25:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-03-23 15:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-03-23 16:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-21 12:46:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-23 16:24:47.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 18:18:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-21 03:25:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-23 15:52:30.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-03-23 16:24:47.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-03-23 16:31:48.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-03-23 17:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"471290085f56716e346cc7834c00778ad0926f11","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-20 18:13:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9758ba5912bb940bbcd9b93d177a5f7dfd63d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 19:47:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d012c1b2087e47b8b0736895c6e5f36e : FAILURE in 5m 18s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c3e3922f5f0e444abe25e835c6037332 : FAILURE in 1h 28m 21s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/a9cc3c5135b34624be24f739062737fd : FAILURE in 53m 17s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/afb98e2a08f644aebb4d61d2cb7cda83 : FAILURE in 1h 25m 28s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/05d0bc5868294fdbb720088b0a60fcc1 : FAILURE in 45m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8643e90bb3bb4755984fd0401eb3d77b : SUCCESS in 1h 04m 24s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c8d73d3bf4840cebf5e66517b8904ca : FAILURE in 1h 27m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/5353e4b35105451287345ff22d2027bb : FAILURE in 28m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c0ba5bfeb6d04a4c9114f05b61e50bb2 : FAILURE in 1h 18m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ae99eb0eb78144fe975a384fad230845 : FAILURE in 45m 06s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/203e4b10703a4e9cacfae518cd194b13 : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e5a34ad7f7bb08d8761c1484cea4530e316fa261","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-20 21:34:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9d54c5db56fc44cc1db737bbf189350fd3e67b4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-20 22:47:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72bda31e232a3f715cab2592d83d4b4dc97cda0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 00:51:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95dceff6afbe4ea49017f28f7d11a6ef : SUCCESS in 4m 47s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/ad8769257a014b4f9f1c2bf9b29d21dc : SUCCESS in 1h 42m 39s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8492c3b41e364e2bbe27db6fdfb20c00 : SUCCESS in 58m 38s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/eaa768f54c6c43a3b5e304f3c39a12d9 : SUCCESS in 1h 20m 42s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/761438ef302546d981bbc8b42686818a : SUCCESS in 57m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4c91be6fb04d4b699cbd3b153d9b2e58 : SUCCESS in 1h 10m 46s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae78bead46c845e49beeecbf8ccb9613 : FAILURE in 1h 52m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/567b0c734a6247e4879990a5e655224c : SUCCESS in 28m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/73ed336c354349588fc8617372097c36 : SUCCESS in 1h 27m 48s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/cc8c863a9e7646aeb60fc49e5e0b0597 : SUCCESS in 48m 55s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a08e58842c774b67a62a142dcb464226 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":3},{"id":"5c050ec6d6b286b7d990f06529e599302c9e88c4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-21 03:25:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/81/714181/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/746579b/ : SUCCESS in 2h 20m 45s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/81/714181/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/eb532ba/ : SUCCESS in 1h 31m 12s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f15887363d3cf4b5d5c44165e394623a63c2ca40","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-21 12:46:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nThanks for this fix.  As we\u0027ve discussed, I didn\u0027t understand the original intent of those test cases.","accounts_in_message":[],"_revision_number":3},{"id":"1237969a8c7c851b59e2e1322cf2bd9121ce2af7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-03-23 15:52:30.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, Goutham! Thanks for working on this!","accounts_in_message":[],"_revision_number":3},{"id":"fdb721514ce64f590fd812395de35a9356432e59","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-23 16:24:47.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9d252c555e2cef66a23fabeffdf8d723e47a0053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 16:24:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c339dd4b38ba30ce4a452a48a2a4919739ae9bc7","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-23 16:31:48.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for providing this enhancement, LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"66a56a794b2d1e858b71a6d77d7490abcdc4177e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:25:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"09cb1ab3b8a8ce62fff60887a9afb1d236a8cdc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:25:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35aa54034cb7477790ebbf39dadf9431 : SUCCESS in 4m 18s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a8954fdb831f405caaa33f12ce51e454 : SUCCESS in 45m 07s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/960f4764f7bc49f2925d20e181873cc2 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c678e21b5e0d0aec2d014a40ceadcd573bac08f9","revisions":{"0823b72d2199a5eae59fd565a07eb55dd0155df2":{"kind":"REWORK","_number":1,"created":"2020-03-20 18:13:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/81/714181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/714181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/1"}}},"commit":{"parents":[{"commit":"39ef895d112f947564d7ca52a73a5b2afb82f741","subject":"Fix link to bug tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39ef895d112f947564d7ca52a73a5b2afb82f741"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 18:13:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 18:13:47.000000000","tz":-420},"subject":"DNM - export locations test fixes","message":"DNM - export locations test fixes\n\nChange-Id: I12a1df36e8d928c54c03ed644dd60557f349ddb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0823b72d2199a5eae59fd565a07eb55dd0155df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0823b72d2199a5eae59fd565a07eb55dd0155df2"}]},"branch":"refs/heads/master"},"70000458d09e30fce0c67e3f6643cb501085dad8":{"kind":"REWORK","_number":2,"created":"2020-03-20 21:34:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/81/714181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/714181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/2"}}},"commit":{"parents":[{"commit":"39ef895d112f947564d7ca52a73a5b2afb82f741","subject":"Fix link to bug tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39ef895d112f947564d7ca52a73a5b2afb82f741"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 18:13:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 21:34:13.000000000","tz":-420},"subject":"Fix export locations tests","message":"Fix export locations tests\n\nThe tempest tests\n- api.admin.test_export_locations.ExportLocationsTest#test_list_share_export_locations_by_member\n- api.admin.test_export_locations.ExportLocationsTest#test_get_share_export_location_by_member\n\nwere both written with the assumption that a user within a\nproject creates a share, and the share\u0027s export locations\nare available to other users in the project. In this specific\ncontext, the user creating the share has an \"admin\" role -\nbut that is just circumstantial. Any user with the ability\nto create shares in a project can do so, and expect that\nthose shares are accessible to other users by virtue of default\npolicy.\n\nHowever, tempest test projects each have only one user,\nand the private share in both these test cases is created\nwithin the original user\u0027s project, and is not supposed to\nbe accessible across projects. This behavior is called out in\nLP bug #1654598.\n\nSo, enhance the test infra to create a user within the same\nproject applying roles specified within tempest.conf and test\naccessibility with such a user.\n\nOnce bug #1654598 has been resolved, we can use the existing\ntest behavior of testing access via a different project as\nnegative test cases (See: https://review.opendev.org/710661/)\n\nRelated-Bug: #1654598\nPartial-Bug: #1655427\nChange-Id: I12a1df36e8d928c54c03ed644dd60557f349ddb3\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70000458d09e30fce0c67e3f6643cb501085dad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70000458d09e30fce0c67e3f6643cb501085dad8"}]},"branch":"refs/heads/master"},"c678e21b5e0d0aec2d014a40ceadcd573bac08f9":{"kind":"REWORK","_number":3,"created":"2020-03-20 22:47:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/81/714181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/81/714181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/81/714181/3"}}},"commit":{"parents":[{"commit":"39ef895d112f947564d7ca52a73a5b2afb82f741","subject":"Fix link to bug tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39ef895d112f947564d7ca52a73a5b2afb82f741"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 18:13:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-20 22:47:12.000000000","tz":-420},"subject":"Fix export locations tests","message":"Fix export locations tests\n\nThe tempest tests\n- api.admin.test_export_locations.ExportLocationsTest#test_list_share_export_locations_by_member\n- api.admin.test_export_locations.ExportLocationsTest#test_get_share_export_location_by_member\n\nwere both written with the assumption that a user within a\nproject creates a share, and the share\u0027s export locations\nare available to other users in the project. In this specific\ncontext, the user creating the share has an \"admin\" role -\nbut that is just circumstantial. Any user with the ability\nto create shares in a project can do so, and expect that\nthose shares are accessible to other users by virtue of default\npolicy.\n\nHowever, tempest test projects each have only one user,\nand the private share in both these test cases is created\nwithin the original user\u0027s project, and is not supposed to\nbe accessible across projects. This behavior is called out in\nLP bug #1654598.\n\nSo, enhance the test infra to create a user within the same\nproject applying roles specified within tempest.conf and test\naccessibility with such a user.\n\nOnce bug #1654598 has been resolved, we can use the existing\ntest behavior of testing access via a different project as\nnegative test cases (See: https://review.opendev.org/710661/)\n\nRelated-Bug: #1654598\nPartial-Bug: #1655427\nChange-Id: I12a1df36e8d928c54c03ed644dd60557f349ddb3\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c678e21b5e0d0aec2d014a40ceadcd573bac08f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c678e21b5e0d0aec2d014a40ceadcd573bac08f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
