)]}'
{"id":"openstack%2Fmanila-tempest-plugin~715675","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie5d776baa7f0cc7f3774dfcedbfc4ef8e3acf44c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"skip_check","hashtags":[],"change_id":"Ie5d776baa7f0cc7f3774dfcedbfc4ef8e3acf44c","subject":"Move skipping conditions under skip_checks method","status":"MERGED","created":"2020-03-29 04:37:18.000000000","updated":"2020-04-22 02:16:38.000000000","submitted":"2020-04-22 02:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715675-1587521798473-dfe5b5ea","meta_rev_id":"9b8de703b883bfb6067cbcfa511a6f5c8fd75d20","_number":715675,"virtual_id_number":715675,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-04-19 11:08:48.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 02:16:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-04-21 21:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-21 17:26:13.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-04-21 21:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-29 05:50:32.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-04-19 11:08:48.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-21 17:26:13.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-21 21:22:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-22 02:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f39adf5d21e792b76d65070027a4cd2d93ccdf3b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 04:37:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7643ccacbb684637d53bfa0e1f4c3f4ed742af9f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 05:33:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ffa94f3037214f3e80fe88f8ed682ce2350a7f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 07:40:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8b8abef72641a283ef3846c6be0c72 : SUCCESS in 3m 11s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f024695473274e4db296aa00e04e6ea5 : FAILURE in 1h 17m 09s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/cf9056ddacf942e8b6c0c96430e9247f : SUCCESS in 59m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b1355ec9bed44665bef9b7e409f4393d : SUCCESS in 1h 28m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/8fd3109b8ddb457681c83f7fc7d08592 : SUCCESS in 53m 17s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/404e5ad1e7ac4c638f3736e3a246eb0d : SUCCESS in 1h 07m 03s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/049881df847340d28adc3b4a7a787039 : FAILURE in 1h 44m 08s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/5975c543a5b64e8997d85b736bfe6dfc : SUCCESS in 32m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/985238510890418599a46a9f3a0f27a5 : SUCCESS in 1h 34m 40s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0fd2fa33ef7c454980583f224cc95171 : SUCCESS in 50m 23s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/68136bfb17eb4e0ab1f8f2b0adedd442 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":2},{"id":"0664e51af4fbf9a3b2382e0be3e59513759e2e3d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 07:43:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a51fd5632c117e6718d5a3a541502e2cd460329d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 09:24:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/028e81df85e74bb09a2babbf67d82e77 : SUCCESS in 4m 08s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b40f00dc274c4e9c935edee2c9d1d879 : FAILURE in 1h 37m 42s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e7cc01afae0845e4aaf5de0a72ae458d : SUCCESS in 1h 03m 09s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/36b28ba5168942e79228fedf3e3bf3c8 : FAILURE in 1h 20m 10s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/daeae690e25d4fe387a93916d5f6f15e : SUCCESS in 1h 03m 00s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/82dddf9573544bac8b446403ef2157ce : SUCCESS in 1h 15m 11s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bac3172b5974d2ab7ffaadb8028daca : FAILURE in 1h 20m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7f4c4dd579f340e087f7eb708aaf2ff2 : SUCCESS in 27m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/689f951806b3440eb4268a5997bee660 : SUCCESS in 1h 17m 26s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/149d0050f9e64f39815a8122b242d092 : SUCCESS in 40m 21s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/bfdf180c9ba843a5830ee7ceac8642ca : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":3},{"id":"93de502f1fe0cd7a933b2a29cfff3bdd1df4a7ca","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-29 11:25:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/75/715675/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c5b1a70/ : SUCCESS in 2h 00m 26s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/75/715675/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/69e90f8/ : SUCCESS in 1h 24m 17s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"3dedb9c14a8b68188445b361f35bcaaa42e9d00b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-18 00:43:06.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\nHelpful, a couple of more instances can be cleaned up in the file","accounts_in_message":[],"_revision_number":3},{"id":"45b2f82ef353a6d555968c45695d287292a4aa0f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-19 04:37:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bb0b0429d9f1e3debeecd3902d5e856215ced913","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-19 04:37:36.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c6f00d904e6754a69ba58fbb9d6419e8e087e419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 06:52:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1aae265eadf4de9bef4f3320a28ffef : SUCCESS in 3m 39s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a51773f65f7d477ba4b7a6c7ddd4b206 : SUCCESS in 1h 30m 39s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/c36832319b504797882b556104096640 : SUCCESS in 56m 09s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/ad3ac20cdc5741abb781c07492f33275 : SUCCESS in 1h 15m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ca8429b965144ad4a614d84edd96c282 : SUCCESS in 53m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4db0f1299d8d4d6e997876d0f9825b65 : FAILURE in 49m 16s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f742ca07fc44e0ca6880d1b0abe0e70 : FAILURE in 1h 27m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c8a65b7d29ab4cb69ad32411b2d9a9cc : SUCCESS in 23m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/802ff9ae30294e0dba6fd0f80b52d643 : SUCCESS in 1h 13m 40s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1bfef72629634a5db0554873ad7f9b5a : SUCCESS in 46m 00s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ee0e3d5cc9c74a82b325b46caa6206c0 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":4},{"id":"3bae5af58f5d119a61f666ba4907ea3aed927172","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-19 11:08:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/75/715675/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8921c5c/ : FAILURE in 2h 07m 49s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/75/715675/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/81f9b93/ : FAILURE in 3m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"85719cac517cc94562935dc9d06f26a5f0ca3721","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-21 17:26:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d0bc7c75430458a7e7324f4f24d1508b0f94010b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-21 21:22:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"eb6a13d6d5a4d2804a71b84d199daffc6e7ba4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 21:22:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d46a27aaa712bb70b681ba91fa98bfc8846be12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 23:32:32.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfdd93e631d4d4c9a654b2b4d3bd625 : SUCCESS in 4m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/012087e73faa48ecaaf403a08643cf1b : SUCCESS in 44m 04s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/10207e7b5892432c9fb899b036af7313 : FAILURE in 1h 33m 00s","accounts_in_message":[],"_revision_number":4},{"id":"d8264b00f52309ce5db5f5b93e2b89143bb1479f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-21 23:43:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1376bb43b9da818d218c86d5d8e4096c0266ed1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 01:20:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ba28f1906b4513827efc51cc809cb4 : SUCCESS in 3m 07s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/cf596ba01dfd4296acc8883c27142a65 : FAILURE in 1h 31m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/afd13ea69f674d01bcd0155e9064f93e : SUCCESS in 51m 07s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/ffba12a5c7be4db7b9aabe3e259b520a : SUCCESS in 48m 53s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e0531a103a8a4f8f9bc85bfe00e28d6c : SUCCESS in 55m 06s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/26f0016e0a9c4b5eae85cf6fe7cd61e0 : FAILURE in 55m 18s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac9c5b95e5f14aeeba335a1d576a96ba : SUCCESS in 49m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/26f4ff162130407ca5665adb04747604 : SUCCESS in 26m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4331d45b2e6b4de0b024c814e7bad611 : SUCCESS in 1h 15m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d36d0213a2d2476693593da9ecc752a8 : SUCCESS in 45m 58s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/79cca2256b98409ab1d78b5f2467f154 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":4},{"id":"faacc48635e2238a66b1761e0adb227e243463ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 01:20:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"262af2061064c3424e8c6afcdb82a08fcd30706e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:16:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665cd0d8c6fa420a8cc33c6d47651e89 : SUCCESS in 4m 04s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/692e229ffb7a4ff8b87597f4fb8e3cd4 : SUCCESS in 48m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e811b5511fcb462cab77724d0c6ec319 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":4},{"id":"4a820ef09eb55a11d3395deb7055e71e4a93fc32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:16:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e3aec90f76a476330d6b02685f3028f1c734f484","revisions":{"f79357da33c1988027855b860d967081e95d45e5":{"kind":"REWORK","_number":1,"created":"2020-03-29 04:37:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/75/715675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/715675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/1"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 04:36:13.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 04:36:13.000000000","tz":180},"subject":"Move skip conditions under skip_checks method","message":"Move skip conditions under skip_checks method\n\nChange-Id: Ie5d776baa7f0cc7f3774dfcedbfc4ef8e3acf44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f79357da33c1988027855b860d967081e95d45e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f79357da33c1988027855b860d967081e95d45e5"}]},"branch":"refs/heads/master"},"617879bda1eab00a065fc837e1f006add7f2239b":{"kind":"REWORK","_number":2,"created":"2020-03-29 05:33:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/75/715675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/715675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/2"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 04:36:13.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:30:52.000000000","tz":180},"subject":"Move skipping conditions under skip_checks method","message":"Move skipping conditions under skip_checks method\n\nSkipping conditions should placed under \u0027skip_checks\u0027\nmethod rather than \u0027resource_setup\u0027 method.\n\nChange-Id: Ie5d776baa7f0cc7f3774dfcedbfc4ef8e3acf44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/617879bda1eab00a065fc837e1f006add7f2239b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/617879bda1eab00a065fc837e1f006add7f2239b"}]},"branch":"refs/heads/master"},"47c3f1c51a1b35a6406afe7a48b673349e840cbc":{"kind":"REWORK","_number":3,"created":"2020-03-29 07:43:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/75/715675/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/715675/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/3"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 04:36:13.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 07:42:59.000000000","tz":180},"subject":"Move skipping conditions under skip_checks method","message":"Move skipping conditions under skip_checks method\n\nSkipping conditions should placed under \u0027skip_checks\u0027\nmethod rather than \u0027resource_setup\u0027 method.\n\nChange-Id: Ie5d776baa7f0cc7f3774dfcedbfc4ef8e3acf44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47c3f1c51a1b35a6406afe7a48b673349e840cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/47c3f1c51a1b35a6406afe7a48b673349e840cbc"}]},"branch":"refs/heads/master"},"e3aec90f76a476330d6b02685f3028f1c734f484":{"kind":"REWORK","_number":4,"created":"2020-04-19 04:37:19.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/75/715675/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/75/715675/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/75/715675/4"}}},"commit":{"parents":[{"commit":"ca4dd9f10178fe9e6d5fe81aac3c3b8f3dae6448","subject":"Update share quota functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca4dd9f10178fe9e6d5fe81aac3c3b8f3dae6448"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 04:36:13.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-19 04:37:06.000000000","tz":180},"subject":"Move skipping conditions under skip_checks method","message":"Move skipping conditions under skip_checks method\n\nSkipping conditions should placed under \u0027skip_checks\u0027\nmethod rather than \u0027resource_setup\u0027 method.\n\nChange-Id: Ie5d776baa7f0cc7f3774dfcedbfc4ef8e3acf44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3aec90f76a476330d6b02685f3028f1c734f484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3aec90f76a476330d6b02685f3028f1c734f484"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
