)]}'
{"id":"openstack%2Fmanila-tempest-plugin~715679","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ieb8082e8314c27b58e0a2d23981068f15b93e336","project":"openstack/manila-tempest-plugin","branch":"master","topic":"skip","hashtags":[],"change_id":"Ieb8082e8314c27b58e0a2d23981068f15b93e336","subject":"Refactoring skip conditions process","status":"MERGED","created":"2020-03-29 05:27:37.000000000","updated":"2020-04-21 01:54:52.000000000","submitted":"2020-04-21 01:54:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715679-1587434092795-795c03e7","meta_rev_id":"c47e4babd82cd0fb057553335821503633523b02","_number":715679,"virtual_id_number":715679,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-03-29 13:40:48.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-21 01:54:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-19 04:51:59.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-04-19 22:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 21:05:31.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-20 21:05:31.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-29 05:49:07.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-03-29 13:40:48.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-19 22:18:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-20 21:05:31.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-21 01:54:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34d860587d5bef1c56ac6cc8096afd875525ee35","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 05:27:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8649be7c488e8978513150eae417bb31e755f34","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 05:39:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4e0c8cb16d40deb608f5d346b01e2776fab3796e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 07:02:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5d242fddc9471b98bb4d90d666e840 : SUCCESS in 3m 00s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5c02aa5e02c649f4a3b259846cfc06d7 : FAILURE in 1h 18m 23s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f4c218defef04ece98c80fe932565757 : SUCCESS in 48m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2bb43b9b316444d78ca0cd9c84737171 : FAILURE in 43m 45s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5dcbc4f7ebaf498bb12e57c98fb8d7ce : SUCCESS in 54m 59s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3ee9c511ad6c4469a4a9c6ef0b092cfa : SUCCESS in 59m 12s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1cb3d3fb615449285414c0b4f5b40ee : FAILURE in 46m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c775b59c91fe4bdab427a2bfd56d63b0 : SUCCESS in 25m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b70522edca434c13923cddbb14eb00a0 : FAILURE in 58m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/bb34b5138a7543f999d37b6a3e0025cb : SUCCESS in 39m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/20631a156f5b4cdd8d51a7113fb2ee63 : FAILURE in 43m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e35e1bbb753b231947f32137cba351008b36f258","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 07:39:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"28b88c81dccdc8621ad96d7ee32ddc5d7e54d852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 09:14:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad34b0dc64354d9bad7942f7043ee426 : SUCCESS in 4m 01s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f8fec7ab894f4b9d91ce1809ce390d89 : SUCCESS in 1h 29m 55s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1287d563a4b14173aaa47ff1bdc9e1cb : SUCCESS in 1h 09m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/226a2d4b9a654a3aa1c76946a9fc2b19 : SUCCESS in 1h 05m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/1653ca1a7aab41d99658ee833013c653 : SUCCESS in 1h 04m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d9c0200a5c4b4c44bbd9aeeb7edac1bf : SUCCESS in 1h 19m 45s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96430ddce1784b3b952c9b84d25acbc1 : SUCCESS in 55m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/8c17bfda9b814ba9951708d23c1c4997 : SUCCESS in 24m 48s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0aecc1879a874d0a9ca53641920f9a08 : SUCCESS in 1h 17m 39s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/12969b5799c34c8a99b3d1da1cbf3042 : SUCCESS in 1h 00m 01s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/7b4eaf9fc4554a7e96f97a53a10dad81 : FAILURE in 1h 18m 15s","accounts_in_message":[],"_revision_number":3},{"id":"a3274ad75182571971fd637d8aedcf0c737f98d7","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 09:49:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a89df4f231ab88de61e479c8be52de17fa5e7290","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-03-29 09:52:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bb8c664cf91d873ce88b4e0f5f1d0db430da9df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 11:34:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00febaa81f2492cb24e3f73cc973c27 : SUCCESS in 3m 31s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b561a3021a3246e988f36704c9cbbaaa : FAILURE in 1h 22m 48s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/15f3fd6fa2fd42eebf7cbef7b3f3d70b : SUCCESS in 55m 52s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9528664f8014ad9ade28755c4bd3384 : SUCCESS in 49m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/82de921172154063b498b3e7b7b3a647 : SUCCESS in 55m 38s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8ba3cc1538b244d9a921ee8a4e7c78ab : SUCCESS in 1h 06m 43s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a3df512b9a04ffe9773c98c8060f4aa : SUCCESS in 46m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/be6ea72593364bbb9a5b7a1ef242846b : SUCCESS in 24m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/39866dddb0634d8f806df7045d691464 : FAILURE in 1h 07m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/423c1bad21274a60bd460c9c1a693926 : SUCCESS in 44m 01s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/fc2c594a6a2c4f648c8f864cae972345 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":5},{"id":"82cd8ca46c074ad56d4e0e60ccba41e66ba3a2b5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-29 13:40:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/79/715679/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/df37813/ : SUCCESS in 2h 06m 18s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/79/715679/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5d6979a/ : SUCCESS in 1h 29m 29s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"e70baa42562234d9761405baa5376bc4fce9bc75","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-18 00:35:33.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nThanks, good cleanup - suggestion inline..","accounts_in_message":[],"_revision_number":5},{"id":"a482fe42ecd9cd0cb4c6f66a55d4b57397ce7eab","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-19 04:51:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4510e015da5091aa08a5ab7c9e9ae77090725bc2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-19 22:18:46.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks, once we start adding CephFS-native testing here, we can refactor again and remove/move the exception to the base class...","accounts_in_message":[],"_revision_number":5},{"id":"ab33b94246b0816f4851afdaf8cb5e4b26aa79ad","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-20 21:05:31.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1962b2646626f334d39171177bfb0792f78f66eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 21:05:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"19a321063de0ed4beb040a77ef019a3f700a3160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 22:20:02.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b8c733a20943f6afdcec8502f76dba : SUCCESS in 3m 59s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9aa0c8e1d61d4e0ba25bc1ec03ae6a5b : FAILURE in 44m 29s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ddd71109a98f4666b4721f1938831f93 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":5},{"id":"c4e360c68680702b6de2492c23ff363606817d6a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-20 22:55:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"81bf2abad2b37a8d95553751784550f05cc4db6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 00:47:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69032f1ecfa479a82edde5fc859050d : SUCCESS in 3m 45s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f46fd1ee11f5474ca14582ada8cc1eea : FAILURE in 1h 18m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/11d3005f183e47febc51ebd495625fce : FAILURE in 34m 00s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/bc9f7a29b89c4fae91811d69afdf0559 : SUCCESS in 50m 53s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/4ba37bb73ca8427db65e2a5f92f856a0 : SUCCESS in 56m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/38d960c5e82149edb66b1e5aef6d6ee4 : FAILURE in 47m 32s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d22e85393ca4ba99ff4a1cffa06ab2e : SUCCESS in 36m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/5d48b2a56ca448faaa682d4784db335e : SUCCESS in 27m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9383702c946a4d7bb038a1db419acceb : SUCCESS in 1h 44m 51s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a4a400aaf8b24c44a0956790ef62f531 : SUCCESS in 48m 25s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/cad1597e286a4352b716782ca5ff6ec0 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":5},{"id":"1ecb9223bce8d9fe4caaeaed1a255d6a93bf518f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 00:47:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b44e4d8b266321735516c72883a0d7766d5e39a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 01:54:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a295f9172f3a09dff804de8a1e7cc7d1c7de4d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 01:54:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b2688a566c411bb25af73f92e19c50 : SUCCESS in 3m 18s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2bc263ac681b4d5e975661f148d9a939 : SUCCESS in 46m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4dbb8146c1a44d079215429a419c4bd8 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5608f1bc70faed221bdbe9a4288a897a109700ea","revisions":{"e493257ef8a55f985fd04d88534bffd3f6f3f268":{"kind":"REWORK","_number":1,"created":"2020-03-29 05:27:37.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/79/715679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/715679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/1"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:12:55.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:12:55.000000000","tz":180},"subject":"Refactoring skip conditions process","message":"Refactoring skip conditions process\n\nChecking ip rules for protocols is placed under \u0027setUp\u0027\nmethod which prevents running other protocols that do\nnot support ip rules.\nThis patch moves the skipping condition under \u0027skip_checks\u0027\nmethod for each class.\n\nChange-Id: Ieb8082e8314c27b58e0a2d23981068f15b93e336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e493257ef8a55f985fd04d88534bffd3f6f3f268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e493257ef8a55f985fd04d88534bffd3f6f3f268"}]},"branch":"refs/heads/master"},"24b67a41f0c91884cc6ac34d2043f58f35a33a3b":{"kind":"REWORK","_number":2,"created":"2020-03-29 05:39:00.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/79/715679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/715679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/2"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:12:55.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:38:45.000000000","tz":180},"subject":"Refactoring skip conditions process","message":"Refactoring skip conditions process\n\nChecking ip rules for protocols is placed under \u0027setUp\u0027\nmethod which prevents running other protocols that do\nnot support ip rules.\nThis patch moves the skipping condition under \u0027skip_checks\u0027\nmethod for each class.\n\nChange-Id: Ieb8082e8314c27b58e0a2d23981068f15b93e336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24b67a41f0c91884cc6ac34d2043f58f35a33a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24b67a41f0c91884cc6ac34d2043f58f35a33a3b"}]},"branch":"refs/heads/master"},"59e5c6311927df0d3d0c33c81eb302a1ce308e8e":{"kind":"REWORK","_number":3,"created":"2020-03-29 07:39:42.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/79/715679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/715679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/3"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:12:55.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 07:39:29.000000000","tz":180},"subject":"Refactoring skip conditions process","message":"Refactoring skip conditions process\n\nChecking ip rules for protocols is placed under \u0027setUp\u0027\nmethod which prevents running other protocols that do\nnot support ip rules.\nThis patch moves the skipping condition under \u0027skip_checks\u0027\nmethod for each class.\n\nChange-Id: Ieb8082e8314c27b58e0a2d23981068f15b93e336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59e5c6311927df0d3d0c33c81eb302a1ce308e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59e5c6311927df0d3d0c33c81eb302a1ce308e8e"}]},"branch":"refs/heads/master"},"196fa0634a445196a0e64d2382fe532ff95a5ff3":{"kind":"REWORK","_number":4,"created":"2020-03-29 09:49:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/79/715679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/715679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/4"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:12:55.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 09:49:33.000000000","tz":180},"subject":"Refactoring skip conditions process","message":"Refactoring skip conditions process\n\nChecking ip rules for protocols is placed under \u0027setUp\u0027\nmethod which prevents running other protocols that do\nnot support ip rules.\nThis patch moves the skipping condition under \u0027skip_checks\u0027\nmethod for each class.\n\nChange-Id: Ieb8082e8314c27b58e0a2d23981068f15b93e336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/196fa0634a445196a0e64d2382fe532ff95a5ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/196fa0634a445196a0e64d2382fe532ff95a5ff3"}]},"branch":"refs/heads/master"},"5608f1bc70faed221bdbe9a4288a897a109700ea":{"kind":"REWORK","_number":5,"created":"2020-03-29 09:52:11.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/79/715679/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/79/715679/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/79/715679/5"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 05:12:55.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-03-29 09:51:59.000000000","tz":180},"subject":"Refactoring skip conditions process","message":"Refactoring skip conditions process\n\nChecking ip rules for protocols is placed under \u0027setUp\u0027\nmethod which prevents running other protocols that do\nnot support ip rules.\nThis patch moves the skipping condition under \u0027skip_checks\u0027\nmethod for each class.\n\nChange-Id: Ieb8082e8314c27b58e0a2d23981068f15b93e336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5608f1bc70faed221bdbe9a4288a897a109700ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5608f1bc70faed221bdbe9a4288a897a109700ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
