)]}'
{"id":"openstack%2Fmanila-tempest-plugin~715758","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Idf34a149e5a975d5fbced0ec78d102c5d90b87c6","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1858328","hashtags":[],"change_id":"Idf34a149e5a975d5fbced0ec78d102c5d90b87c6","subject":"Update shrink scenario tests","status":"MERGED","created":"2020-03-30 02:18:57.000000000","updated":"2020-04-28 01:02:47.000000000","submitted":"2020-04-27 22:35:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715758-1588026939358-5004bf0a","meta_rev_id":"99db5f55302534f7182d6237d5aa207787a5026f","_number":715758,"virtual_id_number":715758,"owner":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"date":"2020-04-28 01:02:47.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 22:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-04-27 21:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-04-27 21:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-04-27 21:23:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 15:28:54.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-23 17:45:32.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-27 21:17:37.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-04-27 21:23:18.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-27 22:35:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 01:02:47.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5b16d801f00190881bf8446f161433b0cb0d598","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-03-30 02:18:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03909ece0513e33302a04f03897d29e0fe27806e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 03:54:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428836596d1b4e128db4c8d8545469ae : SUCCESS in 3m 19s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/43a3a52889ff4e8493e629d04ecae749 : SUCCESS in 1h 27m 30s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8e3314004a3441ec84b0d1ff730c1fee : SUCCESS in 55m 03s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9a18aa311d16489292dacfb96732d31f : SUCCESS in 1h 11m 57s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/708f5ed0731c46db87f59475fd66c71a : SUCCESS in 51m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/391a93dd94e94e1fbe250515052de7ce : SUCCESS in 1h 08m 42s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74d68025cf194dc1b5960debdbbcd123 : FAILURE in 1h 28m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ad1cdcfd73724b829586d239890005aa : SUCCESS in 26m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f6675877e68347bcafcf99413ff4a686 : SUCCESS in 1h 29m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/bb3614e8872249d897bea6ee022ba135 : SUCCESS in 44m 34s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/b85e162dc8f14ce39b87cb5ede5aaadd : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fcd4398a9618c080f5084bc612337371bbeb00d6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-30 06:03:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/58/715758/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a3e96bb/ : SUCCESS in 2h 08m 20s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/58/715758/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/128fa95/ : SUCCESS in 1h 28m 11s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"40e3849b4667a65bfb75e39c4c1813353fd4bf84","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-18 00:25:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0c76c406c00a1fcdb104b6b7031a0f46f486f022","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-23 15:15:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2adbba38aea86a1d06da585e45a2df31ccd8f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:59:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a26eb0faaf459095da61f0b9a03798 : SUCCESS in 5m 56s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/47934c71276942f0aea7eae2b51c7116 : FAILURE in 1h 05m 32s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ac04e05ebe8f49b2b8b138d76b1579dd : SUCCESS in 52m 12s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3388643e0add4a42bc0b228b6ef8af7f : FAILURE in 1h 19m 32s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7dc8e5447e8d4a6180dc2a4a36dc67c9 : FAILURE in 50m 35s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2a157c271bb24d79aa795c4a45340d66 : FAILURE in 42m 48s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7c6825c8f234be79d7e70dd9a934232 : FAILURE in 1h 31m 27s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/66f13921e3eb4a2c8bd7354359834257 : SUCCESS in 27m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cbe7ecae78564b6ba1260638f409b207 : FAILURE in 1h 11m 35s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/596c91bd38604c6b8338f6a8371d872c : SUCCESS in 41m 46s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/2612d18b23e84513b0a56314be1b795a : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":2},{"id":"97d059cf5495e2e2aec289792e01872a036a604b","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-23 17:25:41.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b4217a0dd5399474999be5fcd978b2319b77bd59","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-23 17:45:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nHi, Daniel! Great work on this patch. I have addressed some concerns for you. Please check them inline.","accounts_in_message":[],"_revision_number":2},{"id":"2b2dd5c9e416eb02f29904fe275463ef56d9d02d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 19:13:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/58/715758/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/af22115/ : FAILURE in 2h 14m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/58/715758/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/516b7ab/ : FAILURE in 1h 23m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4090cf4031f48e5a43532a301ca1d98e4628b691","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-24 21:04:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f37766a1aea61d71da027f2b6b1165a2e714fc7","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-24 21:05:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0c979a23a74fc61d9aeef9799ecdc0d16298bc55","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-24 21:06:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c9525c0f7bca2aec16899d99de5fdb671f648256","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-24 21:17:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nThanks for working on this. Please take a look on the comment inline.","accounts_in_message":[],"_revision_number":5},{"id":"2a341922c45ca10f23474afe523ef84f4b57fdbe","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-24 22:39:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"51bddaa7e5216eff00d0f6406712b954decbb268","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-24 22:41:26.000000000","message":"Patch Set 6:\n\n\u003e (5 comments)\n \u003e\n \u003e Hi, Daniel! Great work on this patch. I have addressed some\n \u003e concerns for you. Please check them inline.\n\nHi Carlos, thank you for your review.\n\n\"nit: I think that there is no need to remove this blank line\"\nDone!\n\n\"I think you could filter this using the resource_id as Goutham suggested... So the queried messages would only be related to the resource id you\u0027re dealing with.\"\nI totally agree with that, so we filter in a much better way. I apologize to Goutham for not seeing this before.\n\nIs there a specific reason to set this value to None?\nIf not, i think this could remain as it was:\nbody \u003d self.shares_v2_client.get_share(share_id)\nshare_status \u003d body[status_attr]\nSince at this point the share will possibly be in a state different from available.\n\n\"Then the get_share request would only be executed once by each while iteration and we would avoid it being called at the very beginning and at the end. :). What do you think?\"\n\nThis patch was made to update the tests for the new behavior of the share manager. Since the share-manager update is not yet merged, when I arrive at this part of the code the share status will be \"shrinking_possible_data_loss_error\". With this condition we have no problem. The problem resides in the fact that the update of the share-manager will put the state of the share as available when try to shrink a share to a size smaller than the current used space instead of \"shrinking_possible_data_loss_error\".\n\nThe scenario test must pass in both cases, so in the last one I need to enter in the loop of retries to shrink, but the state is already available. In the case that the backend is slow, the trash put inside it must be not completely erased, so the retry to shrink the share is necessary and it must remain in the loop. So I just added a new logic that tracks the size of the share instead of its state. What do you think now?","accounts_in_message":[],"_revision_number":6},{"id":"cb9c20ef2cf775571724ad6765f241b4ff152a4e","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-24 22:41:55.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n \u003e \n \u003e Thanks for working on this. Please take a look on the comment\n \u003e inline.\n\nHi Douglas, yes, this is a better approach to filter. Done :)","accounts_in_message":[],"_revision_number":6},{"id":"5ac06550e569dc4dcdcfc75ffd5a7966947447b3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-25 13:22:21.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nHey Daniel, thanks for updating this change. Please take a look on the comment inline.","accounts_in_message":[],"_revision_number":6},{"id":"bea0ba0e14c5a09d41f8e29df75d695ecda827a2","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-27 12:56:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4aac49275efc660dd38871a308a4a70811bbaf0b","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-27 12:57:25.000000000","message":"Patch Set 7:\n\n\u003e (1 comment)\n \u003e \n \u003e Hey Daniel, thanks for updating this change. Please take a look on\n \u003e the comment inline.\n\nHi Douglas, agreed!","accounts_in_message":[],"_revision_number":7},{"id":"874e0333f4bb6fc95f593921b8c1aeb70d454952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:12:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1699476e4e9741c89f02f21a19e0b78d : SUCCESS in 4m 45s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b1dd860663ee462c954c930933450724 : FAILURE in 1h 10m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d08bed3027c74791be3aa8a231b7765d : SUCCESS in 1h 10m 57s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8ab484faea0e4b49bf1b7432a5bc5caa : FAILURE in 1h 47m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b51323166cf7468dacbc82a5c1d11f65 : SUCCESS in 58m 00s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/999b50acae324019996fabe0366a187a : FAILURE in 1h 01m 24s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02121c75f6df48a885375408a5301eb9 : FAILURE in 1h 52m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/316f68d7bdfd49fc81c0d1ab31761354 : SUCCESS in 24m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c14497774df54bd2902403d7595a3e09 : SUCCESS in 1h 02m 37s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/06eb017209a84d32a1bd4e5e3e2078ec : SUCCESS in 45m 42s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9c9961715334401ea87f7977f10a5e6c : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":7},{"id":"85f4265d3c99b09cf74eb634546b759e5c4f3133","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 16:19:54.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)\n\nthanks for the changes, couple of questions and a few comments inline.","accounts_in_message":[],"_revision_number":7},{"id":"2ed6af8f92ea013b7e0e17035a09c960092f2930","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-27 18:22:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c0418ea3918f09b85d6bcd4db7b448c4945db798","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-27 18:29:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e151f244d0d10cc6de432d3e27104603e4d2eed9","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-27 18:38:26.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"50d72035a6b55170d951151b55c37ae6ad6da01a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 18:58:52.000000000","message":"Patch Set 9: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"958ffca9a951b8eef5103b31c80410483788c7e8","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-27 19:12:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5291178556cb08d80eee6634c119529eb6265458","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 19:21:45.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c77a01cddf4b37f4cae849338e1e00ba99f3182a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:15:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedcc960713d4b6788496b885db78413 : SUCCESS in 4m 20s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/acb73f0f6a784f7db0c7f2174dbeee50 : FAILURE in 1h 30m 45s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3dc8f3d1e0be42439de35667d36a5c70 : SUCCESS in 55m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3c4f7d6df75542089f41cc79015b4ebd : FAILURE in 1h 12m 11s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d2daf410667341afb513d79942e01f97 : SUCCESS in 52m 21s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/da9bd41193914aac9cc1f38b87c21945 : FAILURE in 50m 13s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02713c4297e14636b532710d3e20bf7d : FAILURE in 1h 59m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d344e44dc54e4a2f8fc7292bd4c9a324 : SUCCESS in 26m 08s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/90d325f8a35e4d7b9d7794985a69e28b : SUCCESS in 1h 11m 55s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d27845accece4d238fcd4dcfd1baa85e : SUCCESS in 44m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/207d8f0102a64a1c8dc0dc9b00980f04 : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":10},{"id":"a9c77f5270c40e8846c74cea88eb58685d82fa61","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-27 21:17:37.000000000","message":"Patch Set 10: Code-Review+2\n\nThanks Daniel, looks good to me.\nMost of the failed jobs are related to timeouts in share migration.","accounts_in_message":[],"_revision_number":10},{"id":"77890b67af26afadf8384a1d3d23516c4c214430","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 21:23:18.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nmanila_tempest_tests.tests.scenario.test_share_shrink.TestShareShrinkNFS.test_create_shrink_and_write passed in ZFSOnLinux and CephFS-NFS jobs, this should be good to unblock the manila side change..","accounts_in_message":[],"_revision_number":10},{"id":"4ce5b5511c96fce2085477c6cecef972b6d06337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:23:32.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4322c9475f2a1bc7d660fef70de6a99c0bcac403","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:35:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"8409be202cd340b33e702567fade76e9ecf7b081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:35:39.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a4b340dfec4c42b03b54a85387a049 : SUCCESS in 5m 25s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/000bb2a935614acfa5c764fe9b039f5e : SUCCESS in 42m 14s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/f8faec96fcdb412b9de419d047e93a40 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":10},{"id":"8a1c1d84b1e8161613a4819e00509bd89c0d7b97","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-28 01:02:47.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/58/715758/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f2ade81/ : SUCCESS in 2h 14m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/58/715758/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ac94245/ : SUCCESS in 1h 23m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"247b0b69791154ea51a84c036168e6bf9cb1208e","revisions":{"c53deffba809feb6b33427617872ffd4ee913aa4":{"kind":"REWORK","_number":1,"created":"2020-03-30 02:18:57.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/1"}}},"commit":{"parents":[{"commit":"39ef895d112f947564d7ca52a73a5b2afb82f741","subject":"Fix link to bug tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39ef895d112f947564d7ca52a73a5b2afb82f741"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"Daniel Tapia","email":"danielarthurt@gmail.com","date":"2020-03-30 02:18:56.000000000","tz":-180},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c53deffba809feb6b33427617872ffd4ee913aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c53deffba809feb6b33427617872ffd4ee913aa4"}]},"branch":"refs/heads/master"},"01d109f9134e3a7694905ef7954b4960cf99b7f7":{"kind":"REWORK","_number":2,"created":"2020-04-23 15:15:55.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/2"}}},"commit":{"parents":[{"commit":"39ef895d112f947564d7ca52a73a5b2afb82f741","subject":"Fix link to bug tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39ef895d112f947564d7ca52a73a5b2afb82f741"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-23 15:12:21.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/01d109f9134e3a7694905ef7954b4960cf99b7f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/01d109f9134e3a7694905ef7954b4960cf99b7f7"}]},"branch":"refs/heads/master"},"9f64291f3489ff1e6680e801b2866ef90f5aecee":{"kind":"REWORK","_number":3,"created":"2020-04-24 21:04:03.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/3"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-24 21:03:51.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9f64291f3489ff1e6680e801b2866ef90f5aecee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9f64291f3489ff1e6680e801b2866ef90f5aecee"}]},"branch":"refs/heads/master"},"c045a9e61211f4fae1bef290aa1b863d1ac9bf9e":{"kind":"REWORK","_number":4,"created":"2020-04-24 21:05:31.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/4"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-24 21:05:16.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c045a9e61211f4fae1bef290aa1b863d1ac9bf9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c045a9e61211f4fae1bef290aa1b863d1ac9bf9e"}]},"branch":"refs/heads/master"},"eb40d3515081c2e2ecd50e93961dc0a990f77571":{"kind":"REWORK","_number":5,"created":"2020-04-24 21:06:37.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/5"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-24 21:06:25.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb40d3515081c2e2ecd50e93961dc0a990f77571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb40d3515081c2e2ecd50e93961dc0a990f77571"}]},"branch":"refs/heads/master"},"4bb86e2bcc7bb5baed50b6f80f0a0c008e2d7034":{"kind":"REWORK","_number":6,"created":"2020-04-24 22:39:19.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/6"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-24 22:39:05.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4bb86e2bcc7bb5baed50b6f80f0a0c008e2d7034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4bb86e2bcc7bb5baed50b6f80f0a0c008e2d7034"}]},"branch":"refs/heads/master"},"945c308eada3d8360a39976a6912e7471d78afa0":{"kind":"REWORK","_number":7,"created":"2020-04-27 12:56:26.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/7"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-27 12:56:13.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/945c308eada3d8360a39976a6912e7471d78afa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/945c308eada3d8360a39976a6912e7471d78afa0"}]},"branch":"refs/heads/master"},"eb6af34a2b451bebfad23d260b19d616c964d4a8":{"kind":"REWORK","_number":8,"created":"2020-04-27 18:22:00.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/8"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-27 18:21:29.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb6af34a2b451bebfad23d260b19d616c964d4a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eb6af34a2b451bebfad23d260b19d616c964d4a8"}]},"branch":"refs/heads/master"},"24024433fd0329134c6e447356a2de58f1a89170":{"kind":"REWORK","_number":9,"created":"2020-04-27 18:29:51.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/9"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-27 18:29:40.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24024433fd0329134c6e447356a2de58f1a89170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24024433fd0329134c6e447356a2de58f1a89170"}]},"branch":"refs/heads/master"},"247b0b69791154ea51a84c036168e6bf9cb1208e":{"kind":"REWORK","_number":10,"created":"2020-04-27 19:12:20.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/58/715758/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/715758/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/715758/10"}}},"commit":{"parents":[{"commit":"b5ed5dfaa5ed4989bdefc30abb00902a46052951","subject":"Fix export location negative tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5ed5dfaa5ed4989bdefc30abb00902a46052951"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-03-18 14:31:39.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-27 19:12:11.000000000","tz":0},"subject":"Update shrink scenario tests","message":"Update shrink scenario tests\n\nThis patch updates the shrink scenario tests to handle with the\n\"available\" or \"shrinking_possible_data_loss_error\" state received\nfrom the manila share manager after a failed shrink operation when\nattempt to shrink a share with a size smaller than thecurrent used\nspace.\n\nChange-Id: Idf34a149e5a975d5fbced0ec78d102c5d90b87c6\nRelated-Bug: #1858328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/247b0b69791154ea51a84c036168e6bf9cb1208e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/247b0b69791154ea51a84c036168e6bf9cb1208e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
