)]}'
{"id":"openstack%2Fmanila-tempest-plugin~716848","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7a048c754cb703eebee709e35d379a4c51a80b3a","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/query-user-message-by-timestamp","hashtags":[],"change_id":"I7a048c754cb703eebee709e35d379a4c51a80b3a","subject":"Add function to test user messages query by timestamp","status":"MERGED","created":"2020-04-02 06:08:01.000000000","updated":"2020-04-16 17:40:35.000000000","submitted":"2020-04-16 17:40:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716848-1587058835025-11ac1cc1","meta_rev_id":"7821d4595df217b01e93723a3321c85e20c32b1e","_number":716848,"virtual_id_number":716848,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-09 01:39:58.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-04-08 10:24:24.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 17:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-04-16 15:42:09.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-08 22:32:37.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2020-04-16 16:24:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 06:34:56.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-04-02 08:40:56.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-08 10:24:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 22:32:37.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-04-16 15:42:09.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-16 16:24:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-16 17:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbc967fe2bde7e17dbc4877e409d865c558a71ad","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-02 06:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"246591d8b276f100aa20263944aed8775ab1961b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 08:24:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b837bfe0defe4f4baa570fd41b2d87e4 : SUCCESS in 5m 26s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f0e60a65d7bf424fb7fceb2f035a0b93 : FAILURE in 1h 23m 12s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/2a8b03b02fa3421e9095bcfc4f66cc5e : SUCCESS in 53m 24s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9589910ac84e403192a8c592fa6a2604 : FAILURE in 1h 14m 29s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/fed44d21046740238b524d477724c152 : SUCCESS in 1h 03m 30s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/6b8a772bbf5f4f189250b68a473c19ab : SUCCESS in 1h 17m 23s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f1941df3dd2400a8cd1f1764129b625 : FAILURE in 2h 03m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/1994874dfba24c4e9401059a78a7578e : SUCCESS in 26m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4b9c28bdd871408c86086983ba11b009 : SUCCESS in 1h 17m 42s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f61d3b2a6e814b6b90281b34585054eb : SUCCESS in 45m 44s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/920a2c9fbd514148a28c64879c51a8c8 : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f98749ae8144f458725439ad17a320342fbcc15c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-02 09:51:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/716848/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f466f90/ : FAILURE in 3m 47s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/716848/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6e66a1e/ : SUCCESS in 1h 26m 53s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"da477cd91628be5338fe86775f6cabe151a1f6eb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-02 23:13:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks for these tests, some suggestions inline..","accounts_in_message":[],"_revision_number":1},{"id":"6845def40e3f341cdf829f66464b9c66f35d6fdd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-02 23:14:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be156e942e0126f503858afc94d17edb287a07cc","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-03 01:58:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e0ba8565e46592e992eceaf4b3af56fe9630561","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-03 02:00:32.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nthank you for your review,Goutham Pacha Ravi.","accounts_in_message":[],"_revision_number":2},{"id":"a41dd7e9c3ab7dc1f4dbb0530143ad85d2d26cc0","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-03 03:45:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67440911489998a8269530fd257fe57158b93801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 05:20:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4e9d91f7e6436f83d91f88460d94a8 : SUCCESS in 3m 41s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/457e158564c84fd0afc387cb98b70406 : FAILURE in 1h 26m 06s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1790d895591e4fde95b5504e170f79b8 : FAILURE in 52m 17s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/97b14fc8a4854523b11ff08c41309fba : FAILURE in 1h 11m 09s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/107aefc69b3f4b83be62ecaa3a2f5fd8 : SUCCESS in 56m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2ae07e10ee654c3cb75fbf2714856442 : FAILURE in 42m 51s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c0002f5e73445da87fa2bdfaa460af1 : FAILURE in 1h 28m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/81da1e710548465b85c9661ecca14365 : FAILURE in 23m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/14ebb53626954ef6b6040c99ac01e11c : FAILURE in 1h 19m 10s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/cf4432b0e30b41cf96fc795e156526dd : FAILURE in 44m 53s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9847942d8dfb46f0a38edbde9e2db0f7 : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":3},{"id":"fbf92fab0e24e6885528e96a804bd38e225bbc47","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-03 05:41:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"61ab5e21c50bf7cfd997af7146e214294854f3a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 07:24:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/787e0dc95a9e4a458550a54494ceb9ad : SUCCESS in 3m 57s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3f634fd417544439acc8f27ca271e78d : FAILURE in 1h 29m 00s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4a4622bb1cfb4c1dab2776a9d26b1517 : SUCCESS in 52m 32s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/0bfa3f31ac0f4a3e8f0f699181034070 : FAILURE in 1h 16m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/90ee7112871b4be7bed15d7af441f32f : SUCCESS in 47m 55s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d5897ce45e5a4bffb9b4b33c05adbaa7 : FAILURE in 46m 36s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6f1351ab493401f83dfae9d948fc9dd : FAILURE in 1h 36m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/894b3b250e4f4042b3c185d2223bda86 : SUCCESS in 26m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/67f87b5c1fbc41d987b588a8fd4bbe1e : SUCCESS in 1h 16m 39s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f0ba35b0c3424283a0f1ce57f842bb6a : SUCCESS in 45m 38s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/70e17030e7624bf6ae651707c372b1ca : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":4},{"id":"1cf77b687357b4bcefd40bf625ad36651895d2cf","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-03 09:27:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/716848/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2daf8f5/ : SUCCESS in 1h 57m 57s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/716848/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1a33a76/ : SUCCESS in 1h 18m 35s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"40bf1d6c094a91d0f3bc39fa76fa9ba47288fa4e","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-03 21:15:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThanks Haixin. This looks good, just a nit. Also, would you be able to add some inline comments to specify what case does and expected output? As the comments you interchanged with Goutham. I think that will make readability better.","accounts_in_message":[],"_revision_number":4},{"id":"364c1287e7ee2f0c1108b4b219e0a1ca07b0374d","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-04 10:49:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0baf3abcedd8fbca0149fd0530c5ae9d5f4bdcca","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-04 10:50:05.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nhi, Victoria Martinez de la Cruz.\nthank for your review.","accounts_in_message":[],"_revision_number":5},{"id":"d64e59dd10b0ec3facb4c5bb2580e0d7b9729d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 12:46:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11b4132e92974d6d94b710394e458691 : SUCCESS in 4m 13s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/69b578fc1b834d6b94733f7188ba0d85 : FAILURE in 1h 08m 17s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/5b32c7ae59b64e12b90fc28dab16493f : SUCCESS in 51m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3268a35c319a4df0bfdc56bbb5ad0ff0 : SUCCESS in 1h 15m 38s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/37bc02d4581a413b8348a9455d5c4064 : FAILURE in 58m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/a7e3b3786b7a49fbb2eb62d45a82f4bf : FAILURE in 43m 52s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6be04be5d7264185b8a2d9925c69a895 : FAILURE in 1h 40m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6ec21b73e262428da20a725705f665c9 : SUCCESS in 30m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/acb1d82f70084056aaf7ca3208f5938b : SUCCESS in 1h 17m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1bf7c8994dac4534abca7ea55914bb9c : SUCCESS in 40m 56s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/3e2ef4f2f63c480f81909310954aede0 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":5},{"id":"286ac9dcdb4ab04c5bc9e773be4e9080ba93af17","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-04 14:51:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/716848/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/304ca95/ : SUCCESS in 2h 05m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/716848/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f09e913/ : SUCCESS in 1h 31m 39s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"5a568979da2c421cc83754626fcd8d0ea32510c2","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-07 13:41:51.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)\n\nThanks Haixin! Looks good, just spotted some nits. If you can address them, good. If not, I think it\u0027s good to merge anyways.","accounts_in_message":[],"_revision_number":5},{"id":"04263f40c22cc862c5d36d97d3c558b72187026c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-07 23:54:23.000000000","message":"Patch Set 5: Code-Review+2\n\n(6 comments)\n\nThanks, Haixin. I agree with Vicky\u0027s comments. If you can address them, we\u0027ll greatly improve the readability of these tests.","accounts_in_message":[],"_revision_number":5},{"id":"250ebe4e29fccf388712a82efa55c2dff3f0bd8e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-08 01:54:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8facec4146eba0d591fc1432e7a65c906615972d","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-08 01:56:50.000000000","message":"Patch Set 6:\n\n(10 comments)\n\nthank for your review, Victoria Martinez de la Cruz and Goutham Pacha Ravi.\ni have update this patch to make it more readable.","accounts_in_message":[],"_revision_number":6},{"id":"cbf845cf1f1a95c2befaebf1630ef911e69f0bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 03:53:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb9454d21d24047a3f39803d3274793 : SUCCESS in 3m 40s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f64afb24cc2648e4ba77d96fdd33667e : FAILURE in 1h 32m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/304b9e23c03d437c8fcb6e02a592f0ed : FAILURE in 52m 38s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b23ab5beddda4412957adb0b84899e7d : FAILURE in 1h 15m 09s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/32af77790cee4d429dc4a5112b8a6b9a : SUCCESS in 1h 23m 34s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8570f7090654455f8740b6234bde96fc : FAILURE in 1h 05m 04s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59183cedf4aa468c9c0543d9d6e84fa8 : FAILURE in 1h 56m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/b1c108da75bb4e85a8438baef0b9cbe5 : FAILURE in 35m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/24843b6e77644affa6e9adf5849cfa2c : FAILURE in 1h 40m 41s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/05f0cd6f673241d68ffbd6cef3c0da59 : FAILURE in 1h 01m 52s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8e5315b74d114afc8f859047ede87f41 : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":6},{"id":"de477dfe5b1d07694ba43d3700530fcd948563cc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-08 04:36:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)\n\nThanks for the changes, suggestions inline","accounts_in_message":[],"_revision_number":6},{"id":"44959d701280e28da854ae0e0285e142c1ed8b01","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-08 05:49:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"89db459aead177a4b2571b3d5143eb6a4ea4cf00","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-08 05:50:11.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nthank for your review, Goutham Pacha Ravi.","accounts_in_message":[],"_revision_number":7},{"id":"d0173fa1a0b2c6c06021cba88a477b76f1497fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:53:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df572818548d40e9acdc7282b6bbf9e7 : SUCCESS in 3m 36s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0d6640641ea14580a7a80e212b177cc1 : FAILURE in 1h 07m 48s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/92d54a9e3f75478a991891f65b036386 : SUCCESS in 56m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d605b095eb82441e9a16c9ae8b58aaab : FAILURE in 1h 18m 47s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/32f92c715f93417da150883a2776c089 : SUCCESS in 1h 02m 29s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/624c17a2e32e4a6cb364ce01179ac6cf : FAILURE in 46m 52s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fca0a87214547e3a78a02b5133a7959 : FAILURE in 1h 44m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/455d3e63df2545b7ba8b5245f3cb034a : SUCCESS in 27m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e45d42908972431aa996b8fce5139af3 : SUCCESS in 1h 24m 04s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1f9f46f748bd42a29e562cbb0f49c279 : SUCCESS in 46m 50s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/23ad25f41d7e451e9fb1fd3a5dc6200d : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":7},{"id":"d28f4e4ea0d299a8bb710cd46f69bf24274100a6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-08 10:24:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/716848/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/1678032/ : FAILURE in 24m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/716848/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3323c69/ : SUCCESS in 1h 31m 02s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d297f1ee70ab212ac7b1f2d490015e319f7aa786","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-08 22:32:37.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks Haixin!","accounts_in_message":[],"_revision_number":7},{"id":"bd1bfa5b85e2cce0dcf51796afd9ffe3ff824bc0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-08 23:29:45.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)\n\nTemporarily holding back workflow on this patch so as to not cause conflicts with things targeted towards Ussuri cycle Feature Freeze (Apr 9th 2020). I\u0027ll revert this when we\u0027re done with feature freeze. In the meantime, please see comment inline.","accounts_in_message":[],"_revision_number":7},{"id":"3be684640cbba2172311ebd08f485ed812cda460","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-04-09 01:39:58.000000000","message":"Patch Set 7:\n\nthanks Goutham Pacha Ravi.\ni got it, I just hope this doesn\u0027t affect the BP state in launchpad.","accounts_in_message":[],"_revision_number":7},{"id":"19d2f8b4c361317d9cb49707d8ef9a9575a544a8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-10 19:52:48.000000000","message":"Patch Set 7: -Workflow\n\nWe\u0027re past feature freeze, I\u0027ve marked the bp as Implemented.","accounts_in_message":[],"_revision_number":7},{"id":"16c3baec17f415ec60da1b31c8e58c8ff0a0aab3","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-16 15:42:09.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"15039d2c599a1965a164972cc386094b6bb4315d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-16 16:24:43.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4d41773c2dccd3d9639a2508e7cfa94dbdb3ad24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 16:25:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2345fee4aa1e5b1abce22f1017ebc19bc47859e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:40:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64aed32d96c4233a00a7c9aa755d021 : SUCCESS in 3m 41s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c774fae4f9bd4cdda0aa24f3a620f5c3 : SUCCESS in 43m 55s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/f8a73074f0484c1b9baf59e77d3cad0e : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b4e74a27895668a94a7624daad9788c01b0f0b46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:40:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"af6f44ad8268eb55954912328f334bb82cf3bf0a","revisions":{"ebe9884b97cd1b529db7438ec6b38f6dd7dec6fd":{"kind":"REWORK","_number":1,"created":"2020-04-02 06:08:01.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/1"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:04:33.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ebe9884b97cd1b529db7438ec6b38f6dd7dec6fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ebe9884b97cd1b529db7438ec6b38f6dd7dec6fd"}]},"branch":"refs/heads/master"},"234dbb98ac4b528d944992139906f6afc4da3b95":{"kind":"REWORK","_number":2,"created":"2020-04-03 01:58:24.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/2"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-03 01:58:05.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/234dbb98ac4b528d944992139906f6afc4da3b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/234dbb98ac4b528d944992139906f6afc4da3b95"}]},"branch":"refs/heads/master"},"cadf14f28afd3a3fb5ce44cf881f7752cc372cfc":{"kind":"REWORK","_number":3,"created":"2020-04-03 03:45:59.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/3"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-03 03:45:25.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cadf14f28afd3a3fb5ce44cf881f7752cc372cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cadf14f28afd3a3fb5ce44cf881f7752cc372cfc"}]},"branch":"refs/heads/master"},"bd4a9cf858cf51c7eafa2616f7d1ac43b45906de":{"kind":"REWORK","_number":4,"created":"2020-04-03 05:41:45.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/4"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-03 05:41:25.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd4a9cf858cf51c7eafa2616f7d1ac43b45906de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd4a9cf858cf51c7eafa2616f7d1ac43b45906de"}]},"branch":"refs/heads/master"},"296442ab22112f818a36756217289f6b33b0dd79":{"kind":"REWORK","_number":5,"created":"2020-04-04 10:49:03.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/5"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-04 10:48:55.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/296442ab22112f818a36756217289f6b33b0dd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/296442ab22112f818a36756217289f6b33b0dd79"}]},"branch":"refs/heads/master"},"043d146a10d001c4c681bf0ef747ed74dc6e1a9b":{"kind":"REWORK","_number":6,"created":"2020-04-08 01:54:57.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/6"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-08 01:54:35.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/043d146a10d001c4c681bf0ef747ed74dc6e1a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/043d146a10d001c4c681bf0ef747ed74dc6e1a9b"}]},"branch":"refs/heads/master"},"af6f44ad8268eb55954912328f334bb82cf3bf0a":{"kind":"REWORK","_number":7,"created":"2020-04-08 05:49:15.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/48/716848/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/716848/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/716848/7"}}},"commit":{"parents":[{"commit":"da28812242789709b65dd30596aac006b5abe9c4","subject":"Merge \"Add knowledge of storage network to scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28812242789709b65dd30596aac006b5abe9c4"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-02 06:03:09.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-04-08 05:48:48.000000000","tz":480},"subject":"Add function to test user messages query by timestamp","message":"Add function to test user messages query by timestamp\n\n-Updates the user messages tests to test query users messages\n by timestamp.\n-Updates the negative user messages tests to test query users\n messages with invalid time format.\n\nDepends-On: I3d94bac4304c236f275abe7ce31432b0e2384247\nChange-Id: I7a048c754cb703eebee709e35d379a4c51a80b3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/af6f44ad8268eb55954912328f334bb82cf3bf0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/af6f44ad8268eb55954912328f334bb82cf3bf0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
