)]}'
{"id":"openstack%2Fmanila-tempest-plugin~721788","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I39ac26fe035af81fbd4cb54d8096fc5d7afc2b0b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ip_version","hashtags":[],"change_id":"I39ac26fe035af81fbd4cb54d8096fc5d7afc2b0b","subject":"Fix ipv6 skip condition","status":"MERGED","created":"2020-04-22 08:12:23.000000000","updated":"2020-04-22 22:21:35.000000000","submitted":"2020-04-22 22:13:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721788-1587593581988-795c03e7","meta_rev_id":"85ad7b73afcb60beb863d21a1ecd2fd98272813a","_number":721788,"virtual_id_number":721788,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-04-22 22:21:35.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 22:13:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-04-22 15:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-04-22 15:28:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":2,"date":"2020-04-22 14:28:55.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-04-22 15:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 14:28:55.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-22 15:21:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-22 15:28:55.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-04-22 22:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 22:21:35.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21e7ef41935daf2fb508dac8ed11b202c7c43287","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 08:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455031f0a6ae25c2f52e68562544a9d86c261d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 10:10:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b59aea147043b498ad0cb498ee5e3f : SUCCESS in 4m 15s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1b14ba70b4a54216b2c8cbc5f8c8aac4 : FAILURE in 1h 31m 01s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6d95c5e8356940969b5c84e9099b8e8e : SUCCESS in 45m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/413070bcbcde4ef09df0c9e5a262b411 : FAILURE in 1h 20m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d50cfd6df5754bd2a6b7e49dce8318f2 : SUCCESS in 1h 18m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d06994937b2f40c78ba051b1db55f7f6 : FAILURE in 52m 42s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7b794be2ebc404d95b6b522bdbba13c : FAILURE in 1h 50m 48s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e82cea03f03e41498e8b579c54bd7951 : SUCCESS in 24m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/71371e17e77b44008df0e98cd52b47f5 : SUCCESS in 1h 23m 34s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7c4ccbd1193a4f98993d7db4b4e2dc62 : SUCCESS in 1h 02m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/3c24ee707fc146bfb06e7265fd78f87c : RETRY_LIMIT in 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1d57b167006d45180a71570fe8f0fb44b14444c5","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 10:46:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"52b098cadafeb27b53a7ae9249c7346c4730e1a9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-22 11:08:16.000000000","message":"Patch Set 1: Code-Review-1\n\nI can\u0027t understand the commit message as written.\n\nIt should say which change introduced this problem, why the way the condition is being checked is incorrect so that it causes ipv4 tests to be skipped, and what the correct check is.","accounts_in_message":[],"_revision_number":1},{"id":"0e3a72c65b53d499b890b9ef7495cbe62f2aba89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e711fa2b9b04eb49fdb1a0da179e5be : SUCCESS in 5m 01s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/485638b8edc74bd8bd1a7e5298a5689e : SUCCESS in 1h 34m 37s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f9011b7ddd68472f87563c57c378e50b : SUCCESS in 54m 13s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2539da7fc57e453ebf04fb10a7ac5b76 : SUCCESS in 1h 16m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d31595411f994f6896dfbe907e5b721a : FAILURE in 58m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/45ce2fc9beb044deaa4c86890f4777db : FAILURE in 58m 03s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecad0bd8e8b04e0481c49a618039027f : FAILURE in 1h 43m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/85d5900dc5e042999c3d9cf468196f49 : SUCCESS in 27m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f6bab6f68861441894e930381acc4389 : FAILURE in 1h 19m 52s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/43696af0ee1e4ed69b6f7ea1ed794447 : SUCCESS in 45m 13s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c662d29a646d40e3a1b815bc3ef42558 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2d2891f3c5e31d654bfe73646c5a9de23f7c4de0","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 12:55:10.000000000","message":"Patch Set 1:\n\n\u003e I can\u0027t understand the commit message as written.\n \u003e \n \u003e It should say which change introduced this problem, why the way the\n \u003e condition is being checked is incorrect so that it causes ipv4\n \u003e tests to be skipped, and what the correct check is.\n\nThe condition checks if the environment is ipv6 by cls.ipv6_enabled instance method:\n\n@property\ndef ipv6_enabled(self):\n        return self.ip_version \u003d\u003d 6\n\nbecause this condition moved to \u0027skip_checks\u0027 class method by https://review.opendev.org/#/c/715679/\nthe condition is wrong.\n\nipv6_enabled will always return True in the \u0027skip_check\u0027 classmethod.","accounts_in_message":[],"_revision_number":1},{"id":"e45c76ba0db23822895b75f03b1e195ef1f28427","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-22 13:36:14.000000000","message":"Patch Set 1:\n\nLiron, my point is that a reader should be able to understand this from the commit message.  When I say \"I can\u0027t understand\", it means \"One cannot understand by reading the commit message\", not that I personally cannot get in there and figure it out :P","accounts_in_message":[],"_revision_number":1},{"id":"edcf2f48cd870a5e89ce2eee7c24058495df0789","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 13:57:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bd298ce4309a0e744c8d7209f4c607d075e95891","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 13:58:29.000000000","message":"Patch Set 2:\n\nI hope it\u0027s better now","accounts_in_message":[],"_revision_number":2},{"id":"abb7efe08ccc1593dede1624d93aab4ccfb9be39","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-22 14:12:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nBetter :D","accounts_in_message":[],"_revision_number":2},{"id":"cbf2da7e01cb61070ac9777c14242c51efaf1b1b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 14:16:16.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"63c796aeedb9cce5517f413d10a000f25dca3e66","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-04-22 14:17:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e22f2da72b43043741debfbd42b2bdcb72013fc","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-22 14:28:55.000000000","message":"Patch Set 3: Code-Review+2\n\nWell we all have stronger and less strong sides, that\u0027s why we\u0027re a team!","accounts_in_message":[],"_revision_number":3},{"id":"7716848d615b851f8261b525a0de8fc1616a607e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-22 15:21:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nGreat find! Thanks for noticing and fixing this, we should look beyond the voting LVM job to notice issues like this - a good candidate that caught this in the original patch is zfsonlinux. However, since skips are not treated as failures, it makes things harder.","accounts_in_message":[],"_revision_number":3},{"id":"0eaeb70e63743383e52c16d207cdd109256a4c7e","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2020-04-22 15:28:55.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good, thanks","accounts_in_message":[],"_revision_number":3},{"id":"c12c332080c2e28093cd27954a3f23d4b27e0713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:55:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db6b22557d447dea8b4d3ff0c3c45b1 : SUCCESS in 4m 53s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/adc7ea31d24c41478612eaa7455c1e65 : FAILURE in 1h 10m 41s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/101d6b764d3e4f3b85cc55a270b6e9a6 : SUCCESS in 51m 17s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/09aba9237f474c4da9102d948403da10 : FAILURE in 1h 22m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0263c79c5f9d46f0b60322797aaaa467 : SUCCESS in 1h 31m 02s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d41d5277452245cb87b6f2f95535b0fe : FAILURE in 1h 33m 52s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/052a087f236a4dccaf2c33d71d6931dc : FAILURE in 1h 59m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f79cb2a3cae74dcb95e6bcaf54468872 : SUCCESS in 58m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4fb196f10723415eac457dcae1c07287 : FAILURE in 1h 59m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/706b87e1f5db4ce5b8395bdb287a7f52 : SUCCESS in 54m 15s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/fdefcdf012a04aa88ced90f37119cb65 : FAILURE in 1h 43m 53s","accounts_in_message":[],"_revision_number":3},{"id":"a4e2acb33a11112c50f848fd65ba6d975404af38","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-22 19:07:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f0603591bf9ad419616232844b4fb45062aeddaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:54:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf428871d854e808db2fe6c7528afe1 : SUCCESS in 4m 47s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0ee7ee75d729435780da36efd90419e4 : FAILURE in 1h 36m 28s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/60840a28a69e463c835104d478ea7366 : SUCCESS in 48m 54s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/b0db39b1414d4d6a8fc11c1d7a09f484 : SUCCESS in 1h 12m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7d0680956f22434b966c4bda44387772 : SUCCESS in 42m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8bd3289550474ca1a81c471c50c710b9 : FAILURE in 38m 51s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08fb520c2bc740058a58ffcc5d9435ab : FAILURE in 1h 40m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7d98a231ac19487eab6ef21a56565d9b : SUCCESS in 28m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ff46a630e9f64e04b1f630b24a2e3f4e : SUCCESS in 1h 11m 45s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/03d8b3f698224049853d838c3071af9c : SUCCESS in 49m 26s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/7a46f718d11e4160843bcda3b36b08b9 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d073f1095670be68036a579bc7a33541d60a831e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:54:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"352103241458996d9b92e07babfbe6ec95e23956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:13:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2368bfa0c846ea963b80f65241bdeb : SUCCESS in 4m 52s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/aa4f417a811144ce9b27afb7ecaa3afd : SUCCESS in 42m 29s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/15a7ed58f7d44b229260a6fe77fd3fe8 : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":3},{"id":"cdcf89f1bbc0dab7008b7dd0978b6a64e6d4a291","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:13:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d1e5f113fa922137744fc533e57791ee0ff199eb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-22 22:21:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/721788/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/66d898e/ : FAILURE in 41m 06s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/721788/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ed42de5/ : FAILURE in 1h 25m 20s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"06665e525844f54e6fb5383e235b99a809fe9331","revisions":{"0a88a689f93c1965abf4b326e87adb61ea8326e7":{"kind":"REWORK","_number":1,"created":"2020-04-22 08:12:23.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/88/721788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/721788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/1"}}},"commit":{"parents":[{"commit":"c8af9190d2356dfbbe3c1458344cecaba31d72a2","subject":"Merge \"Move skipping conditions under skip_checks method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c8af9190d2356dfbbe3c1458344cecaba31d72a2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-22 07:00:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-22 07:00:31.000000000","tz":180},"subject":"Fix ipv6 skip condition","message":"Fix ipv6 skip condition\n\nipv6 skip condition moved under \"skip_checks\"\nclassmethod.\nSo the condition should be adjusted accordingly.\nCurrently, ipv4 tests are skipped because of\nthis condition.\n\nChange-Id: I39ac26fe035af81fbd4cb54d8096fc5d7afc2b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a88a689f93c1965abf4b326e87adb61ea8326e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a88a689f93c1965abf4b326e87adb61ea8326e7"}]},"branch":"refs/heads/master"},"0b46754d77d2651f0cf14bfc291398fc5d49f9c3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-22 13:57:47.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/88/721788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/721788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/2"}}},"commit":{"parents":[{"commit":"c8af9190d2356dfbbe3c1458344cecaba31d72a2","subject":"Merge \"Move skipping conditions under skip_checks method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c8af9190d2356dfbbe3c1458344cecaba31d72a2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-22 07:00:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-22 13:57:06.000000000","tz":180},"subject":"Fix ipv6 skip condition","message":"Fix ipv6 skip condition\n\nipv6 skip condition moved under \"skip_checks\"\nclassmethod by https://review.opendev.org/#/c/715679/.\nThe condition checks if the environment is ipv6 by\n\"ipv6_enabled\" instance method.\n\nChange-Id: I39ac26fe035af81fbd4cb54d8096fc5d7afc2b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0b46754d77d2651f0cf14bfc291398fc5d49f9c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0b46754d77d2651f0cf14bfc291398fc5d49f9c3"}]},"branch":"refs/heads/master"},"06665e525844f54e6fb5383e235b99a809fe9331":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-22 14:16:16.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/88/721788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/721788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/721788/3"}}},"commit":{"parents":[{"commit":"c8af9190d2356dfbbe3c1458344cecaba31d72a2","subject":"Merge \"Move skipping conditions under skip_checks method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c8af9190d2356dfbbe3c1458344cecaba31d72a2"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-22 07:00:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-04-22 14:15:22.000000000","tz":180},"subject":"Fix ipv6 skip condition","message":"Fix ipv6 skip condition\n\nipv6 skip condition moved under \"skip_checks\"\nclassmethod by https://review.opendev.org/#/c/715679/.\nHere we correct the condition that checks whether IPv6\nis enabled.\n\nChange-Id: I39ac26fe035af81fbd4cb54d8096fc5d7afc2b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/06665e525844f54e6fb5383e235b99a809fe9331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/06665e525844f54e6fb5383e235b99a809fe9331"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
