)]}'
{"id":"openstack%2Fmanila-tempest-plugin~723114","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0ee61c3197547201f10340d388baf5e076494be5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1875180","hashtags":[],"change_id":"I0ee61c3197547201f10340d388baf5e076494be5","subject":"Removing duplicated availability zones from tests","status":"MERGED","created":"2020-04-25 20:21:29.000000000","updated":"2020-04-26 22:00:46.000000000","submitted":"2020-04-26 21:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723114-1587935240976-8eb4b95b","meta_rev_id":"85e56c86014f5fec9934b4d5c6694424b3ec5497","_number":723114,"virtual_id_number":723114,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-04-26 18:13:32.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-04-26 22:00:46.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-26 21:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-04-26 19:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-04-26 19:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2020-04-26 20:06:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-26 18:14:15.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 18:14:15.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-26 19:28:01.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-26 20:06:24.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-26 21:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-26 22:00:46.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1e1ea144af1a44a4c656db2457e2250fc7c561e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-25 20:21:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7f1a49b98af98278eb9e41202b78a4425e5ed73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 22:53:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0be14f3276e4e51b7a5dbbada845fa3 : SUCCESS in 3m 03s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/9a2b7b92892a4ce29ad3671f9553240f : FAILURE in 1h 11m 31s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4966040e472741ecb2dcca3ea2079f07 : SUCCESS in 54m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/768cf9355f374202a3b81767d86f1188 : FAILURE in 1h 10m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/c9be901f8a60417cb329fb84a6ef7f8b : SUCCESS in 41m 31s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8f64deb42a5b49e29f0972d789c45dba : FAILURE in 43m 40s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9a2d5060eaa4febaa06914264057d85 : FAILURE in 1h 32m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e7dc53ebc1484aa8959ee3468257baaa : SUCCESS in 23m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bd71f2772eb9407485e79ba9dcc44755 : FAILURE in 1h 20m 50s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/33926c5c6569418b867257308623d3ac : FAILURE in 45m 29s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4d785913a5b64107b6312f0e3605b904 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b610a3c8762614c89dda07a04d48e44d28de71d0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-26 00:26:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0a301aed3066874587969475a39b9739c9eeb29c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-26 00:34:38.000000000","message":"Patch Set 1:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":1},{"id":"a432a28a28a6f167a944e01f02d5c0d2180045d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 01:59:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9925f0ab5445f68cc498b89098e7eb : SUCCESS in 4m 06s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b39965f1aa8f4ddc9e698e92a86c05e1 : SUCCESS in 1h 26m 38s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6c07ecd757974b4b952aa09597b517e7 : SUCCESS in 46m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f1bd0c6a01ef433db101e2016d8f6b36 : FAILURE in 1h 07m 25s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0b4eb9304533474d8ee080cc5580395a : SUCCESS in 52m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/cb68c35e6c7d45a58a2d6bfe427fa7d5 : FAILURE in 49m 57s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/078955a61b3a436593c809becd9ca246 : FAILURE in 1h 33m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0b3d71b6347f46a287993103ea60e4bf : SUCCESS in 30m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/568dc7cf47a844a18b583d4a9c99c9d5 : SUCCESS in 1h 07m 05s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6619556c8b55419a965ac2a1c5ed8212 : SUCCESS in 51m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/41f131c1846842ecb5bbda120990d0be : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b978b04237e342561a1f60ddcba8783577271e58","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-26 03:50:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/723114/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/af92726/ : SUCCESS in 2h 15m 06s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/723114/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f210773/ : FAILURE in 1h 23m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd52ecfa01a31afa13243759c33bdaef94e1b231","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-26 13:04:16.000000000","message":"Patch Set 1:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":1},{"id":"47f7349d14a3f16375300dc68d0b86d0d12d1cde","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-26 15:16:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/723114/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f0b9e13/ : SUCCESS in 2h 12m 09s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/723114/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f869b85/ : SUCCESS in 1h 26m 09s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9e4009c58452ce880a2a14da6e94c4822bbea423","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-26 18:11:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"03a72021d765203edf8be61db0a85d361694f405","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-26 18:13:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"408712db32e2efaf95d413121bb20b0673a54666","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-26 19:28:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dce5221ce4d68fd34ba1d7e7ad2ae4d5917314a8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-26 19:38:12.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"070159d22f1795598809ff61cc948f775c9299e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 20:05:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57cdbc4b61ec49b29146007180f627ec : SUCCESS in 3m 22s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3a490177a2434a50bb0e5f6837341622 : SUCCESS in 1h 37m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/74a0eba3e9ca4401b81554a6ec2482c8 : SUCCESS in 1h 06m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a8bd58c260fe4b93b182ddd93f80595f : FAILURE in 1h 16m 52s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/021f787af3e348bc95ed70f9792e0410 : SUCCESS in 1h 07m 02s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3946b4f4950947f7b99af35fc41bd701 : FAILURE in 48m 15s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1772006a7d74ee6a94002d4cffef56e : FAILURE in 1h 52m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d8cd18d16bab48e692128def07da5856 : SUCCESS in 30m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/45d411a4119e4ce881ac7f25f2a270e4 : FAILURE in 1h 38m 16s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/574a94ad658f469f961bd1802096dfa2 : SUCCESS in 48m 09s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e82d472772d04f828c3055a6a33f14f1 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5e442a9d9589e017e67a6ada34342351833be9c1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-26 20:06:24.000000000","message":"Patch Set 2: Workflow+1\n\nSeems good to go :)","accounts_in_message":[],"_revision_number":2},{"id":"aba41765d2fd4b982e7078c4f1f48d406dbd82cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 20:06:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2109de50771439922f668e1eb229c874c2f2abb3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-26 20:59:45.000000000","message":"Topic changed from fix_add_multiple_rep to bug/1875180","accounts_in_message":[],"_revision_number":2},{"id":"c7033eea35ab2449b09a672245f87cf921b0fe8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 21:07:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336bfadb3fb5430da2dcbecda1485bf6 : SUCCESS in 2m 48s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b762a09c819c4249b2d9dc8a59978e01 : SUCCESS in 39m 08s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/97d0a42e92d24002b4487ef62651758f : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":2},{"id":"bde0602ca090bca01c0a1e7f1e2c72449bc91a5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 21:07:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"85e56c86014f5fec9934b4d5c6694424b3ec5497","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-26 22:00:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/14/723114/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/36096e1/ : FAILURE in 1h 55m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/14/723114/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5a51f20/ : SUCCESS in 1h 21m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"161f180e75fbf3b37ba7585aa19573076eda3069","revisions":{"ef4540aee67ed53c1f6d855ee9352027c6027029":{"kind":"REWORK","_number":1,"created":"2020-04-25 20:21:29.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/14/723114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/723114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/1"}}},"commit":{"parents":[{"commit":"d3432669d2e782734017cc9308e9fd34768f5bb7","subject":"Fix floatingip allocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3432669d2e782734017cc9308e9fd34768f5bb7"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-04-25 20:18:14.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-04-25 20:21:27.000000000","tz":-180},"subject":"WIP - Testing removing duplicated azs","message":"WIP - Testing removing duplicated azs\n\nChange-Id: I0ee61c3197547201f10340d388baf5e076494be5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef4540aee67ed53c1f6d855ee9352027c6027029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef4540aee67ed53c1f6d855ee9352027c6027029"}]},"branch":"refs/heads/master"},"161f180e75fbf3b37ba7585aa19573076eda3069":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-26 18:11:15.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/14/723114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/14/723114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/14/723114/2"}}},"commit":{"parents":[{"commit":"d3432669d2e782734017cc9308e9fd34768f5bb7","subject":"Fix floatingip allocation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3432669d2e782734017cc9308e9fd34768f5bb7"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-04-25 20:18:14.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-04-26 18:10:56.000000000","tz":0},"subject":"Removing duplicated availability zones from tests","message":"Removing duplicated availability zones from tests\n\nThis patch remove duplicated availability zones when retrieving them\nfrom service list.\n\nCloses-Bug: #1875180\nChange-Id: I0ee61c3197547201f10340d388baf5e076494be5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/161f180e75fbf3b37ba7585aa19573076eda3069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/161f180e75fbf3b37ba7585aa19573076eda3069"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
