)]}'
{"id":"openstack%2Fmanila-tempest-plugin~724466","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306","project":"openstack/manila-tempest-plugin","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306","subject":"Add native zuulv3 job for zfsonlinux","status":"MERGED","created":"2020-04-30 05:18:45.000000000","updated":"2020-05-05 12:05:02.000000000","submitted":"2020-05-05 12:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724466-1588680302074-57c03af8","meta_rev_id":"011bb6fa89da3322019c0126cb2f77d762af6b85","_number":724466,"virtual_id_number":724466,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-05 04:12:06.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-05-05 10:42:44.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 12:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 10:46:25.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-05 10:46:25.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 21:38:28.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-05-04 21:57:58.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-05 10:42:44.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-05 10:46:25.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-05-05 12:05:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f73f5cad537bff3f27797920c7a0823bc393e6cf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 05:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e77cbede1e86161f0776390086ee72c4fe5ab96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 05:19:06.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Invalid Ansible variable name \u0027test-config\u0027 for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-tempest-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-base\n      vars:\n        tempest_test_regex: \u0027(^manila_tempest_tests.tests)(?\u003d.*\\[.*\\bbackend\\b.*\\])\u0027\n        devstack_localrc:\n          SHARE_DRIVER: manila.share.drivers.zfsonlinux.driver.ZFSonLinuxShareDriver\n          MANILA_ENABLED_BACKENDS: london,paris\n          MANILA_BACKEND1_CONFIG_GROUP_NAME: london\n          MANILA_BACKEND2_CONFIG_GROUP_NAME: paris\n          MANILA_SHARE_BACKEND1_NAME: LONDON\n          MANILA_SHARE_BACKEND2_NAME: PARIS\n          MANILA_OPTGROUP_london_driver_handles_share_servers: false\n          MANILA_OPTGROUP_paris_driver_handles_share_servers: false\n          MANILA_SHARE_MIGRATION_PERIOD_TASK_INTERVAL: 1\n          MANILA_REPLICA_STATE_UPDATE_INTERVAL: 60\n          MANILA_ZFSONLINUX_SERVICE_IP: 127.0.0.1\n          MANILA_ZFSONLINUX_USE_SSH: true\n          MANILA_CONFIGURE_DEFAULT_TYPES: false\n          MANILA_USE_SCHEDULER_CREATING_SHARE_FROM_SNAPSHOT: true\n        devstack_services:\n          mysql: false\n          postgresql: true\n        test-config:\n          $TEMPEST_CONFIG:\n            share:\n              run_driver_assisted_migration_tests: true\n              run_host_assisted_migration_tests: true\n              run_replication_tests: true\n              run_manage_unmanage_snapshot_tests: true\n              run_manage_unmanage_tests: true\n              run_multiple_share_replicas_tests: false\n              run_create_share_from_snapshot_in_another_pool_or_az_tests: true\n              backend_replication_type: readable\n              enable_protocols: nfs\n              capability_storage_protocol: NFS\n              build_timeout: 180\n              enable_ip_rules_for_protocols: nfs\n              multitenancy_enabled: False\n              backend_names: LONDON,PARIS\n              multi_backend: true\n              image_password: manila\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 50, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e3ad7559183a1b973c747ba504cfd9d35b2ca0f4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 05:22:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd0be0064ee1a44ccc519b5942f84cea71d1574d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:40:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-zfsonlinux https://zuul.opendev.org/t/openstack/build/0a70a52e1b72436bb5f748cd882ad2ee : FAILURE in 59m 27s\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c96398f2070d42f5aabe74b99d39ca98 : FAILURE in 1h 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"600c5ba678eaf4790865597f3809ba85bb3b3845","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 18:53:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"53e272b966196c830bac2b15da4cb4445e2a9628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 19:58:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-zfsonlinux https://zuul.opendev.org/t/openstack/build/618a0b8e087b41af8e110fef14c747b3 : FAILURE in 40m 12s\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f01ae71aa3b24ff09c9952683b40b55b : SUCCESS in 1h 03m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c4205780829a7705dfbc9144acd4d6db4600eba","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-30 22:33:22.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"29fae91cf5f0a02aa9e3153b2e8d779b3de0c927","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 22:56:02.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d727da372d440a02db45923bf097ab457d5d4224","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 23:00:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f4e9e093824314cb43d13917de7cc1deac0f486f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 23:00:19.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Job manila-tempest-tests-base not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-tempest-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-tests-base\n      vars:\n        tempest_test_regex: \u0027(^manila_tempest_tests.tests)(?\u003d.*\\[.*\\bbackend\\b.*\\])\u0027\n        devstack_localrc:\n          SHARE_DRIVER: manila.share.drivers.zfsonlinux.driver.ZFSonLinuxShareDriver\n          MANILA_ENABLED_BACKENDS: london,paris\n          MANILA_BACKEND1_CONFIG_GROUP_NAME: london\n          MANILA_BACKEND2_CONFIG_GROUP_NAME: paris\n          MANILA_SHARE_BACKEND1_NAME: LONDON\n          MANILA_SHARE_BACKEND2_NAME: PARIS\n          MANILA_OPTGROUP_london_driver_handles_share_servers: false\n          MANILA_OPTGROUP_paris_driver_handles_share_servers: false\n          MANILA_SHARE_MIGRATION_PERIOD_TASK_INTERVAL: 1\n          MANILA_REPLICA_STATE_UPDATE_INTERVAL: 60\n          MANILA_ZFSONLINUX_SERVICE_IP: 127.0.0.1\n          MANILA_ZFSONLINUX_USE_SSH: true\n          MANILA_CONFIGURE_DEFAULT_TYPES: true\n          MANILA_USE_SCHEDULER_CREATING_SHARE_FROM_SNAPSHOT: true\n        devstack_services:\n          mysql: false\n          postgresql: true\n        devstack_local_conf:\n          test-config:\n            $TEMPEST_CONFIG:\n              share:\n                default_share_type_name: default\n                run_driver_assisted_migration_tests: true\n                run_host_assisted_migration_tests: true\n                run_replication_tests: true\n                run_manage_unmanage_snapshot_tests: true\n                run_manage_unmanage_tests: true\n                run_multiple_share_replicas_tests: false\n                run_create_share_from_snapshot_in_another_pool_or_az_tests: true\n                backend_replication_type: readable\n                enable_protocols: nfs\n                capability_storage_protocol: NFS\n                build_timeout: 180\n                enable_ip_rules_for_protocols: nfs\n                multitenancy_enabled: False\n                backend_names: LONDON,PARIS\n                multi_backend: true\n                image_password: manila\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 50, column 3","accounts_in_message":[],"_revision_number":4},{"id":"98f7a916f16341611d410492ba5a75df385739ff","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-30 23:04:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9308205208326452075856cc980acbd5dcfcd33a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 23:13:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65caddab9880625e2f16e424168edd412ea55e32","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 23:13:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10a8e79cf6c0de0e41faf7dac3d6410f5653b1c0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 23:14:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"27fe723c56b3b2d4440cd454f17672045fcaefbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 23:14:36.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Job manila-tempest-zfsonlinux not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - manila-tempest-zfsonlinux\n          - manila-tempest-dsvm-postgres-zfsonlinux:\n              voting: false\n\n  in \"openstack/manila-tempest-plugin/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":5},{"id":"6c509a1d3864d9ada341790a28cc26eb6ebccdd4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 23:15:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b72f0e86923515b3994910de4002e38b8113b503","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 23:15:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cb2a56fd5f485ddd315f9ea70e60601d61b66f3c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-01 00:28:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a098f6a3bd9bff08c573aa7d52404c70d9efb61f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-01 00:29:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8040d1d9d3e0c766e967eaf694c498ec3484f603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 02:22:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88c164e4ceb74d7fae05052a63144d59 : SUCCESS in 4m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2fde24ed8fa8460aa1ae008da17018ce : SUCCESS in 1h 13m 38s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/685508668d864cddade2d3975e89d047 : FAILURE in 1h 23m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/700c3d7bd7ee4da78ddc041c40589312 : FAILURE in 1h 12m 29s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7f93052905314384aa67f37e9c7090ac : SUCCESS in 58m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/fe60c3a4c42f497284e89240acca98e1 : FAILURE in 59m 00s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc154175ad0a48148d91f260c91ff80b : FAILURE in 1h 51m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6a339cc83178426cb8cfd45c57ccaadd : SUCCESS in 27m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c136ccb3ec8948e7a4056da0c12f2f00 : SUCCESS in 1h 22m 51s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d51b28bececa4fe29c4f030cc73527ff : FAILURE in 49m 58s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e9b899c53ea5466aa28d9ab2b4d7ef6f : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":8},{"id":"fcb489d4bc044a6c3c646832f661514face43a15","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-01 04:55:00.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"4684c30052593773d9031fac51c1fc30d0458fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 06:59:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87838633cd4d4c6f9bcd6f46b34d3975 : SUCCESS in 3m 25s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bddae9b086694676acbb2ded5c26e3a1 : SUCCESS in 1h 06m 38s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a6e15456fa784973ba743a2d9ca2f750 : FAILURE in 1h 14m 28s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/9dfb59b1867b48bca9254076e4629165 : FAILURE in 52m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/10647ca0943b4879afeefa10e196ac7f : SUCCESS in 49m 15s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/13a3a7574e604a40b0d243c7bed0dd13 : FAILURE in 44m 11s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b66680d267684cefbcec6c7ad6e5e14d : FAILURE in 1h 51m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/42b882c056394c6593863f0e7332f306 : SUCCESS in 26m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8720c0358c4c4023a1bbdad25f66e3f1 : SUCCESS in 1h 29m 12s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5f4f33466b854617b20d8538b8203c59 : SUCCESS in 44m 29s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/8e45b3d82e3f4e14a8ebf0037bdd7873 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":8},{"id":"75c3c2dccabc3c66b7270850dabfa1deee5ca61a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-01 12:34:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/66/724466/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/37f2bfe/ : SUCCESS in 3h 16m 41s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/66/724466/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/704d316/ : SUCCESS in 1h 34m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"a89dc80e5646319264767553bf966ccf891d285d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-04 18:56:14.000000000","message":"Patch Set 8: Code-Review+2\n\nVery nice!","accounts_in_message":[],"_revision_number":8},{"id":"bbc581e3e3a33f9b7abd47f5915d8e5de6846464","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-04 21:38:28.000000000","message":"Patch Set 8: Code-Review+2\n\nThis looks so good Goutham!","accounts_in_message":[],"_revision_number":8},{"id":"172c67a038ffbbb6c1a3f2fc446d15aa3162956a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 21:57:58.000000000","message":"Patch Set 8: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":8},{"id":"360603b4807803e7018cd1f444634b5562fd56ec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-04 23:22:11.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e97849cf9377f67417216993e7b4d2ab5e54a718","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 04:11:46.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"244171ca14d0a95a16a9c0c54b9d23b36291718f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 04:12:06.000000000","message":"Patch Set 9:\n\nJust popped off the dependencies, since they\u0027ve merged on the master branch of manila.","accounts_in_message":[],"_revision_number":9},{"id":"d9e02d6c5ff91651febba344eaeb977d00f2e543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 05:47:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e749577c6040149524bb97402484d5 : SUCCESS in 3m 47s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d85c6dff5bc14c75923346f5b7c0aad5 : SUCCESS in 1h 05m 34s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f953606e08894798b59137737f85bd7c : SUCCESS in 1h 33m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0cd38b813e3c4f5e9543d1802115d54a : FAILURE in 52m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/10f84050d27149d1ae8f1a7fbf097ff1 : SUCCESS in 51m 52s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/202ebb152f4b4da7b5ef61425c06193e : FAILURE in 42m 39s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75aa979e1a5c418ba24d094766a083ec : FAILURE in 1h 26m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/8252a0b248ee4b51b405a3507e55a199 : SUCCESS in 27m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b6c9feb04e74229bafcf23d1872be4d : SUCCESS in 1h 13m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4d1da5495a3d433d9b9725977f9e0185 : SUCCESS in 44m 22s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/75754a3739224988a8ca9e7a6e3205a9 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":9},{"id":"0e8035c70785c86c39d64c57e2bd4b756ad8650f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-05 10:42:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/66/724466/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/820d54c/ : FAILURE in 50s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/66/724466/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1d9e4e7/ : SUCCESS in 1h 27m 07s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"c78e298697fb65b7118f7d3e8c2b067a36db906b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-05 10:46:25.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nMerging: the only diff between this and the last one -- which has two +2s -- is the cleaned up commit message.","accounts_in_message":[],"_revision_number":9},{"id":"0a9a746e2e74b41346058baa82cb3116bb79cf97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 10:46:40.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7a8e832f1b8b5ccb0411c426e76f0fec32afba46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:05:01.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a73fb91e62ac43849c88eb712c38db6e : SUCCESS in 4m 41s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d9a6e1afe3a143fbaf711a641de73589 : SUCCESS in 48m 49s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ac1bef226192421594f7980f70515932 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":9},{"id":"2b49b390a4de225333a7c0892630d6f01e077cd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:05:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5dd6301d5ba0037092ff0f64b5dba899734f0d94","revisions":{"bf6476fe8fba415c29896bb6a709cf42740d1c7d":{"kind":"REWORK","_number":1,"created":"2020-04-30 05:18:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/1"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bf6476fe8fba415c29896bb6a709cf42740d1c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bf6476fe8fba415c29896bb6a709cf42740d1c7d"}]},"branch":"refs/heads/master"},"7dea51ce012c4f4ab942959a65b1c8345807062f":{"kind":"REWORK","_number":2,"created":"2020-04-30 05:22:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/2"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:22:33.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7dea51ce012c4f4ab942959a65b1c8345807062f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7dea51ce012c4f4ab942959a65b1c8345807062f"}]},"branch":"refs/heads/master"},"90a8cb17be18bbc191718e3d415636b380b02308":{"kind":"REWORK","_number":3,"created":"2020-04-30 18:53:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/3"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 18:52:39.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90a8cb17be18bbc191718e3d415636b380b02308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/90a8cb17be18bbc191718e3d415636b380b02308"}]},"branch":"refs/heads/master"},"9884c2f43d9b5a616742a9f421c6295b248f9a88":{"kind":"REWORK","_number":4,"created":"2020-04-30 23:00:02.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/4"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 22:59:28.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9884c2f43d9b5a616742a9f421c6295b248f9a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9884c2f43d9b5a616742a9f421c6295b248f9a88"}]},"branch":"refs/heads/master"},"170b258b55e51d75568243324b93a46a9028e8ae":{"kind":"REWORK","_number":5,"created":"2020-04-30 23:14:12.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/5"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 23:14:03.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/170b258b55e51d75568243324b93a46a9028e8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/170b258b55e51d75568243324b93a46a9028e8ae"}]},"branch":"refs/heads/master"},"835f8804e717806189699773995f2f30de04b904":{"kind":"REWORK","_number":6,"created":"2020-04-30 23:15:47.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/6"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 23:15:40.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/835f8804e717806189699773995f2f30de04b904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/835f8804e717806189699773995f2f30de04b904"}]},"branch":"refs/heads/master"},"990c7eadc45a4057dc8d729c5715632494854ec1":{"kind":"REWORK","_number":7,"created":"2020-05-01 00:28:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/7"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-01 00:27:44.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/990c7eadc45a4057dc8d729c5715632494854ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/990c7eadc45a4057dc8d729c5715632494854ec1"}]},"branch":"refs/heads/master"},"fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972":{"kind":"REWORK","_number":8,"created":"2020-05-01 00:29:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/8"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-01 00:28:59.000000000","tz":-420},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Ic9cad47662f1edf2e5c710dbe64d580bc5f01d44\nDepends-On: Iea90e3f04ae05e5783c3163bbf2d2dabd128c7b5\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972"}]},"branch":"refs/heads/master"},"5dd6301d5ba0037092ff0f64b5dba899734f0d94":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-05-05 04:11:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/724466/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/66/724466/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/66/724466/9"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-30 05:18:14.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 04:11:38.000000000","tz":0},"subject":"Add native zuulv3 job for zfsonlinux","message":"Add native zuulv3 job for zfsonlinux\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: Icc5d16c61b4e4cabf554b2e8bc9f290b6a6d5306\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
