)]}'
{"id":"openstack%2Fmanila-tempest-plugin~725438","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2981f786d21882579c2bbc569615843d253bf5dc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I2981f786d21882579c2bbc569615843d253bf5dc","subject":"Add native zuulv3 job for lvm","status":"MERGED","created":"2020-05-05 00:34:57.000000000","updated":"2020-05-07 13:34:09.000000000","submitted":"2020-05-07 13:34:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725438-1588858449356-ae45bbcb","meta_rev_id":"f35fd41e5ce9435d0049fb0d8b356bdbe8c474e1","_number":725438,"virtual_id_number":725438,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-05-07 11:13:42.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 13:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-07 11:01:30.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-05-07 11:43:52.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-07 12:26:15.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 05:11:08.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-05-06 18:01:50.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-06 18:02:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-07 11:13:42.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-07 11:43:52.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-05-07 12:26:15.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-05-07 13:34:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b49e5a8fea4336457e82f67c8d21ddeb6cfd5cec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 00:34:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ed1b67503a44a942234854b562dae5df640e103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 02:24:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/877bf29db3034b549be74899cf300293 : FAILURE in 4m 54s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9c3ba83ac25d4cadadd2cf459c840886 : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7f343d94445d3cf337b3acb52994d80bef800fdc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 04:10:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63358e358da05b6ceed7715c8bd88024ef0f672f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 04:12:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1aa128891e6b2034c14f3f362028cda8f011e88c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 05:27:03.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2a0e0d62035399a6aed1ce2c8390c21a0d3beff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 07:24:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e2cf8b64b54565b8afd0edcc64c581 : SUCCESS in 4m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8639b404620a4e8c9d00118becf3df19 : SUCCESS in 1h 00m 56s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7d0d4443c9af4d97b76c13737d48e0d2 : SUCCESS in 40m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a00297807b794a9e873878f8571ff81b : FAILURE in 1h 06m 23s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/256c00449221498990720b3c872d1581 : FAILURE in 1h 49m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0762f700ffed434087a9d95fa944953d : FAILURE in 51m 27s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/fdc17455ae194418bb544a086f40ae26 : SUCCESS in 53m 43s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/282f5a573734417aaec3968def902951 : FAILURE in 50m 46s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/937d423cca014175b8ff8ee9d28ce08a : FAILURE in 1h 52m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/492022c7c6d44ba6b29d09bd50a886c4 : SUCCESS in 26m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ff517e7c6bb346689318e72aa6f5da32 : SUCCESS in 1h 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f85a954944d157e7f81bdefc6afb0796e74444b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-05 10:50:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/725438/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e9381d2/ : FAILURE in 3m 02s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/725438/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/c6ae74b/ : SUCCESS in 1h 31m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"87d996a1da72f9e8d77c1c344b287a80fbebdfdc","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-06 11:54:31.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nAwesome!  Pls. see question inline though,","accounts_in_message":[],"_revision_number":4},{"id":"73debf32c2f85269346173d36cf8ddb9c2d7a42e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 14:24:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"345deb4888cede9c70446e72100f4971c805451f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 14:25:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ffb7d9f2c684655896e981799b6f0024f1b049bd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 17:41:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"324fcc656bc004e9d7180d183c9563228e858ec7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 17:45:47.000000000","message":"Patch Set 5:\n\nHi Ghanshyam, \n\nI\u0027m restructuring the old legacy \"LVM job\" in manila to be a zuulv3 native style job. I\u0027m making this new job derive from devstack-tempest-ipv6 because I\u0027d like to keep coverage to the test you added in the Train cycle to ensure we\u0027re deploying an IPv6 service/control plane. I\u0027m however, dropping the \"ipv6-only\" naming from the job. What are your views on that? In manila, we care that this job does data path testing with IPv4 and IPv6; so calling it \"ipv6-only\" is a bit confusing. What\u0027s the reasoning behind this naming? Wonder if we can just drop it and still be okay. \n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"61f6dbeb0d3d3f1ea86c42ca6846d7c8b4bf9561","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-06 17:59:34.000000000","message":"Patch Set 5:\n\n\u003e Hi Ghanshyam,\n \u003e \n \u003e I\u0027m restructuring the old legacy \"LVM job\" in manila to be a zuulv3\n \u003e native style job. I\u0027m making this new job derive from\n \u003e devstack-tempest-ipv6 because I\u0027d like to keep coverage to the test\n \u003e you added in the Train cycle to ensure we\u0027re deploying an IPv6\n \u003e service/control plane. I\u0027m however, dropping the \"ipv6-only\" naming\n \u003e from the job. What are your views on that? In manila, we care that\n \u003e this job does data path testing with IPv4 and IPv6; so calling it\n \u003e \"ipv6-only\" is a bit confusing. What\u0027s the reasoning behind this\n \u003e naming? Wonder if we can just drop it and still be okay.\n \u003e \n \u003e Thanks!\n\ndevstack-tempest-ipv6 base job set SERVICE_IP_VERSION\u003d6 so that every openstack service and DB etc listen on ipv6 address not on ipv4. If you see on the log, you will be able to see manila also communicate on ipv6. I have not checked the tests testing the IPV4 data path but if that is there then we can verify it.","accounts_in_message":[],"_revision_number":5},{"id":"edf1a3dfef3b0fd097bb8bccc6fbc8f69c741546","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-06 18:01:50.000000000","message":"Patch Set 5:\n\n(1 comment)\n\njust a small issue (if I read it correctly):","accounts_in_message":[],"_revision_number":5},{"id":"85703c0ab3577359bdfc2e1953b5d4a99ea65857","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-06 18:02:46.000000000","message":"Patch Set 5:\n\n\u003e \u003e Hi Ghanshyam,\n \u003e \u003e\n \u003e \u003e I\u0027m restructuring the old legacy \"LVM job\" in manila to be a\n \u003e zuulv3\n \u003e \u003e native style job. I\u0027m making this new job derive from\n \u003e \u003e devstack-tempest-ipv6 because I\u0027d like to keep coverage to the\n \u003e test\n \u003e \u003e you added in the Train cycle to ensure we\u0027re deploying an IPv6\n \u003e \u003e service/control plane. I\u0027m however, dropping the \"ipv6-only\"\n \u003e naming\n \u003e \u003e from the job. What are your views on that? In manila, we care\n \u003e that\n \u003e \u003e this job does data path testing with IPv4 and IPv6; so calling it\n \u003e \u003e \"ipv6-only\" is a bit confusing. What\u0027s the reasoning behind this\n \u003e \u003e naming? Wonder if we can just drop it and still be okay.\n \u003e \u003e\n \u003e \u003e Thanks!\n \u003e \n \u003e devstack-tempest-ipv6 base job set SERVICE_IP_VERSION\u003d6 so that\n \u003e every openstack service and DB etc listen on ipv6 address not on\n \u003e ipv4. If you see on the log, you will be able to see manila also\n \u003e communicate on ipv6. I have not checked the tests testing the IPV4\n \u003e data path but if that is there then we can verify it.\n\nto configure the dual stack for neutron, you need to set the IP_VERSION\u003d4+6 in devstack job. so either you set control plan listing on ipv4(devstack-tempest) or ipv6(devstack-tempest-ipv6) and set IP_VERSION\u003d4+6 for dual stack configuration.","accounts_in_message":[],"_revision_number":5},{"id":"8d214c1ad6917249f4278cf3a2e6f89bab46e4af","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 18:06:44.000000000","message":"Patch Set 4:\n\n\u003e devstack-tempest-ipv6 base job set SERVICE_IP_VERSION\u003d6 so that every openstack service and DB etc listen on ipv6 address not on ipv4. If you see on the log, you will be able to see manila also communicate on ipv6. I have not checked the tests testing the IPV4 data path but if that is there then we can verify it.\n\nYes - You can see this in the last run; and I\u0027d like to preserve this behavior. However, we\u0027ll drop the suffix in the job name: \"ipv6-only\".\n\n -  the environment supports both IPv4 and IPv6 (IP_VERSION\u003d4+6)\n - The control plane is on IPv6 (SERVICE_IP_VERSION\u003d6)\n\nSo, in this environment, Nova VMs can still be provisioned on IPv4 only private project networks, and be connected via Floating IPs. This is what matters to manila, since we perform access control via client IP addresses, and test the data path by mounting shares and writing data. For this, we don\u0027t care if the service/control plane is on IPv6.","accounts_in_message":[],"_revision_number":4},{"id":"a9cc78ecfa7fb37b55d4b43e43b6409b98a7e3b5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 18:29:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c87effa6832b2483b9240543bd9fbe29eb48a71f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 19:47:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3867efc04e1332bd96fac439499d8575db26a4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 19:47:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07e9b245ca64846b6f0bc0e618dc941 : SUCCESS in 4m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9308df115f424b21ba99b18c0390d601 : SUCCESS in 1h 13m 19s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4e1262560f3a4e1e8e01ee8fb1b7733b : SUCCESS in 46m 19s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4655971a22f145c3ad05bba4abaab624 : SUCCESS in 1h 04m 17s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/dfe51007733c4389a1711bc04b98426b : FAILURE in 1h 13m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0ca691b48b9643269093a28abf6fdab2 : FAILURE in 53m 02s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0c87d9de8a75464a931cdf0c1dfa08a2 : SUCCESS in 52m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/93a85fb5a0ca4ae0aa12c04fa3b86856 : FAILURE in 47m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07883727d2334fcfbad9ca3a7f1c20fb : FAILURE in 1h 35m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/54ab6a4e26724701856f48e242b5f167 : SUCCESS in 30m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c16a7c7577c546a3bc82c5b211135b8a : SUCCESS in 1h 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"71838cf8df8c542300cafbea383a090f90bdcb4c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 19:53:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a2a58d2466d61af2d5e17d41dd22b1f9796b4ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:49:48.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"24fba96c753ca34b646fcf5bcfb7097a03476bff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 22:01:17.000000000","message":"Patch Set 6: Workflow-1\n\nWill refresh this change after https://review.opendev.org/#/c/725461/ merges - can\u0027t have a dependency on manila while I have backports pending in manila..","accounts_in_message":[],"_revision_number":6},{"id":"1e0ef93f51e7a2cbbdd2a0fa069d58ae86bf03a6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 22:01:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/725438/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f69b7af/ : SUCCESS in 2h 13m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/725438/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e3d9414/ : FAILURE in 1m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"22c7d04d217b347585be74e132a74dd14f71c2e0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-07 03:49:53.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"719c7dd6e394211f63f459e39d08a3ab648334ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 05:36:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022cd4006801477493896d41612a4f5e : SUCCESS in 4m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f212d21a970b495abaf05ca98801409e : SUCCESS in 1h 04m 00s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0392774080144897bb35a8d4c0140e98 : SUCCESS in 57m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/898f5916b04141d19c814b5057396493 : FAILURE in 1h 21m 55s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/8f842683421d4739b95c40a56603ed8c : FAILURE in 1h 40m 59s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ef23c8ab522c47a3a0aab752b78fc1d4 : SUCCESS in 57m 07s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/2953f53b52b84fbe9fc64e2c2c877d4b : SUCCESS in 45m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d40c59ae1aab458a8c274730bb2c8a1b : FAILURE in 46m 57s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/37fe155616f64920afb121aa2931cf3f : FAILURE in 1h 31m 28s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/9508b71da49449ba987b8f3b31ce775f : SUCCESS in 20m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/241ec67830eb468ca587f3bbac8c3279 : SUCCESS in 1h 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0e91bcf85536b04f7e40771c7477855bc87d95b3","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-07 11:01:30.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"75b5725a0301a97e988eb7f8488cf2b122d065e7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-07 11:13:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/725438/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/13841de/ : SUCCESS in 2h 13m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/725438/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f9ed1af/ : FAILURE in 5m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f08f6561a63ee342567fcdb4cf5c77846953e2c2","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-07 11:43:52.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b7af1de88583e45ee1a40fff9bb6c401a8a96c4e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-07 12:26:15.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"85ab7c82eb83b72ec547697845a237d15373e12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 12:26:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"18f391611f0b0f3a17e7e1b768499b08d2c16223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:34:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/558b8105f9b642b495504054a7ccb34b : SUCCESS in 4m 41s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/48ccfb826e2b4b5b9151fae5cc808e26 : SUCCESS in 48m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/beabbbe65af74b509a9072cc2428bf17 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":7},{"id":"e90add592f4ac088e2465d5e5cd427983f9e9980","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:34:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4484d1a514114e41dacb1a5c18e61eb4b6b16d2d","revisions":{"a43abae57a36b2c85a85996c6296b70bcaf63ddb":{"kind":"REWORK","_number":1,"created":"2020-05-05 00:34:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/1"}}},"commit":{"parents":[{"commit":"fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"subject":"DNM: native zuulv3 job for lvm","message":"DNM: native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a43abae57a36b2c85a85996c6296b70bcaf63ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a43abae57a36b2c85a85996c6296b70bcaf63ddb"}]},"branch":"refs/heads/master"},"2ce6e3872c6d276515a938441192035cf3957c4c":{"kind":"REWORK","_number":2,"created":"2020-05-05 04:10:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/2"}}},"commit":{"parents":[{"commit":"fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa71bd3f8cd9c1dca4768c0bb0dff2aeb1447972"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 04:09:55.000000000","tz":-420},"subject":"Add native zuulv3 job for lvm","message":"Add native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ce6e3872c6d276515a938441192035cf3957c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ce6e3872c6d276515a938441192035cf3957c4c"}]},"branch":"refs/heads/master"},"4d9f07835c0c9ebe1fb910ccb29f01a6cfecb280":{"kind":"NO_CHANGE","_number":3,"created":"2020-05-05 04:12:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/3"}}},"commit":{"parents":[{"commit":"5dd6301d5ba0037092ff0f64b5dba899734f0d94","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 04:12:15.000000000","tz":0},"subject":"Add native zuulv3 job for lvm","message":"Add native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d9f07835c0c9ebe1fb910ccb29f01a6cfecb280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d9f07835c0c9ebe1fb910ccb29f01a6cfecb280"}]},"branch":"refs/heads/master"},"b061de6bb1f3fd14fa01f8a41ed1f9b5ac71211f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-05 05:27:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/4"}}},"commit":{"parents":[{"commit":"5dd6301d5ba0037092ff0f64b5dba899734f0d94","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 05:27:01.000000000","tz":0},"subject":"Add native zuulv3 job for lvm","message":"Add native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Icf5f26397e16d7fbfae77aa246ac5e35581f441e\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b061de6bb1f3fd14fa01f8a41ed1f9b5ac71211f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b061de6bb1f3fd14fa01f8a41ed1f9b5ac71211f"}]},"branch":"refs/heads/master"},"da28d4955ff3a298d9605998c61b68dcf7626fe6":{"kind":"REWORK","_number":5,"created":"2020-05-06 17:41:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/5"}}},"commit":{"parents":[{"commit":"5dd6301d5ba0037092ff0f64b5dba899734f0d94","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-06 17:41:25.000000000","tz":-420},"subject":"Add native zuulv3 job for lvm","message":"Add native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Icf5f26397e16d7fbfae77aa246ac5e35581f441e\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28d4955ff3a298d9605998c61b68dcf7626fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da28d4955ff3a298d9605998c61b68dcf7626fe6"}]},"branch":"refs/heads/master"},"f911313dbc3b347c0b1a822ef07fd80dd47eaf73":{"kind":"REWORK","_number":6,"created":"2020-05-06 19:47:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/6"}}},"commit":{"parents":[{"commit":"5dd6301d5ba0037092ff0f64b5dba899734f0d94","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-06 19:47:28.000000000","tz":-420},"subject":"Add native zuulv3 job for lvm","message":"Add native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nDepends-On: Icf5f26397e16d7fbfae77aa246ac5e35581f441e\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f911313dbc3b347c0b1a822ef07fd80dd47eaf73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f911313dbc3b347c0b1a822ef07fd80dd47eaf73"}]},"branch":"refs/heads/master"},"4484d1a514114e41dacb1a5c18e61eb4b6b16d2d":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-05-07 03:49:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/725438/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/725438/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/725438/7"}}},"commit":{"parents":[{"commit":"5dd6301d5ba0037092ff0f64b5dba899734f0d94","subject":"Add native zuulv3 job for zfsonlinux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5dd6301d5ba0037092ff0f64b5dba899734f0d94"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 00:34:38.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-07 03:49:29.000000000","tz":0},"subject":"Add native zuulv3 job for lvm","message":"Add native zuulv3 job for lvm\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I2981f786d21882579c2bbc569615843d253bf5dc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4484d1a514114e41dacb1a5c18e61eb4b6b16d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4484d1a514114e41dacb1a5c18e61eb4b6b16d2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
