)]}'
{"id":"openstack%2Fmanila-tempest-plugin~725487","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I24fca3de3243717928957bbfaa118e8c8aff88be","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cephfs-native","hashtags":[],"change_id":"I24fca3de3243717928957bbfaa118e8c8aff88be","subject":"Preparation for support scenario tests on CephFS Native","status":"MERGED","created":"2020-05-05 07:17:32.000000000","updated":"2020-06-02 01:57:21.000000000","submitted":"2020-06-02 01:57:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725487-1591063041303-3595709e","meta_rev_id":"d1ddc043837b640ccff7ac06c2d2cf6226aac3c7","_number":725487,"virtual_id_number":725487,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":8056,"name":"Ramana Raja","email":"rraja@redhat.com","username":"Ram_Raja"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-06-01 22:20:07.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 01:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-06-01 23:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":8056,"name":"Ramana Raja","email":"rraja@redhat.com","username":"Ram_Raja"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-02 00:53:00.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-06-01 21:02:34.000000000","_account_id":8056,"name":"Ramana Raja","email":"rraja@redhat.com","username":"Ram_Raja"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-02 00:53:00.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":8056,"name":"Ramana Raja","email":"rraja@redhat.com","username":"Ram_Raja"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 07:43:26.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-05-07 18:36:33.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-11 21:47:29.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-06-01 21:02:34.000000000","updated_by":{"_account_id":8056,"name":"Ramana Raja","email":"rraja@redhat.com","username":"Ram_Raja"},"reviewer":{"_account_id":8056,"name":"Ramana Raja","email":"rraja@redhat.com","username":"Ram_Raja"},"state":"REVIEWER"},{"updated":"2020-06-01 22:20:07.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-01 23:33:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-06-02 00:53:00.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-06-02 01:57:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0900bfb55b97c58f24f919cd12305d107c44bfb","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-05 07:17:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3c4a9f06f2b9624aafcae7046acbeb3ff914839","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-05 07:41:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdc4df5b8f6506296906b4d7ef14e5513763fa5e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-05 09:34:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa0bd5630bab7560905dddffd23c7a6be132304d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 11:39:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c8ad7d3e8ae487ab62382f3faf233b0 : SUCCESS in 3m 20s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2b5e296023d946b28d0ba0ba9aa9b3a9 : TIMED_OUT in 2h 02m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/9061b5e6845549c1a9bfda0388fe1b26 : FAILURE in 1h 26m 14s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c6017324d7bd443aab589c2528b64658 : FAILURE in 1h 53m 11s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d398a4307d8742d4bd9cb7299e02cb20 : SUCCESS in 53m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f547e1f9fddb4dbc841036f94b83eedc : FAILURE in 1h 30m 57s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc1ff8a71c64494b88850ce0a9ad3c2c : FAILURE in 1h 54m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/82bfaaefa6a44fee825e939e0c46d16f : SUCCESS in 27m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b398ef1269740d4b02d959e599e92f2 : SUCCESS in 1h 07m 37s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e931cee28e4242e2aa92d4215f43a2e0 : SUCCESS in 46m 53s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/f2f33d8ed59b4a8b9a06eb4704c523d0 : SUCCESS in 1h 35m 47s","accounts_in_message":[],"_revision_number":3},{"id":"18dfb2b45f283830efe497e2997552b9ecf2257c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-05 12:23:20.000000000","message":"Patch Set 3: Code-Review-1\n\nFrom the commit message, I expected the scenario tests to be run for cephfs-native.  The changes in the tests look like good preparation for that though.","accounts_in_message":[],"_revision_number":3},{"id":"ba40e04fd86cdbb8deab1f36dce6c97844ca0941","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-05 13:39:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3013c72/ : SUCCESS in 1h 59m 37s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2703109/ : SUCCESS in 1h 31m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ebbff6b749f53ff29fbd85cbeaf09c0f0f8a1a1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 22:57:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nThank you for starting this effort. Like Tom said, it\u0027d be nice to see these tests running in the cephfs-native job here. Feels like it should work if you install ceph-fuse and crudini via manila-image-elements into the manila-service-image-master test image that we use on the gate... Let me know if you need help with that.. \n\nSome comments inline for improvements..","accounts_in_message":[],"_revision_number":3},{"id":"7c2d1acc35ef71c19a69d57f93ad06a316bb3ffe","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-07 11:29:37.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nOnce we have patches for preparing the image and job for running the tests , this patch will be depended on those patches.","accounts_in_message":[],"_revision_number":3},{"id":"432d0a82e50c0572c861288cc94fa1457f9ba56a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-07 11:37:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e4f82a322438aef9fb5a64d29085c1ea28425dd","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-07 11:40:02.000000000","message":"Patch Set 4:\n\nJust make sure you are aware, I ran those tests downstream and all passes.","accounts_in_message":[],"_revision_number":4},{"id":"570fc7b63e655f9a7841e1cba9801c77a75f0eae","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-07 11:45:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"748f79f1e9fda8dd233c26bf921916eca6dfdc48","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-07 12:29:38.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5d793038daac2103861c35c69d3eb52f2ea998d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:30:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a983f3f3c8374788bf10c6587ad1e7da : SUCCESS in 3m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/106414bc69af4876b5916509bf2c9a82 : FAILURE in 1h 19m 16s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/6d9fd92a39324ab69a8a96ac26127ebe : SUCCESS in 1h 26m 49s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b09b51de32f44a2cbe31a5de43089b10 : SUCCESS in 53m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/65098e685d924c3a8aa6f9b282eb08cb : SUCCESS in 1h 02m 17s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/077a1f48d596410daea378e63170a66a : FAILURE in 53m 47s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5cf025c3570046bead68443d582aafc1 : FAILURE in 1h 42m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6aabaa49c5e048db821b242907b8551b : SUCCESS in 36m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f123c4d4e7e14ebd99e9be4d9b4bfc5f : SUCCESS in 1h 09m 21s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4c336daabcba465396efbfbd187870b1 : SUCCESS in 40m 02s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/6a93ddbcb4f44bbfb9e4cf506f5026d5 : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":5},{"id":"8721fc5402fddb88d622f436d8012fc219e81ee3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-07 15:34:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3158fda/ : FAILURE in 2h 03m 27s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e45966c/ : SUCCESS in 1h 32m 10s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"1f008b4580def026d012ef6657402e4a86ebb5dd","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-07 15:47:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30f4e0825144cf453915d394ee52c296aeccc053","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-07 15:54:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e1faf52d297ad73f09cd083fe66f81d1b20d0a2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-07 18:36:33.000000000","message":"Patch Set 5: -Code-Review\n\nIt seems I misread the test code, sorry.","accounts_in_message":[],"_revision_number":5},{"id":"07132492423a9b7bb3f99620eeb1e6420511643c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-07 19:33:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"296aa9b23dc3820f8f4c984d5cb52750d2bc3588","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-10 08:52:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"750c8268848aaecf44598bb59b5eb60530208abb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-11 20:27:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e653095b90689d87b907304e6056de00c94b6a93","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-11 21:47:29.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nCode looks good Liron! Just a minor nit and also I endorse Goutham\u0027s comment on adding checks for both ceph-fuse and ceph kernel client if possible. I\u0027ll run this on my env and submit feedback later today.","accounts_in_message":[],"_revision_number":5},{"id":"099c313aa89fb907fdc02a994cf239bb3a7cb530","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-12 05:03:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce355ab0bec8e48f9ae50cad6d55fd2b9ddd7ae4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-12 05:36:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"30650e6b7c763c1c9e46c606dcaeee69fd25bee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:26:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053f8293b8a94d88803327463a261516 : FAILURE in 3m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93b47d73999f4ad58a402ed4e90fe533 : SUCCESS in 1h 00m 51s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ed63bd6db6004edb91541cefca06cb48 : SUCCESS in 55m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4e1ca2d31c3242fbba581bae3ce6cd5a : SUCCESS in 1h 06m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c04d2ec4207345969441ba17af60d696 : FAILURE in 1h 19m 04s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/331bd7a7b1cf4b769ca7453d5e701d0a : SUCCESS in 1h 28m 51s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5a4921bbd1a54cc398265164560671b9 : SUCCESS in 1h 01m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f4125e860d9946ab9bae19de9bf829e1 : FAILURE in 1h 14m 49s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c51cba9d2d2d4809818fe7d1624c2970 : FAILURE in 1h 48m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/25db13ded49d4c43bc7b385130165350 : SUCCESS in 23m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/98406bfb38c948debe70b62db30ecd7c : FAILURE in 1h 10m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_quotas.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_actions.py","accounts_in_message":[],"_revision_number":6},{"id":"23c1de0a6438ee2919a69a3c453aa1cf672797eb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-12 17:26:54.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6b7f9cc32de20c9047cb545d370001ebedadd88f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 22:32:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795fbcdec6074f60a1a37f535d8ae1f9 : SUCCESS in 6m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4926d5cb2eb2408f9110068d587a6aa4 : SUCCESS in 1h 25m 26s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/dec77112523c48b78f1012f8344b5803 : SUCCESS in 48m 36s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/accbec89145849a5a39bb00b6a9911ac : SUCCESS in 1h 04m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1bdd5378e03f493fbc88eccd2d249ebb : FAILURE in 49m 48s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b7acdd3543504686b3ea195edd668058 : FAILURE in 1h 17m 00s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/813e28f3d1e24e67a9822e244a120fc7 : FAILURE in 54m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e8491955467441ae83f9f5b8f5a0f615 : FAILURE in 50m 47s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05743116326142c382803171a53c850a : FAILURE in 1h 53m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f413ea398b62471c93a785ddf61485c8 : SUCCESS in 28m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ac57f7e7b5e44431a9ec93b2229698aa : SUCCESS in 1h 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7e64805832789971836742ddc18c390bf41807c4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-13 00:47:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5a9aada/ : SUCCESS in 2h 14m 46s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e0ed068/ : SUCCESS in 1h 28m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"1e7aaed1bb26c96d6fad15ccf9d0fdfcf7c36fbd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-27 21:58:01.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)\n\nThanks for the changes. It\u0027s great to finally have coverage for the cephfs protocol; atleast with a subset of the scenario tests. The tests have now run against this change, with manila-service-image based VMs mounting cephfs shares: https://review.opendev.org/#/c/730996/\n\nDo you have plans to follow up with necessary changes in manila_tempest_tests/tests/scenario/test_share_extend.py and manila_tempest_tests/tests/scenario/test_share_shrink.py ?\n\nI\u0027ve also left some comments inline regarding the use of ddt and f-strings. I\u0027m okay with merging this and working on those concerns later. What do you think?","accounts_in_message":[],"_revision_number":6},{"id":"fa465f02eb6d68de4f1bc818fd84ac3d571b644e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-28 08:57:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b85909acd8205568908977f6d41249ca6aa413f4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-28 09:41:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e6d9c42e7cf195bbf3dd7a0760440e99f2a3728b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-28 09:46:02.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"abbbc170561fdb05f1473be821b2134d264caa56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:21:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7624ecb56a65449aba18ac5f07a41ac0 : SUCCESS in 3m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9c0e8602a15741f996e15ec1aba58e6c : FAILURE in 1h 05m 11s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d1205ed9bf3142a7bd96e92b3c166f88 : SUCCESS in 49m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bfadd34d4cdb4f0796da0777db178e5b : FAILURE in 1h 13m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/399271658a454150b4128bf4da6a20fc : SUCCESS in 54m 05s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e90c8cd4c1dd47c8970932f24016a1f3 : FAILURE in 1h 22m 26s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ad09a0e7c5c14c47aee70f194aa3dd96 : SUCCESS in 1h 00m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2a163f965c7640be89ef83b305096c37 : FAILURE in 56m 17s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2c18d686b5e4d10aa05a0dc33ec6e09 : FAILURE in 1h 37m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/fcfe1ba2fb62421684db151a6a1ebc90 : SUCCESS in 26m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/83377dd1850c475990e1a5840793ceb5 : FAILURE in 1h 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9b8dbef471d9fcd93b4cad129fb8b7adc1f58239","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-28 11:41:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7b7d9be4c9bfae73c9c4b4ec94f41446b9b564fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:15:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3627510bed214467b5328b548b9813e2 : SUCCESS in 3m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/209d02b89c2a407eac3e40c15309018c : SUCCESS in 1h 18m 07s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5d9bd09d001d4528ae49da765cc7edb7 : SUCCESS in 44m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0d5951aee6d74711908869aab4ba2cc1 : SUCCESS in 1h 03m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1e1c5d0061844ac3850a13acd81a68a9 : SUCCESS in 1h 07m 19s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/32f24d60d700429fb42d23b3bbe9dab8 : FAILURE in 1h 19m 34s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/78c3ddf5117c4ca98e5dbe94a1b7f95f : SUCCESS in 1h 00m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/0f3148bbf2a44aed9767c9ef80b21ca8 : FAILURE in 52m 25s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32d777d7634a4025bb932f7b95016f39 : SUCCESS in 1h 32m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/22e51b2abb224c7dae0ff54285bfe925 : SUCCESS in 28m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/48a0dfa5a71349e6b33ad3f43fb4c927 : SUCCESS in 1h 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4dcdcf633e67ae12e981438add6061fca705a3e9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-28 14:26:04.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks, Liron, LGTM.","accounts_in_message":[],"_revision_number":9},{"id":"9261adc3fa79b0b1b5c5b87473b33e36f39b7e10","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-28 15:22:12.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/cbad20d/ : SUCCESS in 2h 06m 06s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d4f58e0/ : FAILURE in 3m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"abfd2ddb6d5a989cef8e526d0213529c550b5bcd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-29 06:04:47.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"daf844324a73f3ea894ae7f2c57a71b352617268","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-31 06:13:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c245033fb01616cfeb688f31559e63a927ce620e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 07:57:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7971539e2c744a489cd2c1a5c7272dbe : SUCCESS in 5m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d7e7337238634cd88323ee2e8a38d2b4 : SUCCESS in 54m 05s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/721cb29b9a4748bc94da40992adcd82b : SUCCESS in 40m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a4f5a2bf3284e8e8aff05fbada0c795 : SUCCESS in 1h 06m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aa5eb7c43ef7493e9705e63c6df3649d : SUCCESS in 52m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fe24e40607f94bf5bdeb31ff5043bb44 : SUCCESS in 57m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8784366ec626437dac9ec784b87b5678 : FAILURE in 1h 37m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8157b749a1c04dbab5f9853a4e065396 : FAILURE in 1h 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b0aae4cb1c3af885aed01cbf13cef811a7da3457","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-31 08:02:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3c287aad9e06d5253a3daa53ab4cdacea2e7c4d5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-31 12:08:29.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/80717a1/ : SUCCESS in 2h 07m 12s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/27cafaa/ : FAILURE in 1m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"657868cfaf111b08a4c567b0c1c343d3de16b4cf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-31 15:41:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"033bc49361f2f920b321f834e1a8718e2dbf1429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 17:59:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0d53e15ebe43d587455f1d0c8b18a6 : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/00f0b67e2ebf4e1ca114075da6ac1b18 : SUCCESS in 40m 56s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b198ad1b5a8c4aa8bb933adaf1af69b0 : SUCCESS in 38m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e6dd3c40ef724df4ad36f3e56d497e19 : SUCCESS in 1h 19m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/078efc892021415ba98b276f6377fdc0 : SUCCESS in 48m 36s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b8e187d03d094265959560bc68e8b7d2 : SUCCESS in 59m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d83ae3e8600e4488b7bebbe658089751 : FAILURE in 2h 17m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5bb2fb7408f34f139428ff9a3b3ca9e0 : FAILURE in 1h 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0ea47b103594761b3301627889c211fe352b5367","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-31 19:34:15.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6aa539698868a98cf2e3d0be3133703f536ba721","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-31 20:13:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"205c7af322beae6b940b4dc456f2d2ab3ad6082e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 22:26:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07353b0b1fff4b478ded9990fe7affa8 : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/10c021fd25754fd58b3a9bd7e6b05627 : SUCCESS in 1h 04m 40s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0d070e726c5b43a38a98e74abbdae6fc : SUCCESS in 48m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/28a779c15d6d4ab6bb0ce35ebce08298 : SUCCESS in 50m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/31329212b3c44ff3883ff567e1b18064 : SUCCESS in 53m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/69d509a0cb824fb69561ccf0ee84cb20 : SUCCESS in 51m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8aec593274b1493bb604af22550be8d0 : FAILURE in 2h 07m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/15d770b3901b462d9662d37051e8f499 : FAILURE in 1h 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f3488e39ef0db69733aeac83617d5b030dbfd9cd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 00:45:23.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/12/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d755de3/ : SUCCESS in 2h 18m 45s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/12/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/bfdc93b/ : SUCCESS in 1h 30m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"02858ac4f581499f3b3930495f11a7cf168caf35","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-06-01 08:37:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1bff991edfa14f37e3ef2dbbff78161266e256ab","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-06-01 08:54:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0737967839056794db9a315c387bf00f19acbc59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 10:48:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d77f08990c048b39b07b9c50cd43115 : SUCCESS in 4m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/91db008060a9491c93cccc9c8e53937a : SUCCESS in 1h 21m 01s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3cb8851453a0474ca3361c3faccf2cd8 : SUCCESS in 51m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/06e0f69382194b109cfb22be8796dff1 : SUCCESS in 1h 12m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c42e37f56cae44848df63fb20c37452d : SUCCESS in 57m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a17988e9a2034c559d9ebd035d27acc8 : SUCCESS in 1h 06m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dae557d608d0411ca9bfb5d8d3f1ee48 : FAILURE in 2h 10m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ce174f92fcff46758dbb8c713d0c7655 : FAILURE in 1h 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"525203181ebff70c7c47d3c19c54287089a9b02a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-01 12:10:41.000000000","message":"Patch Set 13: Code-Review+2\n\n(3 comments)\n\nThis looks good to me. Kernel tests are failing due to incorrect quota enforcement at the moment. However, that\u0027s not because of this patch: https://review.opendev.org/#/c/732014/\n\nWe can file a bug to see if we should expect those to work on Luminous at all; if they don\u0027t, we can consider blacklisting them if running against Ceph Luminous, and enabling them only when testing against Ceph Mimic, Ceph Nautilus or beyond.","accounts_in_message":[],"_revision_number":13},{"id":"0faf73838d930c09d24fc26df668d71f7d74044b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 12:56:56.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/13/upstream-check/manila-cDOT-manila-tempest-plugin-ss/cbacad7/ : SUCCESS in 2h 08m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/13/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/670b651/ : SUCCESS in 1h 27m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"03f678605c15cfa4ad72c71de89c9ec1a2e970d2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-06-01 13:14:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"58740bda14f64971f4d30635cfb0b408ef025180","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-06-01 13:15:02.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6c7e34a4236b71a374ee321341ba1dde2d79a27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:28:00.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40cd68c40134f0384af87f14c5e17d9 : FAILURE in 7m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a55449e7993d409e8c65a362a8e5f2e9 : FAILURE in 1h 02m 13s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c2a7af8edb01467884e052f67a2ec229 : SUCCESS in 44m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2fa11bf6f7cb49cdbfec6328ea7eb414 : FAILURE in 1h 06m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d05dd96ae25f49948e87bc5bd36aad09 : SUCCESS in 57m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/27e315b064584322b306a5906d6a9ef3 : FAILURE in 59m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/912674c18aec4b2098470beeab48b74a : FAILURE in 2h 07m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4ccf09ec111a44ed825d05ecbe8a9ca3 : FAILURE in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4f41ee3d7d4b24d484f6d0656b49610eec48e20e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-06-01 15:48:39.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"df14943830dd493f0fe32ddc9b68502ba2b37b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 17:23:39.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84894adbbe4a41dfae105d1086d9b356 : SUCCESS in 3m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7191e04005d94cf69323f04d26dc86bc : FAILURE in 1h 01m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/48b8ef13ab6c493486819e23387416ce : SUCCESS in 40m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dbca0490293f4326855d7a9200a5d1cb : FAILURE in 1h 02m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4e1e23b27b5d4983a6fa850fc50ad163 : SUCCESS in 56m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4ccb6be44354c1497b8070d0595e944 : SUCCESS in 59m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e271e04cf144a84ba5df480aff52b77 : FAILURE in 1h 31m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/889f630e47814f91917aac4ddd0ad5e8 : FAILURE in 1h 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b887cee34f07c77ca611c29c596ddf2a8705d1a2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-06-01 18:11:03.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9f6b8220a80d46f53fbc45257a74b16724e74381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 20:11:49.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d3a212d6594599a3444441eca2991d : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/69576c0311a447c69f672d4f95109f23 : SUCCESS in 1h 17m 29s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a54191bdcf5d44188de59cf7c2455aa4 : SUCCESS in 42m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58888ff0df7040f2972416d91e1651b6 : SUCCESS in 57m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bb5f2bd8cd334e44b749ea74349d4471 : SUCCESS in 1h 01m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a27ecb38b60d414aa747339f525e9732 : FAILURE in 52m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62ea88a710bd4d5494f9ff6ac558613f : FAILURE in 1h 59m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0cee2ce5149f400c93e7e5f53ff5592a : SUCCESS in 1h 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"533f5904185ea3bb7a99f0c3f8f19a8a924e3433","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-01 22:00:43.000000000","message":"Patch Set 16: Code-Review+1\n\nLGTM, waiting on the latest test runs on https://review.opendev.org/#/c/732014/","accounts_in_message":[],"_revision_number":16},{"id":"efae67057a72fb9dcaa5b2b7fa96465bc7a50d17","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-01 22:20:07.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/87/725487/16/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5c5436d/ : SUCCESS in 2h 08m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/87/725487/16/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3222fac/ : SUCCESS in 1h 22m 21s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"b23495e15e8d729a62c6b075f8e72f91a323550c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-01 23:33:27.000000000","message":"Patch Set 16: Code-Review+2\n\nAll these tests pass with the Ceph-fuse client and the extend/shrink tests  fail when running with the Kernel client: https://review.opendev.org/#/c/732014/\n\nThose test failures expose a bug that\u0027s not in manila or in these new tests. This is a great improvement, and we can blacklist these tests until we understand what\u0027s going on...","accounts_in_message":[],"_revision_number":16},{"id":"cd144ef9ae220c505bcdd691534db77574d11939","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-01 23:34:42.000000000","message":"Patch Set 16:\n\nLiron, can you propose running the tests here as a follow up https://review.opendev.org/#/c/732014/\n\nYou can supply the failing tests as a blacklist, and a bug reference will help. The ceph community tracks its bugs here: https://tracker.ceph.com/issues","accounts_in_message":[],"_revision_number":16},{"id":"62fb7c5eebb7400ddde5ea77a3a52966ed594d38","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-06-02 00:53:00.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nGoutham and Victoria and I discussed earlier today: we all agree with merging this and blacklisting the kernel client tests and opening a bug with Ceph w.r.t. the kernel client.","accounts_in_message":[],"_revision_number":16},{"id":"1a137b28a255beb9e088df7785e743c3900da91a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 00:53:11.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"97c00de79925cdae487ad4bc0da0d2ce391906c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 01:57:20.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbda27098d13475d9766f3ca1e3251c3 : SUCCESS in 4m 33s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e7e677099a9f4f9d89aecd42fe97b75c : SUCCESS in 47m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e84bb73c327f4714b9ba2316de2b7853 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":16},{"id":"42c8ac0cc6030d2667948d7df1e526c7e24fcdab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 01:57:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"8e68e447817a198070b96a276e25df19cf81ec19","revisions":{"876d7ea4ac14f6a937b6d1fc5a0c62e579d5c1f1":{"kind":"REWORK","_number":1,"created":"2020-05-05 07:17:32.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/1"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"subject":"Add CephFS Native support for scenario tests","message":"Add CephFS Native support for scenario tests\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/876d7ea4ac14f6a937b6d1fc5a0c62e579d5c1f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/876d7ea4ac14f6a937b6d1fc5a0c62e579d5c1f1"}]},"branch":"refs/heads/master"},"2b3fec25dd5f72c512f8a0b82d3364ede87d3087":{"kind":"REWORK","_number":2,"created":"2020-05-05 07:41:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/2"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:41:07.000000000","tz":180},"subject":"Add CephFS Native support for scenario tests","message":"Add CephFS Native support for scenario tests\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fec25dd5f72c512f8a0b82d3364ede87d3087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fec25dd5f72c512f8a0b82d3364ede87d3087"}]},"branch":"refs/heads/master"},"1d67006f80bb518f97a51cbc00128dcc15d21322":{"kind":"REWORK","_number":3,"created":"2020-05-05 09:34:20.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/3"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 09:34:05.000000000","tz":180},"subject":"Add CephFS Native support for scenario tests","message":"Add CephFS Native support for scenario tests\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1d67006f80bb518f97a51cbc00128dcc15d21322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1d67006f80bb518f97a51cbc00128dcc15d21322"}]},"branch":"refs/heads/master"},"8e5b98f14a3e188cd36ef38ca2ec403917980e75":{"kind":"REWORK","_number":4,"created":"2020-05-07 11:37:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/4"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-07 11:37:19.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e5b98f14a3e188cd36ef38ca2ec403917980e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e5b98f14a3e188cd36ef38ca2ec403917980e75"}]},"branch":"refs/heads/master"},"ee8a6970689c6b7bd664daf99a9149cf35cde1bb":{"kind":"REWORK","_number":5,"created":"2020-05-07 11:45:37.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/5"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-07 11:45:24.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee8a6970689c6b7bd664daf99a9149cf35cde1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee8a6970689c6b7bd664daf99a9149cf35cde1bb"}]},"branch":"refs/heads/master"},"81c76990c889fee2f185aa8dace89a3d6b491df9":{"kind":"REWORK","_number":6,"created":"2020-05-12 05:36:27.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/6"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-12 05:36:16.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81c76990c889fee2f185aa8dace89a3d6b491df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81c76990c889fee2f185aa8dace89a3d6b491df9"}]},"branch":"refs/heads/master"},"48879fee28e034f22fb9ca9983f14476fc19af29":{"kind":"REWORK","_number":7,"created":"2020-05-28 08:57:56.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/7"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-28 08:57:42.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48879fee28e034f22fb9ca9983f14476fc19af29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48879fee28e034f22fb9ca9983f14476fc19af29"}]},"branch":"refs/heads/master"},"fe4f9de0810d2668ec35e74a1e7498c19188c391":{"kind":"REWORK","_number":8,"created":"2020-05-28 09:41:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/8"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-28 09:41:35.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe4f9de0810d2668ec35e74a1e7498c19188c391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe4f9de0810d2668ec35e74a1e7498c19188c391"}]},"branch":"refs/heads/master"},"7f5df8ef100450fc36808007977304dd00297a28":{"kind":"REWORK","_number":9,"created":"2020-05-28 11:41:27.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/9"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-28 11:41:10.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f5df8ef100450fc36808007977304dd00297a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f5df8ef100450fc36808007977304dd00297a28"}]},"branch":"refs/heads/master"},"cc40afc9a93501e25888d4df77fda2bb7dc2647e":{"kind":"REWORK","_number":10,"created":"2020-05-31 06:13:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/10"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-31 06:13:23.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc40afc9a93501e25888d4df77fda2bb7dc2647e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc40afc9a93501e25888d4df77fda2bb7dc2647e"}]},"branch":"refs/heads/master"},"716a0ef00fe126aef0e5034577e3d21d1fa45383":{"kind":"REWORK","_number":11,"created":"2020-05-31 15:41:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/11"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-31 15:41:36.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/716a0ef00fe126aef0e5034577e3d21d1fa45383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/716a0ef00fe126aef0e5034577e3d21d1fa45383"}]},"branch":"refs/heads/master"},"a034ec9ccf6d736d719ca9981f98a702e335cfbc":{"kind":"REWORK","_number":12,"created":"2020-05-31 20:13:46.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/12"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-31 20:13:32.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a034ec9ccf6d736d719ca9981f98a702e335cfbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a034ec9ccf6d736d719ca9981f98a702e335cfbc"}]},"branch":"refs/heads/master"},"5fe2104f4c01926418111d1551f11d4374b29bc9":{"kind":"REWORK","_number":13,"created":"2020-06-01 08:37:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/13"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-06-01 08:36:58.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5fe2104f4c01926418111d1551f11d4374b29bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5fe2104f4c01926418111d1551f11d4374b29bc9"}]},"branch":"refs/heads/master"},"236a0edeb08e78685df9035f001a43e152257bfa":{"kind":"REWORK","_number":14,"created":"2020-06-01 13:14:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/14"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-06-01 13:13:50.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/236a0edeb08e78685df9035f001a43e152257bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/236a0edeb08e78685df9035f001a43e152257bfa"}]},"branch":"refs/heads/master"},"7579544d1fd5105cdebe241009e9d96bfd6c90bc":{"kind":"REWORK","_number":15,"created":"2020-06-01 15:48:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/15"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-06-01 15:48:26.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7579544d1fd5105cdebe241009e9d96bfd6c90bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7579544d1fd5105cdebe241009e9d96bfd6c90bc"}]},"branch":"refs/heads/master"},"8e68e447817a198070b96a276e25df19cf81ec19":{"kind":"REWORK","_number":16,"created":"2020-06-01 18:11:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/87/725487/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/87/725487/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/87/725487/16"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-05 07:13:44.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-06-01 18:10:51.000000000","tz":180},"subject":"Preparation for support scenario tests on CephFS Native","message":"Preparation for support scenario tests on CephFS Native\n\nTo support Cephfs Native driver a new method was created for\nconfiguring Ceph clients to access the shares.\nFor Ceph native backend \"ceph-fuse\" package must be installed\nwithin the instance for mount operation.\nAnother package called \"crudini\" is required within the instance\nfor handling ini files.\n\nChange-Id: I24fca3de3243717928957bbfaa118e8c8aff88be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e68e447817a198070b96a276e25df19cf81ec19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e68e447817a198070b96a276e25df19cf81ec19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
