)]}'
{"id":"openstack%2Fmanila-tempest-plugin~725635","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie460cbe4779005e284af0652a14112c1939d4855","project":"openstack/manila-tempest-plugin","branch":"master","topic":"TrivialFix","hashtags":[],"change_id":"Ie460cbe4779005e284af0652a14112c1939d4855","subject":"TrivialFix: Remove unnecessary import","status":"MERGED","created":"2020-05-05 15:41:19.000000000","updated":"2020-05-05 20:08:25.000000000","submitted":"2020-05-05 20:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725635-1588709305496-23512575","meta_rev_id":"9e57078d5d28f95ae8336077174d6c7ea707b6e2","_number":725635,"virtual_id_number":725635,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-05-05 19:40:35.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 20:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2020-05-05 18:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-05-05 18:49:05.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-05-05 18:49:05.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 18:08:16.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-05-05 18:49:05.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-05-05 19:40:35.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-05 20:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ccbe09800f935e61976ae05e0bf1015f99deac7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 15:41:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07c5d8e6d4a14b5b6387f825d9ba5ae949f4aa93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 17:24:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9cfb3eb1454db88092ed9c6ca4e52b : SUCCESS in 6m 54s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b4a7f66d0df04c28b10d444a03b42815 : FAILURE in 1h 06m 21s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5be9d793d62a4948a589058285af3f41 : FAILURE in 1h 17m 54s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f0120df5c7064807bc6878a9f1ec53f1 : FAILURE in 49m 32s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ee4802bfbf0649b0a5db2254dc4364d3 : SUCCESS in 47m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/894ad4e3df2547639104172bf8f2c715 : FAILURE in 44m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d01bbac6c8b040fe861aa6b781d2c254 : FAILURE in 1h 40m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/8690da163e934cd1b833a671f2434eb3 : SUCCESS in 32m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57d4a67cfb994116ac53f46b95706c44 : SUCCESS in 1h 18m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b6c1cc4b91444fce97a3a389f6f03f4e : SUCCESS in 42m 05s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9383c415dce04f9892f3b5e757ffb7db : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"418437774d91fe27d740713d84fe25f913cce92c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-05 18:08:16.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"cda8393b5f74c987d1fee233b0f0c752de253e07","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-05 18:49:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ae223f53de0d89cfca3d7e1a131b331acd1bc18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 18:49:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd7dbe1614275eb9dab0c53ed1755f6149d306ac","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-05 19:40:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/35/725635/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e59d2c3/ : SUCCESS in 2h 15m 16s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/35/725635/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1a761aa/ : SUCCESS in 1h 37m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fb777b97498d8488fa87454bf5568712253d0619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 20:08:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d7521243bf1440bb208fd1fa621a02f : SUCCESS in 3m 44s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/aac928d8a2974480a6ddc6ef3216b89c : SUCCESS in 42m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/45eff603e772434da4eda2514ec21631 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"13f578b4d6983c4179d600eb6af56f46b60fcf16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 20:08:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"696a057ca2db5558dfcfdf73bcbd4a0f34f2aeb2","revisions":{"696a057ca2db5558dfcfdf73bcbd4a0f34f2aeb2":{"kind":"REWORK","_number":1,"created":"2020-05-05 15:41:19.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/35/725635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/725635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/725635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/725635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/725635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/725635/1"}}},"commit":{"parents":[{"commit":"ce66f004c89e5f04e85f8eec96170d1460d701fe","subject":"Merge \"Update shrink scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce66f004c89e5f04e85f8eec96170d1460d701fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 15:41:07.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-05 15:41:07.000000000","tz":-420},"subject":"TrivialFix: Remove unnecessary import","message":"TrivialFix: Remove unnecessary import\n\nprint_function was being imported, but\nnot used in the module.\n\nChange-Id: Ie460cbe4779005e284af0652a14112c1939d4855\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/696a057ca2db5558dfcfdf73bcbd4a0f34f2aeb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/696a057ca2db5558dfcfdf73bcbd4a0f34f2aeb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
