)]}'
{"id":"openstack%2Fmanila-tempest-plugin~725680","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d","subject":"Add native zuulv3 job for cephfs native","status":"MERGED","created":"2020-05-05 20:22:25.000000000","updated":"2020-05-29 20:45:39.000000000","submitted":"2020-05-29 19:15:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725680-1590779719409-2a4b48c7","meta_rev_id":"849cd31fc7382a5d8215d9104a2c542ee15d40d3","_number":725680,"virtual_id_number":725680,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-05-29 20:45:39.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 19:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-05-29 17:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-05-29 16:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2020-05-29 18:32:27.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-29 16:46:02.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2020-05-29 18:32:27.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 16:46:02.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-29 16:50:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-05-29 17:44:12.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-05-29 18:32:27.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-05-29 19:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-29 20:45:39.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45a4181069c9d52b9f7c9db56239b28f7da633a9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-05 20:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cefe267ba9e6253b9f8653a224cb950820eefb9d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-05 20:26:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"188bb42b6291d423a29547951a2d9170ce70a6da","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 21:17:24.000000000","message":"Patch Set 1:\n\n(2 comments)\n\ngood stuff, excited to see this working!","accounts_in_message":[],"_revision_number":1},{"id":"e3b21340e66414140b8e46dc0149073c3f147ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 22:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cee939621746099e3566930225faca : SUCCESS in 3m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/59b84ec034f448b1bd8a8f743602a91b : FAILURE in 1h 02m 19s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/14828bb196854e4f97ac2bace6a69df4 : FAILURE in 1h 19m 43s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/314eb713c3154982b082cd78311dd809 : FAILURE in 50m 36s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/94c7a919be814cfea7cd42e96b80d5b0 : SUCCESS in 58m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/cdebbcfa97a24784a53785636fae0386 : FAILURE in 46m 51s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f07ea091a9e47a3b60bd208dab8c04f : FAILURE in 1h 35m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/60cab5343e04470e9bca6d6e0d10e3c0 : FAILURE in 9m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/72a2fc8387e140b7912fb8fe9a9fcb7c : SUCCESS in 1h 10m 10s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6e3d34a68f5e4593bd9f13618e8e41e3 : SUCCESS in 41m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/2f98ba0043254232b2c1d7f0eb52ffd0 : SUCCESS in 1h 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"852da9e4913e8ed5fd95b6a6aa5fdfd005122799","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-05 23:03:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da7cff9301b75c45d29f2eb40cda02a370cb7466","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-05 23:09:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92b8a1cb935043be9e85c2302b387201b71a0d98","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-05 23:11:20.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"aeebc7aacfeb02ca174954ef81f5e4e98280243d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 01:15:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0fff605cece4a3583ff476f6d87b20e : SUCCESS in 3m 20s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6691e6704cff4ca2baa816bcdc55b3f3 : FAILURE in 32m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a70600e0233a40c58de976e18cb73c8b : SUCCESS in 1h 02m 10s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/62a39b74b24543ce907c21e2a1348844 : FAILURE in 1h 10m 59s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b85be59de5524aaa841ea06c623e4a2e : FAILURE in 51m 41s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/c49f8c489e12466f904d521382249c44 : FAILURE in 52m 43s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/7ade7517468e412ba1cce201e6dec59d : FAILURE in 44m 57s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9ff520e5c9345da83127aba04ee7c5e : FAILURE in 1h 43m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3e65f077963c42a6bc671ede2480efc2 : FAILURE in 1h 26m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e01f11706b5e46d08956164d1345ea5e : SUCCESS in 48m 25s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/f1ed1d866e3e447fab26ab30514e2a4f : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":3},{"id":"1d06852219f704d31a8557ed61c968b15df4cb46","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-06 03:31:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/bc10981/ : SUCCESS in 2h 15m 02s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/28e8314/ : SUCCESS in 1h 26m 07s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5b0d53290680c04db7c4884c872bd403261c32d6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-06 21:55:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef1e7eb64180a483f1208fb57bd1874ed90595dc","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-06 23:46:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8fa80969b1b47a12db0d3f051edf021ac34412df","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-06 23:51:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b7755151d5ad324da6299a7af2da2b72ed8c93a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 01:39:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255d35da4f5340ceb0b4a4b2b73a6d2c : SUCCESS in 4m 51s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/af82c56f71244a9fbe71b3ca41496efc : SUCCESS in 55m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/de01929e26f24835b1137f3df0279b08 : FAILURE in 1h 20m 32s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/670bff6572f842a4a8253d92c3fe628c : FAILURE in 1h 22m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0e0a0d2cab514622b3061e7b355cda79 : FAILURE in 1h 00m 45s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/91f069db43864eb681fd0f7d7a2aea75 : SUCCESS in 1h 03m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4221caee234840f19a8aaee4d8a11bbe : FAILURE in 55m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad1a4443ad1a473aa61475be7cefb718 : FAILURE in 1h 39m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ffd5b1f44445438a8112a83dabeb9882 : RETRY_LIMIT in 35m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/365b71b2827d4d53bc11bf63df4119cd : SUCCESS in 52m 35s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4c173890eda943c1aa4074fa917725c0 : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f53abc7a53d814ee781b9c91e5628c31b7e1554f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-07 03:58:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/0b6802b/ : SUCCESS in 2h 18m 23s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/12f061e/ : SUCCESS in 1h 34m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"485fdc24afaf79a3e930090f32a3ea07b7c4f0f6","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-07 09:48:44.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"68315d4f8a9103e0382044b72454a03e266e7f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:33:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e6ff4c353d45b19759b1e222756814 : SUCCESS in 4m 22s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2f2810d1c2db4a568ef289d9ffbde55a : SUCCESS in 33m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fb4804fa206f46508f5a5c5f85eced93 : SUCCESS in 1h 12m 49s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/25a5f3ea04304b1ea9e0c7964956ce7f : SUCCESS in 1h 33m 23s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/974fa77091d14c668f3cf9a5215fe253 : SUCCESS in 57m 40s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/1d154198a8444d64bd4769adf9bad03c : SUCCESS in 1h 08m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3f88a2f8dfea4f56b331de70ca4b144f : FAILURE in 50m 19s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/272746cd094a436d850da58114b2bbb0 : FAILURE in 1h 39m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d4fb5af0f71541a3a0917757afd1bbe0 : SUCCESS in 1h 32m 08s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f7f096580f544f70bf28580ee3962f5f : SUCCESS in 54m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/372f0b13c8bb4f0080bd8ebf24f978a8 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":5},{"id":"f624709a3ab8d21f1188ad31c268fc0c69d84e78","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-07 13:42:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/4b7b4d6/ : SUCCESS in 2h 08m 20s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b4610df/ : SUCCESS in 1h 34m 11s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"1d9f2991d5744bb160f37ee25441219e76686cec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-07 17:41:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)\n\nThis needs to be rebased, some comments inline..","accounts_in_message":[],"_revision_number":5},{"id":"f1b61ea3282c7612950450482f67d2406d70354f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-07 17:55:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"39884e775c0fefec99009ac8eeaab2023a3afe0a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-07 19:28:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"10f08fd5aab22c338b9f5223f0e573e365cd2114","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-08 14:59:54.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cad8fe5d41d22d59a5a7a2b11de27ba75a349684","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-08 15:00:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"35a11c405c262acbe5e5693ecd78e0f0fc6bd9bc","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-08 15:01:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9cd3e791a6a74c6f9e8773648314e7b69b0d3d4a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-08 15:09:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1c621945dcbc61822e94f3584d7043245d0c8fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 16:58:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c614c25503c4b2e9c34e40919e82296 : SUCCESS in 4m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/357512ebeafc4276bc816e682192ca03 : SUCCESS in 1h 10m 11s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9f0ed93d94c54932b0a9d0a780886a23 : SUCCESS in 44m 59s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3ae71f793ea84d3ca883c3e0d1fe12dd : FAILURE in 28m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6bdba7a1b34d45b49a470fc21e4f4111 : SUCCESS in 1h 16m 46s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/6fba5c06d402456e9b5b953edc746a9e : SUCCESS in 1h 42m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/22adce3478d34744a8acc5b38a754896 : SUCCESS in 52m 15s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/899334ff7ac445409ee47c7294445bdb : SUCCESS in 50m 21s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3c8f7880f0d24f02ad24f678ac4ba827 : FAILURE in 43m 22s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c532a06f1fc14f15ad0e0749d1a43ded : FAILURE in 1h 35m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a0668f707bde4f669cdd4368ae4c955c : SUCCESS in 1h 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"68700ceae70f1516589cd85a676ea7794a975ea2","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-11 16:51:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5ec8e782d84cefae9571462a27bd53466b3978fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:55:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3b66052fdf4d5693e0f37f986faebe : SUCCESS in 4m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7919040704644ef2bfb1107464250a2b : SUCCESS in 1h 10m 09s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a15ba10f52c54b8bac95eaa290ff5aa2 : SUCCESS in 45m 11s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/201041c7a58e48ed89e040072edcbb33 : FAILURE in 30m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c871e6eb3b0b41cb84bba2a23324f573 : SUCCESS in 1h 01m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/238792df86a0463cb766bd373cb5af5a : FAILURE in 54m 21s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/8b903442c6eb48daa60ef0c6d0ad66c4 : FAILURE in 1h 34m 58s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/1576c0d55355432c96800cb4801b2a72 : SUCCESS in 57m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/96118230fbf44e0992198bf3e40adffc : FAILURE in 52m 54s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3259b50eef984fd4870b443631dcb504 : FAILURE in 1h 54m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ba91b8a4fe884ab9bc298c412af08a11 : SUCCESS in 1h 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"68fee1f67d71172be11973b8ef948d3c0fcf2925","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-11 21:19:03.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6edd75a/ : SUCCESS in 2h 08m 48s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/447d853/ : SUCCESS in 1h 40m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"7f47f1589ce7a09140954269b1e5aab3111da0f4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-13 06:00:58.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1f3a00b242891324baf5c3d00b94c5e3fb2f0d29","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 10:08:31.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"e0a19bc91fa94178580d08fc329e88dde0d3c403","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 21:51:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8f50ec803611091ff3a2c803eb5955c9c71f0af0","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 21:52:15.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"23d1e3cf2cc95823658898756482f2fbcb331a98","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 21:52:49.000000000","message":"Patch Set 11: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":11},{"id":"214f13552a0658837ef5e64263a494da6687daa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 22:34:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f136dcbc5ad944fbae0243c549625b25 : SUCCESS in 3m 43s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ab2fa6fc8f704552bf02fcc6a90d6a36 : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e57bf64c2e744c89854054ceb59da0c91f7954fc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-14 00:41:15.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/11/upstream-check/manila-cDOT-manila-tempest-plugin-ss/74834e9/ : SUCCESS in 2h 06m 31s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/11/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/21dcc13/ : SUCCESS in 1h 34m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"f695fb03f2a2c3adedfeddded9647ca107a50b36","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-15 00:07:28.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b3b68016c65286daa83feebfac6e2f1b677233d7","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-15 00:16:13.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"8f2ca9a81d35dca962a755611bffaefc699eb510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 01:02:01.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f32c3af89834f88b30e90cc8e4b8a7d : SUCCESS in 3m 34s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d40ff5d1d5b0451399ad85f7176d7078 : FAILURE in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"66e4209161f57f868213f2b3a38abfc658314966","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-15 02:54:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/12/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6860c2d/ : FAILURE in 4m 00s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/12/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d3a83d8/ : SUCCESS in 1h 52m 11s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"468d5b5936a8ec513b4f6bb9aa8a8e2a922237c3","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-26 20:50:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e6fc4d8f65acb2c5317d0a00344e7a37f124809a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-26 21:38:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a00b1822ddf5042afcf0c406a9a64f60dda213f2","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-26 22:01:42.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3884d51b8361e186611ad13788de50c590b8b059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:41:18.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9fa906f0d94e1bab8f20184bf3b3a4 : SUCCESS in 4m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/52b569892c2f454bb70d572893c3b61a : SUCCESS in 1h 05m 46s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4fce37143003406b86ea77a0379e248c : SUCCESS in 52m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/41fb22a2db654806a1865dfdfda92cb5 : SUCCESS in 1h 03m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f358df6de1794752aae8d8877d565a6a : SUCCESS in 52m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/daaf47917bdb413bbeab2427c0277285 : FAILURE in 1h 05m 44s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/22fdec0f46014e9ea6b391314c47fb67 : SUCCESS in 1h 31m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/291e321dd4754e0fb33c306df59f9366 : SUCCESS in 53m 03s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/7ccb315aeacb44d8820955cbd7418a32 : FAILURE in 49m 17s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d271fbf07594f52a0bfa37dcd834c23 : SUCCESS in 1h 33m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9b6a1a66947548f6aad2a1fde2a52514 : SUCCESS in 1h 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"deaacca45cf81a273694b72c5d36c8aaf8ed882c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-26 23:41:34.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/15/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2828f06/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/15/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5909670/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"6a971e412ceabb2509112de401022e7055b660dc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-27 21:13:36.000000000","message":"Patch Set 15: Code-Review+2\n\nLooks great thanks for working on this!","accounts_in_message":[],"_revision_number":15},{"id":"ab7bc356f82b02e76c0e34db5752543643fcbb99","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-05-29 11:34:07.000000000","message":"Patch Set 15: Code-Review+2\n\nLooks good! Thanks Victoria\nI\u0027ve noticed that it needs a rebase... Could you please do it? :)","accounts_in_message":[],"_revision_number":15},{"id":"1ae9493cb3402312e305a6cd7ec21d7b47070f34","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-29 13:38:45.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"cf3bff8ce043bd15896582acbf0a9f641fae4213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 15:23:24.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed18b2f533d45df912d78f436b79461 : SUCCESS in 5m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/50360d9bcc014c919d7537ab2bc0fa52 : FAILURE in 1h 05m 56s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3f0de37428694cba860acd59d145d618 : SUCCESS in 44m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/73530e0e018a43dea3010c3e239ee720 : FAILURE in 1h 15m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/381e5409c4364d03bf6cd84172411719 : SUCCESS in 52m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/675903f009d94ca6a1446cd6a7845d14 : FAILURE in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/404c2b85162c41ce8e236009fec27585 : FAILURE in 1h 22m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f98a46815d7d48c2a42bbf973a8baddd : SUCCESS in 1h 41m 36s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"39368316e48efc6b87563ba1dd186fc44526a4fe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-29 16:45:15.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e041b2677ed6c7f61cde13b135aa931b3a6ebfcb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-29 16:50:35.000000000","message":"Patch Set 16: Code-Review+2\n\nThe CephFS Job passes, and the change LGTM... thanks Victoria! \n\nThe voting LVM job failed due to known sporadic issues with booting a nova VM, and test_revert_to_latest_snapshot_1_2_55 failed likely because the snapshot wasn\u0027t in an \"available\" state from its \"restoring\" phase - we\u0027re aggressive about cleaning up right after a test here, something that\u0027s unlikely to be done in a real cloud","accounts_in_message":[],"_revision_number":16},{"id":"ac5e9f176656e8dfaf224e2c872ee1b1e17fc536","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-05-29 17:44:12.000000000","message":"Patch Set 16: Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":16},{"id":"14ac0d192b28df46203648652b7dae82acd36c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:29:04.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6f40ff3b104a35b9298d3a60c73d20 : SUCCESS in 5m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f7829d1307c04fe2b0aee89a29de4bba : SUCCESS in 1h 03m 31s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/46d0efce15d1496289da0a6a4c48551b : SUCCESS in 47m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4066e89276664659bfeb4f3570b915c2 : SUCCESS in 56m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/22e1593bd8a64e23808b8a9d4f7a3756 : SUCCESS in 47m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d02eb15df6c4cb7b7cffb3e1b7b17f5 : FAILURE in 1h 17m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5c95d343b90e4e4089f6ea20d076ae06 : FAILURE in 1h 39m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/39273173f3d04361a7c9cf99e1bec93c : SUCCESS in 1h 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9f95a7b8d0cd7b022c6c814f39b8f4f0571cbd5f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-05-29 18:32:27.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nlet\u0027s merge this, thanks Victoria!","accounts_in_message":[],"_revision_number":16},{"id":"bf53029d2198d843ab8dfc31284916f6d8fd19b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:32:49.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"cc7fba3167d4f2fa5b1c8d08d79924c93d300cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 19:15:19.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c880363cb524bc2a16f95c5a7090264 : SUCCESS in 3m 05s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a93be5778cab423ea2aff0b13dca483b : SUCCESS in 34m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/16815ed1e04b45f6961aa12ac82179c3 : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":16},{"id":"fd5bd2d303c95196ad761c7d2a6d56326c987a32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 19:15:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"c5a41e97681e936818b37dd0e4ad873cd8f26c3c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-29 20:45:39.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/725680/16/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a1e7392/ : SUCCESS in 2h 16m 05s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/725680/16/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9522f90/ : SUCCESS in 1h 35m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"62ff787526020108608859d03ed83bb1300da280","revisions":{"bb0daba2ac909d48fc25f67e4a28409e21146814":{"kind":"REWORK","_number":1,"created":"2020-05-05 20:22:25.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/1"}}},"commit":{"parents":[{"commit":"d86d85f852599e5f0ba7d99d44953ec1d124984f","subject":"Merge \"Add native zuulv3 job for zfsonlinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d86d85f852599e5f0ba7d99d44953ec1d124984f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:22:21.000000000","tz":-180},"subject":"[WIP] Add native zuulv3 job for cephfs native","message":"[WIP] Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb0daba2ac909d48fc25f67e4a28409e21146814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb0daba2ac909d48fc25f67e4a28409e21146814"}]},"branch":"refs/heads/master"},"a60ad63e4bbf4168e2330b6c26ca42ba4e7e2684":{"kind":"REWORK","_number":2,"created":"2020-05-05 23:03:42.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/2"}}},"commit":{"parents":[{"commit":"d86d85f852599e5f0ba7d99d44953ec1d124984f","subject":"Merge \"Add native zuulv3 job for zfsonlinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d86d85f852599e5f0ba7d99d44953ec1d124984f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 23:03:26.000000000","tz":-180},"subject":"[WIP] Add native zuulv3 job for cephfs native","message":"[WIP] Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a60ad63e4bbf4168e2330b6c26ca42ba4e7e2684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a60ad63e4bbf4168e2330b6c26ca42ba4e7e2684"}]},"branch":"refs/heads/master"},"f324ddeb57f58534a90707b2a0a117bd1d9996ec":{"kind":"REWORK","_number":3,"created":"2020-05-05 23:09:15.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/3"}}},"commit":{"parents":[{"commit":"d86d85f852599e5f0ba7d99d44953ec1d124984f","subject":"Merge \"Add native zuulv3 job for zfsonlinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d86d85f852599e5f0ba7d99d44953ec1d124984f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 23:09:05.000000000","tz":-180},"subject":"[WIP] Add native zuulv3 job for cephfs native","message":"[WIP] Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f324ddeb57f58534a90707b2a0a117bd1d9996ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f324ddeb57f58534a90707b2a0a117bd1d9996ec"}]},"branch":"refs/heads/master"},"f7c7dc38c6009a6e65d2ba5370db02e11edc59dd":{"kind":"REWORK","_number":4,"created":"2020-05-06 23:51:44.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/4"}}},"commit":{"parents":[{"commit":"d86d85f852599e5f0ba7d99d44953ec1d124984f","subject":"Merge \"Add native zuulv3 job for zfsonlinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d86d85f852599e5f0ba7d99d44953ec1d124984f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-06 23:51:29.000000000","tz":-180},"subject":"[WIP] Add native zuulv2 job for cephfs native","message":"[WIP] Add native zuulv2 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7c7dc38c6009a6e65d2ba5370db02e11edc59dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7c7dc38c6009a6e65d2ba5370db02e11edc59dd"}]},"branch":"refs/heads/master"},"63abd831d2fe16139e545248a14d9e10340fe40c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-05-07 09:48:44.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/5"}}},"commit":{"parents":[{"commit":"d86d85f852599e5f0ba7d99d44953ec1d124984f","subject":"Merge \"Add native zuulv3 job for zfsonlinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d86d85f852599e5f0ba7d99d44953ec1d124984f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 09:48:26.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/63abd831d2fe16139e545248a14d9e10340fe40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/63abd831d2fe16139e545248a14d9e10340fe40c"}]},"branch":"refs/heads/master"},"6bdd21efa0a72fdf169d75efe2bf088d2c4d11b1":{"kind":"REWORK","_number":6,"created":"2020-05-08 15:00:38.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/6"}}},"commit":{"parents":[{"commit":"38d392969e1325e633d752f1a96571a9793fbf99","subject":"Merge \"Add native zuulv3 job for lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38d392969e1325e633d752f1a96571a9793fbf99"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-08 15:00:12.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bdd21efa0a72fdf169d75efe2bf088d2c4d11b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bdd21efa0a72fdf169d75efe2bf088d2c4d11b1"}]},"branch":"refs/heads/master"},"c2edf7c57deb4735aac6fc3154bfb62cdd878099":{"kind":"REWORK","_number":7,"created":"2020-05-08 15:01:55.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/7"}}},"commit":{"parents":[{"commit":"38d392969e1325e633d752f1a96571a9793fbf99","subject":"Merge \"Add native zuulv3 job for lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38d392969e1325e633d752f1a96571a9793fbf99"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-08 15:01:42.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c2edf7c57deb4735aac6fc3154bfb62cdd878099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c2edf7c57deb4735aac6fc3154bfb62cdd878099"}]},"branch":"refs/heads/master"},"1f55a50f1a6d57baaa90a3d1cd1a6bbaf43ef4a2":{"kind":"REWORK","_number":8,"created":"2020-05-08 15:09:43.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/8"}}},"commit":{"parents":[{"commit":"38d392969e1325e633d752f1a96571a9793fbf99","subject":"Merge \"Add native zuulv3 job for lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38d392969e1325e633d752f1a96571a9793fbf99"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-08 15:09:34.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1f55a50f1a6d57baaa90a3d1cd1a6bbaf43ef4a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1f55a50f1a6d57baaa90a3d1cd1a6bbaf43ef4a2"}]},"branch":"refs/heads/master"},"2fa6c444248fbf573ecaf5259fd411f5f8f33adb":{"kind":"REWORK","_number":9,"created":"2020-05-11 16:51:10.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/9"}}},"commit":{"parents":[{"commit":"7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de","subject":"Add native zuulv3 job for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-11 16:50:57.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2fa6c444248fbf573ecaf5259fd411f5f8f33adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2fa6c444248fbf573ecaf5259fd411f5f8f33adb"}]},"branch":"refs/heads/master"},"bd8b174127cd6576b070b196e621472368b926d6":{"kind":"REWORK","_number":10,"created":"2020-05-13 21:51:30.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/10"}}},"commit":{"parents":[{"commit":"7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de","subject":"Add native zuulv3 job for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-13 21:51:18.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd8b174127cd6576b070b196e621472368b926d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bd8b174127cd6576b070b196e621472368b926d6"}]},"branch":"refs/heads/master"},"71cda995caad9db62fe0dff5cb89134299fe0ac4":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-05-13 21:52:15.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/11"}}},"commit":{"parents":[{"commit":"17fcfb8965df323ce7876827d1633bd321e21291","subject":"Merge \"Fix dict.values usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17fcfb8965df323ce7876827d1633bd321e21291"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-13 21:52:06.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/71cda995caad9db62fe0dff5cb89134299fe0ac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/71cda995caad9db62fe0dff5cb89134299fe0ac4"}]},"branch":"refs/heads/master"},"867494ab9c0e17f8ea9e214703d9c96045c064cf":{"kind":"REWORK","_number":12,"created":"2020-05-15 00:07:28.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/12"}}},"commit":{"parents":[{"commit":"17fcfb8965df323ce7876827d1633bd321e21291","subject":"Merge \"Fix dict.values usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17fcfb8965df323ce7876827d1633bd321e21291"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-15 00:07:10.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/867494ab9c0e17f8ea9e214703d9c96045c064cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/867494ab9c0e17f8ea9e214703d9c96045c064cf"}]},"branch":"refs/heads/master"},"b709656354d2ffeaadc08ea9c0b57e29815d4e2b":{"kind":"REWORK","_number":13,"created":"2020-05-26 20:50:47.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/13"}}},"commit":{"parents":[{"commit":"1727f7a18b903cd0082b58796580ff698aa9d728","subject":"Cleanup access rule test skip conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1727f7a18b903cd0082b58796580ff698aa9d728"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 20:50:37.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b709656354d2ffeaadc08ea9c0b57e29815d4e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b709656354d2ffeaadc08ea9c0b57e29815d4e2b"}]},"branch":"refs/heads/master"},"c35102ba7ff8ae4b334501ab100bd57f59f23c55":{"kind":"REWORK","_number":14,"created":"2020-05-26 21:38:28.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/14"}}},"commit":{"parents":[{"commit":"1727f7a18b903cd0082b58796580ff698aa9d728","subject":"Cleanup access rule test skip conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1727f7a18b903cd0082b58796580ff698aa9d728"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:07:31.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c35102ba7ff8ae4b334501ab100bd57f59f23c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c35102ba7ff8ae4b334501ab100bd57f59f23c55"}]},"branch":"refs/heads/master"},"6aaa54a55a6691c496525d1a0d060d6866612de8":{"kind":"REWORK","_number":15,"created":"2020-05-26 22:01:42.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/15"}}},"commit":{"parents":[{"commit":"1727f7a18b903cd0082b58796580ff698aa9d728","subject":"Cleanup access rule test skip conditions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1727f7a18b903cd0082b58796580ff698aa9d728"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:53:41.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6aaa54a55a6691c496525d1a0d060d6866612de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6aaa54a55a6691c496525d1a0d060d6866612de8"}]},"branch":"refs/heads/master"},"62ff787526020108608859d03ed83bb1300da280":{"kind":"REWORK","_number":16,"created":"2020-05-29 13:38:45.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/80/725680/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/725680/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/725680/16"}}},"commit":{"parents":[{"commit":"0a40f10dc81ae0c4d0d2ae056416870d0a76a84b","subject":"Merge \"Zuulv3 generic driver testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a40f10dc81ae0c4d0d2ae056416870d0a76a84b"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-05 20:21:14.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-29 13:38:13.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs native","message":"Add native zuulv3 job for cephfs native\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I5fa4071b9fe3927a0c8e1ba7137af824ddfaf66d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62ff787526020108608859d03ed83bb1300da280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62ff787526020108608859d03ed83bb1300da280"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
