)]}'
{"id":"openstack%2Fmanila-tempest-plugin~725692","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib1067db3771bacb287517e370c83a13d8bea89b5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable-jobs","hashtags":[],"change_id":"Ib1067db3771bacb287517e370c83a13d8bea89b5","subject":"Add stable branch jobs on the plugins master gate","status":"MERGED","created":"2020-05-05 22:16:03.000000000","updated":"2021-06-23 22:35:15.000000000","submitted":"2021-06-23 21:34:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"725692-stable-jobs","meta_rev_id":"e3b50f270d60f20d4cbacbeec9bea47cde94a073","_number":725692,"virtual_id_number":725692,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-23 21:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-06-21 22:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-06-23 12:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-06-23 12:39:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 22:21:18.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-05-06 01:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 21:33:26.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-17 23:42:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-21 22:14:16.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-06-21 22:14:16.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-06-23 19:40:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"adbfae046a107671e1a334ad9452c8d6a6517d6d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 22:16:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f056596bc65fb4e2d973bbbddf8cb00c16733de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 22:16:14.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"bfb6e166dfe2b03257bb323aefbc048d9dd061b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 22:21:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d584eefd2f36f8318d4303b7bf52fc0b6fdf0d99","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 23:56:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea7a1bc8521f22cab35bcd1c34ae85a97bb05eef","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 23:59:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e1a127e7ad0b2522e30f1a60a18081cf86f2336f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 01:52:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8696e21735dd4259957552a6db8acf28 : SUCCESS in 3m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/721a24d96f9d4334bb99ad00f67834d3 : FAILURE in 1h 03m 16s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a33220793dde457fbbf547f1e542cc6e : SUCCESS in 1h 33m 24s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4f228618095d497ba59b0a30baede15f : FAILURE in 1h 05m 02s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e0e45bcbab7e4cb98da7b56d77a95d25 : SUCCESS in 51m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/acc399d8d76d44b8a3d1a9b8c809f976 : FAILURE in 58m 22s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f70e268a4c94c13b57633e2ae5f5406 : FAILURE in 1h 50m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6f99ab897fd8426bbf5bdbc2858cfad1 : SUCCESS in 26m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2191ef51c64a4da9ae279f6aed8b1ae8 : SUCCESS in 1h 09m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/44e56fa703014c6fb07adcd43670d43a : SUCCESS in 44m 56s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/6502c327418b4e8ca413bcf2ef1aa65d : SUCCESS in 1h 20m 03s\n- manila-tempest-minimal-dsvm-lvm-ussuri https://zuul.opendev.org/t/openstack/build/9de5472b0300486988811040639626ca : FAILURE in 1h 42m 57s\n- manila-tempest-minimal-dsvm-lvm-train https://zuul.opendev.org/t/openstack/build/8c65607745e54eecb74be8327a0dbfe9 : FAILURE in 1h 49m 35s\n- manila-tempest-minimal-dsvm-lvm-stein https://zuul.opendev.org/t/openstack/build/386f8572eed04774ac3e06b7c7083e68 : FAILURE in 1h 42m 39s","accounts_in_message":[],"_revision_number":2},{"id":"26214bab41df8fe8bee043bdeeced04cde5a8d54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-29 23:25:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5e7acfc67b66a419a4bc52f60b72afe2e36c4b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 01:26:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d9e6d2e05249f095dcf2ca65e4ee73 : SUCCESS in 3m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/654c7b629264425e83f89b02e98b9936 : SUCCESS in 32m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/783270905ca24f868787de11eb93b4f6 : SUCCESS in 41m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99804558321c468faf0c869a01933826 : SUCCESS in 56m 27s\n- manila-tempest-minimal-dsvm-lvm-ussuri https://zuul.opendev.org/t/openstack/build/ec05dd0d1a5a4ac59b277bede95505bd : FAILURE in 30m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8b676507e79f4b449f17c61be8177a92 : FAILURE in 1h 07m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/60a1b241e23f479a804d55877a4ea943 : SUCCESS in 1h 19m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0093a60477304152b824a47e2d9dca59 : SUCCESS in 1h 37m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6fad602a225240119756b6540da6c046 : FAILURE in 44m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/47bd86c19ac2475383fafa744c910f8d : FAILURE in 1h 56m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/eac6614985674475a91ed7916fdfe380 : SUCCESS in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94d50d38d77d83f78eab29fb2b4af1fcf697c57d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-30 13:52:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b3b60faede6b86d28e991d8c2ff78766a24ca5bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 15:55:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9657b0fd125647b7b5c388dee8769e04 : SUCCESS in 3m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/80e7f32f90d44d51b454e41020605974 : SUCCESS in 28m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e0f58a2725c1481cbd8a9c5e53fadc4a : SUCCESS in 35m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6a9b42a5ccb84ed4958f68d5eefe05fd : FAILURE in 53m 06s\n- manila-tempest-minimal-dsvm-lvm-ussuri https://zuul.opendev.org/t/openstack/build/8ca6a2e4096f4dde9fc62bade0283693 : FAILURE in 32m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3f214ef37eba437991481e38be815dfd : SUCCESS in 1h 31m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/675c9047bc094991ac5a567539dc80f8 : SUCCESS in 1h 10m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b93346f9a26c4a73af065ff1f43575a1 : SUCCESS in 1h 34m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ddafe588368d41d8a2950905d541164b : FAILURE in 47m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/607688245f944f6db979cf5aa3e25c42 : FAILURE in 1h 55m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/80dd02772e104bd38e0606a1cba38d60 : SUCCESS in 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b99b19394530b405d0340b662f7f9183920f625","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-05 00:48:59.000000000","message":"Patch Set 3: Workflow-1\n\nnot sure why victoria and wallaby stable jobs are not running. will debug in morning.","accounts_in_message":[],"_revision_number":3},{"id":"b78ed2be5720a6dc2ec85df9bf2232096e091221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-14 20:39:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5bdda22fcca361322b065ad937f9208ee0cb7207","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-14 20:41:44.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 3: Workflow-1\n\u003e \n\u003e not sure why victoria and wallaby stable jobs are not running. will debug in morning.\nParent job name was changed since the last review. \n\nI suspect we may need to adjust some configuration options; will monitor this run. Thanks Ghanshyam!","accounts_in_message":[],"_revision_number":4},{"id":"8e83982f81fc3dff25b31aa45f62ce868f35d94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 22:50:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c86f0189ed4e63bfd7aa4561b10f2f : SUCCESS in 3m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/81ad9a845b12428aad05f539895dc5a8 : SUCCESS in 47m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/78857dfc1502429caff36c37d7f9497a : FAILURE in 30m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0a9e69c19ce44de292457f5b20aad954 : SUCCESS in 51m 41s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2136029e833b4e83a75efa10d539fa0e : FAILURE in 1h 24m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9318abc95aff45078d5aa488726ded8c : FAILURE in 38m 03s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/c010dab6724b4875b51fbb91e180dc5b : FAILURE in 35m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0b23164762624ef69aeb7207f85a1c0e : SUCCESS in 1h 16m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9801462ac5ed43a69230b15d212c91eb : SUCCESS in 1h 20m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2232694567ca4dfaaa176edb26f07e0e : FAILURE in 1h 48m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bc95b89cdf244b6aacefaca92b3185f3 : FAILURE in 44m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd39ec1131e8472e9f22e683381893f1 : FAILURE in 2h 08m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1713712e90ce48ffa4e3dfb5fd9370a9 : SUCCESS in 22m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b7ad657f45bd3a13d7610c56346bea9fe0ffe998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-17 18:12:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b007de05e6ef520a4927ac64526bbd0602b7e172","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-17 19:20:23.000000000","message":"Patch Set 5: Workflow-1\n\nStill testing, will merge with https://review.opendev.org/c/openstack/manila-tempest-plugin/+/763282","accounts_in_message":[],"_revision_number":5},{"id":"b8914921e45dd8b653a22772643f126944bf20e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 19:47:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c7cb510b8f4ce09d4a4196b9529f18 : SUCCESS in 3m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ab872367358e48dc90fb77bb4eb7d68f : SUCCESS in 35m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/001d46d6a38941749b345a8840271887 : SUCCESS in 33m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1cea0624ac114cc9ac5d199a204bef92 : SUCCESS in 1h 02m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ce8544424ba848b6890431cadff479fd : SUCCESS in 1h 18m 05s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9aae6821f65042b997f88b4a01799236 : SUCCESS in 1h 32m 12s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/cc2f54d0df4c40aab1e09b330b8e3634 : SUCCESS in 57m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/12b61423fbb244178030fb1917c28d1a : SUCCESS in 1h 04m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2e103ac1c6584f8abb0ae5af004b762e : FAILURE in 55m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6be40d470f1c4bbf92d2c91fdc4fb55e : SUCCESS in 1h 21m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0f4b6cee4071469393582d6bf0c18927 : FAILURE in 44m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f198e88c84b44640871675a42af4580c : FAILURE in 55m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4c8b1b67e6284ad2939f011391ef4c0d : SUCCESS in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e8decf9cbf9847b110aa57ee31c8f7ce8ef308b4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-17 21:33:26.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/92/725692/5/check/DellEMC_VNX_Manila/39ba818/DellEMC_VNX_Manila/38 : SUCCESS in 1h 11m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"2f938a704f940fab33b6c9aaa52e490a1227f849","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-17 23:42:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/92/725692/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/dfd3b25/ : SUCCESS in 2h 22m 04s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/92/725692/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/09c3bad/ : SUCCESS in 2h 31m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"33df5fef59b19c692f9cc75468d7e063deec7d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-18 01:50:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"568a248a4fae34b907497588906d088755022396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 03:50:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092fe4de3310484ba144f56ce0b4ea1f : SUCCESS in 3m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6e114fa0eace4b91a7910baa9a247a5b : SUCCESS in 24m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3b47714f84214fd5b9fa500407e43ac0 : SUCCESS in 36m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aa47c40da4a243aa8540d9e1b94a7b00 : FAILURE in 1h 45m 32s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f169da7be9734389b23144d6ac9c0a89 : FAILURE in 1h 14m 02s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ae8e9c2762e8412c8b9c515ddfb2a17c : FAILURE in 35m 21s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/2989e94b702e4ea0a8c0414833f65939 : FAILURE in 1h 10m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aef1c19fa57d4b798c5bb69d451cbc40 : SUCCESS in 1h 03m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5e9f2f060f58411294410c7272689734 : FAILURE in 1h 37m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e32fe72282ae4c0f94997ac7ed4b92fe : FAILURE in 1h 40m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2d88095aee0b43f2b800f122b7d1f9f6 : FAILURE in 37m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/46390b9dd5854c458385bd26dab51eaa : FAILURE in 1h 57m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/eb3872ffdc03474a96fe4c5d01faf55e : SUCCESS in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"437cac32b3e79bc768f616c8972fa7dac09e9f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-18 05:00:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"470af97b36870af613696224c4df25963c2ed2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 06:44:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cfc0513ebf40ce8e1472ca0916cad3 : SUCCESS in 3m 31s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a0a44a57d27547969cdefc6a6392b5da : SUCCESS in 31m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7d45191cff9b41a7b30717fe2eb03c5e : SUCCESS in 42m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/111f677b98de4a0199182584f5ecbc3e : SUCCESS in 1h 11m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5954d0878e8d45a8b2e93a433054ed97 : FAILURE in 1h 38m 29s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ae683dbc7c2a48f5891ac6d36ec2ea04 : SUCCESS in 1h 11m 01s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/fd4b19aa7adf46aeb8078481a83da783 : SUCCESS in 1h 15m 54s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/940b38599a2b4beeb49205c09bc769d9 : FAILURE in 1h 21m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9aa4458e59074fbf88fe22bdaf2ce2c5 : SUCCESS in 1h 10m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a69599036cd1478ba3a97b9206e6d972 : SUCCESS in 1h 37m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/246eef37e6734d27b63e49c6a8fd9c63 : FAILURE in 37m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c267f8b285ff4e2bbc4e7f0c09def3b5 : FAILURE in 1h 26m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3224d08364734cc4b98c48e72fd2ca0c : FAILURE in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd700e62d879643536dcd557291abf7767abdac0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-18 15:32:16.000000000","message":"Patch Set 7:\n\nrecheck\n\nKnown sporadic failures..","accounts_in_message":[],"_revision_number":7},{"id":"5519bd1f85849eb7f8e1a48e11e7d4f466555114","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-18 16:07:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9ca7bd6194ba9ffd46f19fce3cc6accb5ab0f4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 17:14:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69955635ec18422a8c12029ebcc6c816 : SUCCESS in 3m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/98e3a2109ab74ba4b218ef596893ee12 : SUCCESS in 40m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/78741139c5c548af864cb95b3d7aaced : SUCCESS in 30m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/92ef3b2198ef4f0db0e0280a9db0da38 : SUCCESS in 1h 02m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7830f4b8c90c4d7cb5b6bb97d903ca36 : SUCCESS in 1h 02m 51s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/bbf23f4b3c8c460dbe5f1d075fb039ce : SUCCESS in 1h 02m 06s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/08428c40da304cb29879060a905b193d : SUCCESS in 57m 22s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a36c0d0e818a426d867a2f0b22655355 : SUCCESS in 1h 08m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6475dcfd2b1b443fab8dde9dbda77fe4 : SUCCESS in 1h 02m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e6ef00b04cb5414d931fd7134d1d7e35 : SUCCESS in 1h 27m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4e4248576da645c599ef336c5e9415c5 : FAILURE in 41m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/820db502b5f34542920cc60422acf722 : FAILURE in 1h 40m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5539726d502d40c586dcabdb9932a7a2 : SUCCESS in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"313a36b04d73272907c6acdfcb9de01585a88dc1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-18 17:31:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/92/725692/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6b8b8a3/ : FAILURE in 17m 01s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/92/725692/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/82c9992/ : FAILURE in 8m 13s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"00f79830e1a73e6c72bb7429dc679bc85db29665","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-18 18:05:16.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/92/725692/7/check/DellEMC_VNX_Manila/82e42df/DellEMC_VNX_Manila/40 : FAILURE in 50m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"5f65de4954420bb7e062b30781f250d1002812be","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-18 19:12:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"95e9adacc2b12cf8433913b4b8d751d54785f9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-19 17:19:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"28f7597c8b9464e52f5053b555f7490f1e812f66","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-19 17:59:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dc3a814a7076b67aeb355c9f300bfa15fbb4f14c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:59:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf2cc4effe44e1197afae2dbc5e3230 : SUCCESS in 6m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e57250aac4b74e4e8b7de2cb395fdb0e : TIMED_OUT in 1h 03m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f228713ff7644309f80c31c3d9dec90 : SUCCESS in 30m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bb9bc2d1fb6f49cea5ab9a37db8d6681 : SUCCESS in 1h 27m 57s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/69249a7446f645839bb07762c3612dbb : SUCCESS in 49m 58s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/954ebdf2188e4aaf8fef7c2e5adeb7ed : SUCCESS in 57m 37s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/19045238b88a4194a1a3d0215d6af8e8 : SUCCESS in 1h 16m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b1bfcd43c0ba4d9abaa668813275cf0b : SUCCESS in 1h 17m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/acf406ae969e4b1dbc8f27c0dbffeae7 : SUCCESS in 1h 23m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7ded6498e0ff4c2593ef46f706f04b1f : SUCCESS in 1h 46m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0ce73caea1b8437f8e3fc1b5f2d5c4ba : FAILURE in 44m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/61a8e9d943e64d67a6e5f61b6733a806 : TIMED_OUT in 3h 34m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/32a6cdec488745ad8d3d56f6098577d4 : FAILURE in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"08ac140bfabd683845a5e5b69b0d6635f45f7a00","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-20 15:40:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"6e94afb841a53b7d51d506f50151234cc1013b71","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-20 16:56:33.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cc1ff92985d9c31ef886f443b2e1d07f7f7904a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:43:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf41e4bfd93c40b4860dd3477950102b : SUCCESS in 4m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7cb65638300a4e398849f91af8de1a6f : SUCCESS in 22m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8c6ff9c041754f50bad48a46a6f4e3e4 : SUCCESS in 45m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4cf3bc27771f446fa0ef65649c088ba7 : SUCCESS in 53m 21s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8d1cd68b2d074fd49cfce424d8358231 : SUCCESS in 1h 24m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/095413f32c714a59a31c0d696050ed79 : SUCCESS in 1h 20m 11s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/bda4dbee69094bdba8609c2715762c14 : FAILURE in 1h 18m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5ea6bac2d8fb4e439b1896b6d44116f8 : SUCCESS in 59m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2b3ea8e7c3414fa39d6e89c5284af0b8 : SUCCESS in 58m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b15cff7d4fea4f77b9c441f1375944bb : SUCCESS in 1h 54m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/af177fd8aacc4b58894ec9d20c88ad05 : FAILURE in 49m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e43607d070d74869a55025320c6fd980 : FAILURE in 2h 54m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bce58338c25d4dd19109e097e0a06ab1 : SUCCESS in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8442804d75a925dfe2cb4360adc2e18e25052ba3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-20 19:17:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"314713db7ef1e82f3b9c6ffc3d71ce29cbd39060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-21 18:59:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e32dbdf20e9c85e8e7e13c06eaaca55bf0f8a903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 21:53:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91f60a12a4d4c2ba5ce7f95d928a058 : SUCCESS in 3m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8eeccb85668e4444adcd2f618bd3476f : SUCCESS in 35m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/87ea99d267d7451ca209c7ea102a7a32 : SUCCESS in 37m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f170f7eb2f6249d4ad6386abce9c8961 : SUCCESS in 1h 17m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9ebb32ad3a17469ba85b6afae32c7f7c : SUCCESS in 1h 14m 29s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/92021d77345f4721b99ae35826e1b368 : SUCCESS in 1h 20m 05s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/e923e0e82d56435f901878e82f09a59c : SUCCESS in 1h 09m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec7970fdc185400a87b24e37432cc4f8 : SUCCESS in 1h 21m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cbeeecb3ae584032b821e0ea9c3eec4f : SUCCESS in 1h 19m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7a7c6afb5d654c80afdbd0289cdbec0d : FAILURE in 31m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/204dccdb4d3a4f429bbd357d5cb51381 : SUCCESS in 43m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0a9466682db34413ac614d6c063b48d7 : FAILURE in 2h 46m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a3b7bbf8628e4fb4ab99641efa63bfdc : FAILURE in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8ab130788ad5a0fa9d65ec527c579ca238a3724c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-21 21:53:25.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/92/725692/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/41cef3f/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/92/725692/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/99d7ba6/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"e647610b90f22ff0066527a929fc0548c43f212a","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-06-21 22:44:18.000000000","message":"Patch Set 9: Code-Review+2\n\nLooks good, thanks!","accounts_in_message":[],"_revision_number":9},{"id":"f16df95c449d7a36ec698ed559ab569b266e8ca5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-22 00:11:14.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/92/725692/9/check/DellEMC_VNX_Manila/76c83cf/DellEMC_VNX_Manila/74 : FAILURE in 59m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"31f7b7d279d42d582f3e286296b00b4193dbac92","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-06-23 12:39:35.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nThanks Goutham,\nLGTM!","accounts_in_message":[],"_revision_number":9},{"id":"c697ea6a4191ce2ccdafe486efde49270ebfe328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 12:39:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7c1b87391ae9d806d11636839a4f1032c44a9e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 13:40:45.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67876b9161044da2bda5738c16815864 : SUCCESS in 5m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bf17a066e99b4440bb93fbd66ae2dc58 : SUCCESS in 33m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cdbfd2a569ed4f29891dc84c14c90f51 : FAILURE in 37m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/587c2f6bbea34b32b96148446b60e099 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":9},{"id":"6f6fabc085b0f695247e9735518bbfe457816c82","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-23 15:18:24.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"45f05d1f95b582874fc877b37e53369e5d9a84f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 17:08:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6d29b323ae496d838b78c0a36d3f7c : SUCCESS in 6m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/59ad5592ccf24d1d89b83c1530bdfa4e : SUCCESS in 29m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c46ec09dbf3649ddb5d2248a8b0040f9 : SUCCESS in 35m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1ba4c4b00e9440a9bffc2e8fe088cf84 : SUCCESS in 1h 12m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a019933b82614e459683ddbbb309e94e : SUCCESS in 1h 11m 44s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1b9fa718817b48f3856da0310dcc6f17 : SUCCESS in 1h 05m 12s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/ffb5772600c74a5fb9d1469f7e0a07f3 : SUCCESS in 1h 10m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7720b1533de42b596cc8bd791f92527 : SUCCESS in 1h 15m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e4712d6ea8a44ef08a15999a77f43a01 : FAILURE in 1h 18m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ba522f1cd8934decafe439fe0e2389eb : FAILURE in 26m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c1a6ae9b1b754a4dbb4cbedd2ae3c528 : SUCCESS in 49m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8c9ea2cd1c6428d8ff487ac6a2f0db4 : FAILURE in 1h 48m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0e963b66fe7248bba1089ada8ef89f87 : SUCCESS in 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2eb4d3eb781291c67ac8a93873ec4d5aac089f57","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-23 17:09:18.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/92/725692/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d81dae6/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/92/725692/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/371a948/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"ff1c8a15a82b1df53d7e4ccc8bd66019b210ccf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 17:09:27.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a05660f46c41f0c2d5b6cd25941ec5fc3243e1f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 18:16:08.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8949057c2e48b4a8b96f9e98e1b8d3 : SUCCESS in 4m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/51b4c29258d74b6db6f16ace950f2cbd : SUCCESS in 41m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/68e6d9a51bb9421a921b8491d90423af : SUCCESS in 34m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6a1072297c7141b2ba27d3a1f621536c : FAILURE in 1h 04m 43s","accounts_in_message":[],"_revision_number":9},{"id":"6c99bde04ee30f4d4595de44da9efe22f4a2bea9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-23 18:29:13.000000000","message":"Patch Set 9:\n\nrecheck\n\nThird time lucky?","accounts_in_message":[],"_revision_number":9},{"id":"cb66b7aaba6b71907a651a1f3553fe9f58245224","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-23 18:31:20.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/92/725692/9/check/DellEMC_VNX_Manila/4021206/DellEMC_VNX_Manila/79 : FAILURE in 1h 21m 59s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"79d5fc41fa24de83749afdab92782cb7d732515b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-23 19:40:04.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/92/725692/9/check/DellEMC_Unity_Manila_dhss_true/c7cc68b/DellEMC_Unity_Manila_dhss_true/136 : FAILURE in 2h 10m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/92/725692/9/check/DellEMC_Unity_Manila_dhss_false/4e6631f/DellEMC_Unity_Manila_dhss_false/134 : FAILURE in 1h 21m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"73fa3951970bc8c4fffdf9979c469b5e0544b56a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 20:30:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98aa6b948d1f47199b47ba61d952c0ea : SUCCESS in 4m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eda9a9e4a29e4e51a1345529b4a6b77a : SUCCESS in 31m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a2bc16df902744e3affa1d69de118a2d : SUCCESS in 35m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5da982ecf4e341ab90ef437419ca3501 : SUCCESS in 1h 09m 46s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d1d4df108ed046d49804be4a65685380 : SUCCESS in 1h 01m 00s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/3eca8998500342299151a554863bf8d0 : SUCCESS in 1h 00m 22s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/09c138831e10401e8cbc4067ee48c0be : SUCCESS in 1h 03m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d6d67271afb14cf6be334f0cd2eb4c53 : FAILURE in 1h 17m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2d8084a69b54447eb7ba651ea10ea3d7 : SUCCESS in 1h 31m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e4300bbe4e884266b9049ec60a861f14 : FAILURE in 29m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e5d380fd527845e687574446968837a6 : SUCCESS in 48m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/acadf3fe65074c288fc4d866359feb5c : FAILURE in 1h 54m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/210be3218cb74d50836c19780734d6f0 : SUCCESS in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1ce9b24df1e918b21bb2f7912a0662031037d9fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 20:30:39.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"59fc408a9353d230a62dce6a9dc8688fd24965bc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-23 20:30:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/92/725692/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3708264/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/92/725692/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3877da0/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"da5c0045a8cc674b3682562c3ecfbcc0ccd31c6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:34:37.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a7b6de45bc43e8b056335803500606 : SUCCESS in 3m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/49d08db630dd4686b65763fcd111bb66 : SUCCESS in 27m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9075efa545a7486d9995fd42070f6c0c : SUCCESS in 34m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/789f9a2733bb4000b937d2a4eebf8b57 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":9},{"id":"a785b8295f05c57bd6ef2cca445e55455cac718c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:34:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"0d90693f23c997300474db9af43e6142ea4541dd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-23 21:37:08.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/92/725692/9/check/DellEMC_VNX_Manila/98e4d60/DellEMC_VNX_Manila/80 : FAILURE in 1h 06m 32s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"e3b50f270d60f20d4cbacbeec9bea47cde94a073","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-23 22:35:15.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/92/725692/9/check/DellEMC_Unity_Manila_dhss_true/c99201e/DellEMC_Unity_Manila_dhss_true/137 : FAILURE in 2h 04m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/92/725692/9/check/DellEMC_Unity_Manila_dhss_false/3b51c7b/DellEMC_Unity_Manila_dhss_false/135 : FAILURE in 1h 10m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"cf95d4089769650bd4da4254dbd578b33b8d77d3","revisions":{"69b8a5d24c50aa51dec013f42c9377dc31bdf301":{"kind":"REWORK","_number":1,"created":"2020-05-05 22:16:03.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/725692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/1"}}},"commit":{"parents":[{"commit":"7636a1fce4c2b49a03338097f110fadbb3f81206","subject":"Specifies the appropriate ip address in case of multi-networking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7636a1fce4c2b49a03338097f110fadbb3f81206"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:04:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:14:39.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69b8a5d24c50aa51dec013f42c9377dc31bdf301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69b8a5d24c50aa51dec013f42c9377dc31bdf301"}]},"branch":"refs/heads/master"},"425f91e92dfc5dbbb99d1c61fba9876336cb8404":{"kind":"REWORK","_number":2,"created":"2020-05-05 23:59:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/725692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/2"}}},"commit":{"parents":[{"commit":"11b8687ed11cf8b6e08bb2adeff6f84f0d87f1b4","subject":"Merge \"TrivialFix: Remove unnecessary import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/11b8687ed11cf8b6e08bb2adeff6f84f0d87f1b4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 23:58:06.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 23:58:06.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/425f91e92dfc5dbbb99d1c61fba9876336cb8404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/425f91e92dfc5dbbb99d1c61fba9876336cb8404"}]},"branch":"refs/heads/master"},"f50a24c88fc998a973488aadd46d0b9f4a05d0e1":{"kind":"REWORK","_number":3,"created":"2021-04-29 23:25:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/725692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/3"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:25:40.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby- train moving to EM next week)\nuse the Tempest and plugins master version for their testing.\nTo make sure master version (new tests or any changes) is\ncompatible with stable branch testing, we need to add the\nstable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f50a24c88fc998a973488aadd46d0b9f4a05d0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f50a24c88fc998a973488aadd46d0b9f4a05d0e1"}]},"branch":"refs/heads/master"},"4a4f755b202f531383072398991baeddb9c8d96b":{"kind":"REWORK","_number":4,"created":"2021-05-14 20:39:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/725692/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/4"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-14 20:39:36.000000000","tz":-420},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby- train moving to EM next week)\nuse the Tempest and plugins master version for their testing.\nTo make sure master version (new tests or any changes) is\ncompatible with stable branch testing, we need to add the\nstable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a4f755b202f531383072398991baeddb9c8d96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a4f755b202f531383072398991baeddb9c8d96b"}]},"branch":"refs/heads/master"},"2f4c17412b5388bf04c5112585fb241559c69ade":{"kind":"REWORK","_number":5,"created":"2021-05-17 18:12:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/725692/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/5"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-17 18:12:09.000000000","tz":-420},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby) use the Tempest and\nplugin\u0027s master version for their testing.\n\nTo make sure master version (new tests or any changes)\nare compatible with stable branch testing, we need to\nadd the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the\nstable branches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we\nare running only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f4c17412b5388bf04c5112585fb241559c69ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2f4c17412b5388bf04c5112585fb241559c69ade"}]},"branch":"refs/heads/master"},"2950d99c5c1fbd1e6c7745ce367b5ffc87800dea":{"kind":"REWORK","_number":6,"created":"2021-05-18 01:50:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/725692/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/6"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-18 01:50:28.000000000","tz":-420},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby) use the Tempest and\nplugin\u0027s master version for their testing.\n\nTo make sure master version (new tests or any changes)\nare compatible with stable branch testing, we need to\nadd the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the\nstable branches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we\nare running only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nThis change also sets branch specific tempest config\noptions for the stable branch test jobs.\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2950d99c5c1fbd1e6c7745ce367b5ffc87800dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2950d99c5c1fbd1e6c7745ce367b5ffc87800dea"}]},"branch":"refs/heads/master"},"186f4511530ce4ead156eeaf572e63cb0f3dda82":{"kind":"REWORK","_number":7,"created":"2021-05-18 05:00:49.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/725692/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/7"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-18 05:00:38.000000000","tz":-420},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby) use the Tempest and\nplugin\u0027s master version for their testing.\n\nTo make sure master version (new tests or any changes)\nare compatible with stable branch testing, we need to\nadd the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the\nstable branches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we\nare running only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nThis change also sets branch specific tempest config\noptions for the stable branch test jobs.\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/186f4511530ce4ead156eeaf572e63cb0f3dda82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/186f4511530ce4ead156eeaf572e63cb0f3dda82"}]},"branch":"refs/heads/master"},"0da317a438fb8f7385787dcaf1e89d14fa789774":{"kind":"REWORK","_number":8,"created":"2021-05-19 17:19:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/725692/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/8"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-19 17:18:51.000000000","tz":-420},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby) use the Tempest and\nplugin\u0027s master version for their testing.\n\nTo make sure master version (new tests or any changes)\nare compatible with stable branch testing, we need to\nadd the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the\nstable branches in same way [1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we\nare running only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0da317a438fb8f7385787dcaf1e89d14fa789774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0da317a438fb8f7385787dcaf1e89d14fa789774"}]},"branch":"refs/heads/master"},"cf95d4089769650bd4da4254dbd578b33b8d77d3":{"kind":"REWORK","_number":9,"created":"2021-06-21 18:59:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/725692/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/92/725692/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/92/725692/9"}}},"commit":{"parents":[{"commit":"602f101c14ea9130226359942636fb5bf0c9e782","subject":"Fix CephFS snapshot cloning scenario test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/602f101c14ea9130226359942636fb5bf0c9e782"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-04-29 23:24:19.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-06-21 18:59:32.000000000","tz":-420},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(ussuri, victoria, wallaby) use the Tempest and\nplugin\u0027s master version for their testing.\n\nTo make sure master version (new tests or any changes)\nare compatible with stable branch testing, we need to\nadd the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the\nstable branches in same way [1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we\nare running only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n[1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: Ib1067db3771bacb287517e370c83a13d8bea89b5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf95d4089769650bd4da4254dbd578b33b8d77d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf95d4089769650bd4da4254dbd578b33b8d77d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
