)]}'
{"id":"openstack%2Fmanila-tempest-plugin~726088","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I700731e714db09f757d21d31ba3ddce62d70e48d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I700731e714db09f757d21d31ba3ddce62d70e48d","subject":"Add native zuulv3 job for cephfs nfs","status":"MERGED","created":"2020-05-07 10:38:02.000000000","updated":"2020-07-13 18:37:22.000000000","submitted":"2020-07-13 18:02:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726088-1594663326886-773feb59","meta_rev_id":"0f7f564586209006f3f3f5951d225d6a60599e78","_number":726088,"virtual_id_number":726088,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-13 17:05:48.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-07-13 18:37:22.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 18:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-13 10:34:39.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-13 10:34:43.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 23:57:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-13 05:16:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-07-13 10:34:43.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-07-13 17:05:48.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 18:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 18:37:22.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a74fe0b9b537f3ffbd300e3c38dd70da544625f","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-07 10:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4e33fff5ff7251a11bd589ae9a5acc7a190ad11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 12:35:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ad37174df94a198c43df34f545158c : SUCCESS in 4m 05s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e3e07f1322be457f8b51a158e6919383 : FAILURE in 40m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7105924f0ace42e897234c98f51b484d : FAILURE in 1h 30m 23s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3e24f01b610f445c9dc472dbde064aba : SUCCESS in 1h 49m 17s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ac2e18128eba46c7917796389e7d181f : SUCCESS in 58m 29s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d751a07ae50e4495830d6daa3bac5b20 : FAILURE in 1h 02m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/82bd04d66f324ab1b50d3ef052df4a6a : FAILURE in 45m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6eb0482c79c14edd897a176324d17a35 : FAILURE in 1h 52m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f3f8e7cd6a95481aaf904ff330f05038 : SUCCESS in 32m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ba8f7158162f40e7b8096fe45c164741 : SUCCESS in 49m 15s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5d002165e2c243c499aa02332e478335 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4a1a618ca2523cb6c0d270fa0549c5c3c91e4512","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-07 14:42:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/dae5030/ : SUCCESS in 2h 06m 09s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6b98947/ : SUCCESS in 1h 33m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"119e8e85b7469d78d98e5d7aad80ba01ae652977","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-08 15:08:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c031bcdb7d6731d55c294d1ee8d9a499af351ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 15:09:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 107, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 108, column 1","accounts_in_message":[],"_revision_number":2},{"id":"7f75680bf93b28ca8fff6f7b6ba7d67664cf82da","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-08 15:11:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4d16dcde3b69c40943e08582e999530ac4928f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 17:01:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f183dd8aeb342a598b5dce1e83c39e0 : SUCCESS in 5m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0992399c154c453ba154240dc12d4cca : FAILURE in 1h 10m 39s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c922f001fa29446ea9ebd2090faea0cc : SUCCESS in 47m 36s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/06b4e96e5e454a85b72e6dce577a0e0c : FAILURE in 34m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/308bd38feb054b30ac81f96a4237e0bf : FAILURE in 1h 00m 20s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/511d8679328b4b589f30cd4ed17a1614 : FAILURE in 1h 47m 27s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/2288f3515ecb47afa194d7fd8c897bc9 : SUCCESS in 1h 10m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/be2c0564f32440149cbd082fd487d643 : SUCCESS in 1h 11m 18s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/27ee79c64133498a93554c7eb5353edf : FAILURE in 45m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07572d0e916c42f58cb951d901c39c2c : FAILURE in 1h 43m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/975d6bd629d84647ad1fcd38385ac4d7 : SUCCESS in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7869bca3d24bfe84d23cbcd9ead70cb9c594cdc7","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-11 17:04:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e30e449f75d10597cd561a67162e64e71fb55316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:47:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc931da16b549e49f8c0293bc242e36 : SUCCESS in 9m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ed498757ff9f4de28f983287901e5088 : SUCCESS in 1h 14m 10s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/735e1f18b93240ceaaeb5ba4e3e8dc98 : SUCCESS in 48m 01s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bed26985de4148f3a54127b0f4aeeeec : FAILURE in 29m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cf940288beaf4f4d97a4017afa71df29 : SUCCESS in 1h 07m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7249ea1748964cf4bca1749fac64ce26 : SUCCESS in 1h 03m 07s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b1920d080d4e47b9aa9a164912f32cef : FAILURE in 1h 39m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e2f6273044bb42b980f51da2645af981 : SUCCESS in 1h 09m 54s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/9c3f87278fa247729c869bc75caa5cc3 : FAILURE in 43m 12s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ad4d0b1be4a4c1c98db0fc2edc9bcd9 : FAILURE in 1h 33m 54s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/87b936cdc77440489817251a6eb90d75 : SUCCESS in 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52efb73ad1911e40e380248a9626233a4f377355","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-11 21:05:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3af5f64/ : FAILURE in 2h 17m 15s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/adc3cdb/ : FAILURE in 1h 40m 38s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"2bf363727d85bb93a9fd90cb5f8e3e05b43f5c11","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-13 06:16:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)\n\nThank you for working on this! Please see comments inline..","accounts_in_message":[],"_revision_number":4},{"id":"eed4361e9376480c709e975a7b790cd2a783e091","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 10:09:00.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"2ffbd803f68ad6df5510f8eff37ca9f0dc6f3301","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 22:05:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"564823dba10bc8f4789794f6580658c8cfaeeceb","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 22:06:11.000000000","message":"Patch Set 5: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":5},{"id":"f93fd6d1be12217a3e2625239e95d032c3fd5ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 22:25:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b4fb6a733e43ba880a7b28aa3e5396 : SUCCESS in 3m 55s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e436bf55457841bb9a84452752218c82 : FAILURE in 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"74480392cccf882067e354fdf556b415b00fcffd","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 22:35:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d4898dd71c8679610f4ed27e1d0e43dac34b5ee8","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 22:45:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b3848c86cd0c199905833e995a18f1ace5c1c50e","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-13 22:46:55.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"29ecc7b9e1cf071973ed43e75fe8c0c45dbe25db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 00:21:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c878f0da493145dda0589a61a1f0fa6f : SUCCESS in 4m 21s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a832de0bc3e44b9e821d45233353db13 : FAILURE in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c3111110084f110da6076f3accb95f34ce7f4211","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-14 04:08:09.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/71065bb/ : SUCCESS in 3h 46m 06s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d73ff45/ : SUCCESS in 1h 36m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"b11d6a833167df7142266776d7cd490c0f1dfa9d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-14 14:41:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b3d6618350ff59855b642512640d33ca7a21d2f3","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-14 14:43:56.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"235f817f0743ce6b355a997827e30e759a8f8de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 15:29:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fea1ef980c4442596836b77e357c20f : SUCCESS in 3m 51s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bace705fedd944c68fd2a7eca9eb92d1 : FAILURE in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bfe269c70f2e6703037bbed15e4f133239611a93","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-14 17:33:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3263086/ : SUCCESS in 2h 03m 45s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/68ddcb7/ : SUCCESS in 1h 31m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"77f01d1935b0c8e998d355ab78ecce06154b4076","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-14 18:48:51.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7e26239fc93ce0ff7f58198fbb9d9f835a570405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:35:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b54d6ae345442d9a7d4056714686606 : SUCCESS in 3m 35s\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/234088acdc384db480ca237f19a8103b : FAILURE in 45m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"233a6ec8d62842c8cc7baee2dadd0873a09daf6b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-14 21:46:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/99074c9/ : FAILURE in 2h 10m 12s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2492da0/ : FAILURE in 1h 37m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"57ed910683cb12ce8cd0c92dd6833f9f9cf9a70e","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-06-23 19:00:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"33711b5ce561fde66d595f6135565f4072920f35","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-06-23 20:10:13.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"a4f48cb3fc90423a14b05fab3f3b3fbe562f643d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 20:49:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16baf8d76e3b4ecbb67851c8679ab07a : SUCCESS in 4m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/275a05a168b74e009dead3959a81cd60 : SUCCESS in 1h 27m 16s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3acf3a1265534a21805ac25bc8e79fea : SUCCESS in 52m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b89978bba1524bff875b1aeaf3683379 : FAILURE in 1h 05m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9c4ff2b10f46480f85b9555ad06b4f0f : SUCCESS in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7c22a5fb2fc34bf29861a1edda634af8 : FAILURE in 43m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/88449720009c46d68232911ebd5bda1f : FAILURE in 55m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2ecaf648138a4898b3040444034ab8a4 : FAILURE in 1h 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0ce0777a97ca7f62c14771e13281b8cfecfbd448","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-06-23 22:01:01.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/726088/9/check/DellEMC_VNX_Manila/baa8200/DellEMC_VNX_Manila/58 : SUCCESS in 1h 11m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"e745c2f6132ebf03c5ab335ec4415b652f42d670","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-23 22:52:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/0c83779/ : SUCCESS in 2h 03m 12s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e3162a5/ : FAILURE in 21m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"4a881bb028986784af01984a2ae9431913d8b39b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-23 23:57:44.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/88/726088/9/check/DellEMC_Unity_Manila_dhss_true/6d12e29/DellEMC_Unity_Manila_dhss_true/91 : SUCCESS in 1h 41m 59s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/88/726088/9/check/DellEMC_Unity_Manila_dhss_false/6933a78/DellEMC_Unity_Manila_dhss_false/90 : SUCCESS in 51m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"c8e43e18346e7d033e587cf3b590e52c5f61cd56","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 06:22:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5e468e4471b77299574d30a707d4c2e6ab938dd6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 06:24:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f3f586e41370a5c08a34ce841e88cc6746ff39c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:41:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fb4f8fde0743ba88d708680adab949 : SUCCESS in 6m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3846d897b96c4bdbb2fb7e48b06b841d : SUCCESS in 1h 30m 50s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a5ec84192e76439b9401e8615f76b083 : SUCCESS in 52m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4dcc1ea59cc4472aa5a52e9e2db2861f : FAILURE in 1h 56m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bc0c2bacef9646dbbd95571d14f5c344 : FAILURE in 1h 29m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5741f848cce747cea227ab0272839546 : FAILURE in 1h 00m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2ba4018e368a44a29bcb30058969a3aa : SUCCESS in 51m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d6df33dfdd934baa97a94db261c7e7ce : FAILURE in 2h 37m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"01ea4b12ac94a0231fccba297a6545af1f2ee4c1","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-08 12:41:03.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nLooks great, but we need to indicate that the cephfs back end is through NFS rather than native CephFS.","accounts_in_message":[],"_revision_number":11},{"id":"a396f0e026f38561981c2b40ed1056f1a114e2d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-08 13:46:17.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/11/upstream-check/manila-cDOT-manila-tempest-plugin-ss/53be76c/ : SUCCESS in 2h 04m 01s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/11/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/7eca1a2/ : SUCCESS in 1h 25m 50s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"c10faadd205e057f02c0c38bef94709aa11f0f31","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 13:53:55.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"44c96859fa35c0ccdecc078ed997650ed0c66aa5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 13:53:58.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5493dc545e5fff7740b132aec07d9ee6a8c88b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 14:02:07.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 291, column 9\ncould not find expected \u0027:\u0027\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 292, column 9","accounts_in_message":[],"_revision_number":12},{"id":"32db4469cea7c0626c0555134918307f46f3e4ff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 14:07:33.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"b95450ac9c9ae0fc5450e3777b0a19a461a17a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 16:07:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f4caa2e991440bb118bd853051cc0c : SUCCESS in 5m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9d3c67bfe78e4f19a80785c1cfdc769c : SUCCESS in 1h 29m 27s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/60218e6e9b934badaa815117b7923692 : SUCCESS in 50m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ed124409723f4d9aa06b4512bbf7739a : FAILURE in 1h 48m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0253695ac4064b7389d7834ef0bc27c7 : FAILURE in 58m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/89f4e58292cb4d40b4c9242c04bbf2a3 : FAILURE in 1h 25m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d48a1805d87f4ba78070511e16f9e769 : SUCCESS in 58m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e794af0c199d4db3af6d72a82a1f36e5 : FAILURE in 1h 44m 15s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9a0ac3d98ee097c1b0cf1699baaabdcf800b9359","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 17:05:59.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"a4dc574da503c0ec791464830a4c58963a9157e3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 17:08:12.000000000","message":"Patch Set 14:\n\nNative CephFS job will fail, it is missing some options to declare and test snapshot support: https://review.opendev.org/#/c/736806/ addresses this - if I make another patch, I\u0027ll include it as a dependency here, but otherwise you know what\u0027s happening :)","accounts_in_message":[],"_revision_number":14},{"id":"e2abd2310191b0c3b3a8026c45acc5bb026bcdaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 19:22:56.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b3ebfe55d74be7a1bdce61c2e0eed6 : SUCCESS in 6m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aefcaf442b714032ba409ba7a6dc7d1e : FAILURE in 1h 16m 39s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5bd0e9b40bc14c3c887f9b4bd417a9fc : SUCCESS in 46m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/064e06d3fc414bcf8fe627c25b410fd9 : SUCCESS in 1h 09m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8735bee57c2247f9923cf24751d09835 : FAILURE in 55m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/da960b92ab35421584281e0f724359fe : FAILURE in 1h 26m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/89a4c467c4a94dd9b3d5926ed32badd1 : SUCCESS in 51m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/693c109ad8bd4bcd9ff5a151c5a40416 : FAILURE in 2h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1b950af2a22c6c269ffc01cf83b6c05d8ead1db9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 21:47:44.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"253e8893bbb38f1daa1ee0fce624c0714ab84551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 00:05:49.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389c5166e98e475cb96921bd20d34653 : SUCCESS in 5m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb6fa4bbfa8244a693d6921e10f55be6 : FAILURE in 1h 12m 59s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/12b65b7bc2974be3ac96cc3a6fac935a : SUCCESS in 47m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e535282c083f4e37894fb4e75fe8ba54 : FAILURE in 1h 46m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/57508184064146b7b89cfe7059e0f72b : SUCCESS in 1h 16m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/11f5390a85804c6e88cf10d0b3547ab5 : SUCCESS in 1h 41m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8643b0cc019d4c5c9151fb9adba787d4 : SUCCESS in 56m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/532e9db551ed4a3499de9b6ba1cd8db7 : FAILURE in 2h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6361e205ae65a5a8b5ba903dccaf54c250806b4b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-10 21:26:55.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"f3bf26157fae65c29d0efbe91e2db277ae32883f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:09:19.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdc8908659734902a1e832bdd4390f4d : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/759c1fbfbc994832aa96f1096e0917db : SUCCESS in 27m 48s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2e76379302794bc9bf7d39f71380489c : SUCCESS in 34m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bb6164b1c08941d5a7a0693b06752b78 : SUCCESS in 54m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/65c5c97f03c24197b43c52537446668b : SUCCESS in 53m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ace87ef44b3341199f8d5541677a5f5c : FAILURE in 43m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e61dcf04f6084c3c8cbca64313f05b5e : SUCCESS in 32m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63a87be1314745a19fadb6a5c0c71b6e : FAILURE in 1h 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4e7d74f62b7887f786f0434b84f9dccb64621f22","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-11 02:45:50.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/15/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c3324af/ : POST_FAILURE in 4m 24s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/15/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0d749dc/ : SUCCESS in 1h 39m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"a50702ac04426d615f617e2b7ff1b20cf8ece0ac","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-13 05:16:03.000000000","message":"Patch Set 15:\n\nrecheck\n\nsporadic I/O error in the Ceph-nfs job","accounts_in_message":[],"_revision_number":15},{"id":"459f9a3ccbe8325e6d82aa734536d1ebd216b911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 08:25:39.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14eae9e9f9954bb097354721f07419bc : SUCCESS in 6m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b8ad5f00c78145919c0c7c568d24b3bd : SUCCESS in 1h 46m 04s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ebf466bd33204512b19e873363bb4063 : SUCCESS in 1h 52m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/203e4a2272394928a61e7bbc98977fd7 : SUCCESS in 1h 07m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c64cd23e98aa434181c4f15020a6e139 : POST_FAILURE in 1h 41m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8b3299ad111e4d989b3ab78b7ce1a407 : SUCCESS in 2h 23m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/36f0b0306e5e4744909a476a8cb8afb6 : SUCCESS in 1h 57m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/170fd7f0e059462b981bc22341e0c7a4 : FAILURE in 3h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"83fdc49abff1ef396962915a2039755810ca465b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-13 08:25:53.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/726088/15/check/DellEMC_VNX_Manila/d412d02/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"762cadece362efcb58c6c623c4a189089cd5541a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 10:31:15.000000000","message":"Patch Set 15: Code-Review+2\n\nThis is great progress!","accounts_in_message":[],"_revision_number":15},{"id":"15d1396fa07d2584ecfa5a59f52ee59484a33655","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 10:31:41.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"11f1861cecb5ee5a2d4330bc3e43eae7226005d0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-13 10:34:15.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"4b53693e1064bb279acd428ccc84202c4dc83adf","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 10:34:39.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"dcb03e927c2dacabed76c5e766e92446c16c1906","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 10:34:43.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"484faeeaa5effb2e6eebf3026324b1a37fc07bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:09:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f42d0958cc42eca72084d53ba0674d : SUCCESS in 5m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c18332af077947a2b7d04cc3277fcc62 : SUCCESS in 47m 26s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/71d9edf9a5324357bca72ec09f54b5b1 : SUCCESS in 59m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cc69d0d942d9468d81aef7fedd5a712f : SUCCESS in 1h 10m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b3aa07a8341f445ea29cc9eeddc3c36f : SUCCESS in 1h 15m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e73862d968e44a16882e8573064f0663 : SUCCESS in 1h 12m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8ee3637410ee45d1b008c7ee4a150b04 : SUCCESS in 54m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fdc16f73418f416cb741a82f87d5ff33 : FAILURE in 1h 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"aa9609f96c33251df995dc09e1852026efad636c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-13 12:10:01.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/726088/16/check/DellEMC_VNX_Manila/45b18e0/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"a60ec067c54c0b058b4259504e7d3dc828840768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 13:14:56.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"aabf7616489e2b47f90cdf14c01aba7380eaab2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:09:59.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ba94d35e154e73b60f642df893b0dc : SUCCESS in 4m 07s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/27170747f05c44e6accd9d082646d3f9 : SUCCESS in 44m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e2a819468f254c0a95ab45b79eea7894 : FAILURE in 38m 32s","accounts_in_message":[],"_revision_number":16},{"id":"8a7f048afe959f888661531f0e659d7b26355dfe","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 14:24:37.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"58d37fbcff6ef76b9884eff6271a6bbf34181eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 17:05:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a0ed42779e42f08f8387918365c96c : SUCCESS in 5m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eaf62c4d913f4905abb7bb2031c6d392 : SUCCESS in 42m 42s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/dafb5f065b9a4e9eb90f832f7372d45b : SUCCESS in 50m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/48c93a2be7f44d0ba17db4187144f1d6 : FAILURE in 1h 09m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bcac2d46297b4b5f9b88cec771d288f1 : SUCCESS in 1h 19m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f5846e5b416842d78af75ae38c1d6c98 : SUCCESS in 1h 10m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c0b5f101b3704e5386c7470fe05cd648 : SUCCESS in 1h 03m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cd391dd204fe44e2923f1840f2e406cd : FAILURE in 2h 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c91dc98486f77753cc806bebb65b3310a1a808ac","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-13 17:05:48.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/88/726088/16/check/DellEMC_VNX_Manila/5d39f6e/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":16},{"id":"6325acfdadbb686b7fbc0a5e37b44a25545ff35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 17:05:51.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"d1e1569ecec1aebbe3f3192b568d146aeeb93717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 18:02:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"f60c88b5d106e720b1651d1b028384928a99c04b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 18:02:06.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dead4f714b284d47aade20cf0abcf537 : SUCCESS in 5m 24s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/abe35bbee6384ead8ead6c0410bdd1b7 : SUCCESS in 49m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/868fb91cd5b445bebc85ba1ba1e3ac7c : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":16},{"id":"901fb57061d8a62abb02237707a1c373040a300a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-13 18:37:22.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/88/726088/16/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f3ba3cb/ : FAILURE in 1m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/88/726088/16/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2b31141/ : SUCCESS in 1h 31m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"b28c6cd31cbd12a677c94047b74684d032e662dc","revisions":{"7b5f6ff2888a964e2930fb572c145e63cf172fbe":{"kind":"REWORK","_number":1,"created":"2020-05-07 10:38:02.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/1"}}},"commit":{"parents":[{"commit":"d86d85f852599e5f0ba7d99d44953ec1d124984f","subject":"Merge \"Add native zuulv3 job for zfsonlinux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d86d85f852599e5f0ba7d99d44953ec1d124984f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"subject":"[WIP] Add native zuulv3 job for cephfs nfs","message":"[WIP] Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b5f6ff2888a964e2930fb572c145e63cf172fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b5f6ff2888a964e2930fb572c145e63cf172fbe"}]},"branch":"refs/heads/master"},"692100137bed88feded99cdf9d7f79a94a1e2c9c":{"kind":"REWORK","_number":2,"created":"2020-05-08 15:08:56.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/2"}}},"commit":{"parents":[{"commit":"38d392969e1325e633d752f1a96571a9793fbf99","subject":"Merge \"Add native zuulv3 job for lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38d392969e1325e633d752f1a96571a9793fbf99"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-08 15:08:30.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/692100137bed88feded99cdf9d7f79a94a1e2c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/692100137bed88feded99cdf9d7f79a94a1e2c9c"}]},"branch":"refs/heads/master"},"bca6a863539a2539bf644056e7f66a9a5c3a62f7":{"kind":"REWORK","_number":3,"created":"2020-05-08 15:11:21.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/3"}}},"commit":{"parents":[{"commit":"38d392969e1325e633d752f1a96571a9793fbf99","subject":"Merge \"Add native zuulv3 job for lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38d392969e1325e633d752f1a96571a9793fbf99"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-08 15:11:07.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bca6a863539a2539bf644056e7f66a9a5c3a62f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bca6a863539a2539bf644056e7f66a9a5c3a62f7"}]},"branch":"refs/heads/master"},"df7eaad5f744df11078d55616f06544ad9e80ca3":{"kind":"REWORK","_number":4,"created":"2020-05-11 17:04:22.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/4"}}},"commit":{"parents":[{"commit":"7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de","subject":"Add native zuulv3 job for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-11 17:03:58.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df7eaad5f744df11078d55616f06544ad9e80ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df7eaad5f744df11078d55616f06544ad9e80ca3"}]},"branch":"refs/heads/master"},"9414ba3b954042e03e3a7a3a3424e45cb1474740":{"kind":"REWORK","_number":5,"created":"2020-05-13 22:05:05.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/5"}}},"commit":{"parents":[{"commit":"17fcfb8965df323ce7876827d1633bd321e21291","subject":"Merge \"Fix dict.values usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17fcfb8965df323ce7876827d1633bd321e21291"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-13 22:04:55.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9414ba3b954042e03e3a7a3a3424e45cb1474740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9414ba3b954042e03e3a7a3a3424e45cb1474740"}]},"branch":"refs/heads/master"},"5fac24e27cedb86c1ae1a4426e82ff96ae07e389":{"kind":"REWORK","_number":6,"created":"2020-05-13 22:35:15.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/6"}}},"commit":{"parents":[{"commit":"17fcfb8965df323ce7876827d1633bd321e21291","subject":"Merge \"Fix dict.values usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17fcfb8965df323ce7876827d1633bd321e21291"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-13 22:35:00.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5fac24e27cedb86c1ae1a4426e82ff96ae07e389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5fac24e27cedb86c1ae1a4426e82ff96ae07e389"}]},"branch":"refs/heads/master"},"7b202a928370b57d8c5c2e2577b18f30c26fc97b":{"kind":"REWORK","_number":7,"created":"2020-05-13 22:45:36.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/7"}}},"commit":{"parents":[{"commit":"17fcfb8965df323ce7876827d1633bd321e21291","subject":"Merge \"Fix dict.values usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17fcfb8965df323ce7876827d1633bd321e21291"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-13 22:45:26.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b202a928370b57d8c5c2e2577b18f30c26fc97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b202a928370b57d8c5c2e2577b18f30c26fc97b"}]},"branch":"refs/heads/master"},"cd9882f0d07e2e3e51aadfee695870d9700fb09b":{"kind":"REWORK","_number":8,"created":"2020-05-14 14:41:51.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/8"}}},"commit":{"parents":[{"commit":"17fcfb8965df323ce7876827d1633bd321e21291","subject":"Merge \"Fix dict.values usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17fcfb8965df323ce7876827d1633bd321e21291"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-14 14:41:41.000000000","tz":-180},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd9882f0d07e2e3e51aadfee695870d9700fb09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd9882f0d07e2e3e51aadfee695870d9700fb09b"}]},"branch":"refs/heads/master"},"22e7780422fd7587a126e8aa7b4933bcf9c49bd3":{"kind":"REWORK","_number":9,"created":"2020-06-23 19:00:53.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/88/726088/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/9"}}},"commit":{"parents":[{"commit":"f8894b6f0b3d98394d6c573a065ac10baa7e621c","subject":"Skip snapshot cloning if there is no capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f8894b6f0b3d98394d6c573a065ac10baa7e621c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-06-23 19:00:48.000000000","tz":0},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22e7780422fd7587a126e8aa7b4933bcf9c49bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22e7780422fd7587a126e8aa7b4933bcf9c49bd3"}]},"branch":"refs/heads/master"},"ca030b93fb6c6ab9d4fa8e9e84f5aca0c00ad22d":{"kind":"REWORK","_number":10,"created":"2020-07-08 06:22:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/726088/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/10"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-08 06:22:26.000000000","tz":-420},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I1ef42b414f8af0b8f9a912655757ec200a75c309\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca030b93fb6c6ab9d4fa8e9e84f5aca0c00ad22d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ca030b93fb6c6ab9d4fa8e9e84f5aca0c00ad22d"}]},"branch":"refs/heads/master"},"c36391bb06c0101bd0de5ed8fdd2b67d14326062":{"kind":"REWORK","_number":11,"created":"2020-07-08 06:24:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/726088/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/11"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-08 06:24:17.000000000","tz":-420},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I1ef42b414f8af0b8f9a912655757ec200a75c309\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c36391bb06c0101bd0de5ed8fdd2b67d14326062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c36391bb06c0101bd0de5ed8fdd2b67d14326062"}]},"branch":"refs/heads/master"},"a434b35e8b6ff53ec5f378ac01777d986803f953":{"kind":"REWORK","_number":12,"created":"2020-07-08 13:53:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/726088/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/12"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-08 13:53:53.000000000","tz":0},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I1ef42b414f8af0b8f9a912655757ec200a75c309\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a434b35e8b6ff53ec5f378ac01777d986803f953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a434b35e8b6ff53ec5f378ac01777d986803f953"}]},"branch":"refs/heads/master"},"7d3879309dc3c41ffef343d20d720878396a275a":{"kind":"REWORK","_number":13,"created":"2020-07-08 14:07:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/726088/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/13"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-08 14:07:31.000000000","tz":0},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I1ef42b414f8af0b8f9a912655757ec200a75c309\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7d3879309dc3c41ffef343d20d720878396a275a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7d3879309dc3c41ffef343d20d720878396a275a"}]},"branch":"refs/heads/master"},"01a4d5a48dabc5524f318e5c50bc300648dbefc6":{"kind":"REWORK","_number":14,"created":"2020-07-08 17:05:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/726088/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/14"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-08 17:05:55.000000000","tz":0},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I1ef42b414f8af0b8f9a912655757ec200a75c309\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/01a4d5a48dabc5524f318e5c50bc300648dbefc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/01a4d5a48dabc5524f318e5c50bc300648dbefc6"}]},"branch":"refs/heads/master"},"7369eca3767a7b7cc0971377eb4db7ae253a6dff":{"kind":"REWORK","_number":15,"created":"2020-07-08 21:47:44.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/726088/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/15"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-08 21:47:36.000000000","tz":-420},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7369eca3767a7b7cc0971377eb4db7ae253a6dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7369eca3767a7b7cc0971377eb4db7ae253a6dff"}]},"branch":"refs/heads/master"},"b28c6cd31cbd12a677c94047b74684d032e662dc":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-07-13 10:34:39.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/88/726088/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/88/726088/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/88/726088/16"}}},"commit":{"parents":[{"commit":"c80e538a47c0be47eea193deaa8ce7eb971792ae","subject":"Merge \"Test different users in same tenant can use same cephx id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c80e538a47c0be47eea193deaa8ce7eb971792ae"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-07 10:36:28.000000000","tz":-180},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-13 10:34:39.000000000","tz":0},"subject":"Add native zuulv3 job for cephfs nfs","message":"Add native zuulv3 job for cephfs nfs\n\nThe old/legacy job depends on components\nfrom OpenStack infra that are not maintained\nany longer.\n\nThis job also enables testing CephFS snapshots.\n\nDepends-On: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nChange-Id: I700731e714db09f757d21d31ba3ddce62d70e48d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b28c6cd31cbd12a677c94047b74684d032e662dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b28c6cd31cbd12a677c94047b74684d032e662dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
