)]}'
{"id":"openstack%2Fmanila-tempest-plugin~727072","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8dede49a32f432d05637e63177f9f5feae245adc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cephx_id_negative","hashtags":[],"change_id":"I8dede49a32f432d05637e63177f9f5feae245adc","subject":"[Negative] Test different tenants cannot use same cephx id","status":"MERGED","created":"2020-05-12 07:08:41.000000000","updated":"2020-07-11 00:15:34.000000000","submitted":"2020-07-11 00:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727072-1594426534326-72fcc681","meta_rev_id":"2403165a6bb9cdc83a4fd645e03c38a8d7393337","_number":727072,"virtual_id_number":727072,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-05-13 11:26:53.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-11 00:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"date":"2020-07-08 11:05:17.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-07-08 06:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-08 11:31:40.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-08 11:31:40.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 07:09:50.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-05-12 07:10:18.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-05-13 11:26:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 06:01:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-07-08 11:31:40.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-07-11 00:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdfb0cfa1b98d20826190fc94bac7adb10b24b48","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-12 07:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15439aa4180e38103063d660b7b9256e66c5e34c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:37:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd3f8bfc6a441cf9cd481ef104806bd : FAILURE in 4m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4bc9e9605a72430c95dfef247f93e67e : SUCCESS in 1h 00m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e2339acbc71247969f55689fad0cbbe0 : SUCCESS in 41m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/db3457cf0c7b49b8bec4aacb05940b31 : SUCCESS in 1h 04m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d955f24b43bf4691bb8e52941fb5820b : SUCCESS in 58m 01s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5807ae8879384460a745b0f4fd30c7eb : SUCCESS in 1h 38m 28s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/a0242869d2e0432b8cf51be47799fbd5 : SUCCESS in 50m 26s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/a3009c787c8d480cbbab48678550b351 : FAILURE in 54m 10s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9babc5b9bd194ba28df8e4f2c374a882 : FAILURE in 1h 54m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/85e11d710a4843dda0fcf1365d76cd30 : SUCCESS in 25m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ecc7bc1a60ac4ed0afb65c7f22400bd7 : SUCCESS in 1h 18m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_quotas.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_actions.py","accounts_in_message":[],"_revision_number":1},{"id":"677bb55d37f530cb9af9861d842c09dee7b5e108","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-12 17:26:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a33c22271d2361fc68607148858c36d2f07a153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 20:37:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5a2c77f8f74fde9b56110a721ffae9 : SUCCESS in 4m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5ae029afd0dc4dcbba962cbab7af80f0 : SUCCESS in 1h 07m 26s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9b65ce07beb54ea5868475d9119e734e : SUCCESS in 49m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c94672c5e5d40509a2b54bd7f4d5f8c : FAILURE in 1h 21m 49s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5cb4fdb8f5624161b3e8a3bf30c23842 : FAILURE in 54m 51s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/7382a0d9b1fe47d0b757d83332e68603 : SUCCESS in 1h 43m 53s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/2b18c5efdaf3402b833bf10f31adaa4e : SUCCESS in 49m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5f222e9bfa084ef09cfb85c0a29c459e : FAILURE in 44m 29s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/037465e8cb8f4f0f9068958a392d0f05 : FAILURE in 1h 51m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/49845539ad3e42eba8b7f6ed4237b2a8 : SUCCESS in 28m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cc9e3c82efa84b92ac9987242a2c1878 : SUCCESS in 1h 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99a0da22b2724a786df906ed044cdbd5b21d3471","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-13 05:28:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eaafeabb8355a45d2f95af38c1911fcbfced2c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 07:11:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c844aa5821141c19b70f4e6e898905e : SUCCESS in 5m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/85ddd4f75479438abaae7aacf140b6ba : SUCCESS in 1h 36m 21s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2f2f8d4b613e4b719f5bb958651ee4e1 : SUCCESS in 47m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4f497f1647cc4df2b362b731903c952f : FAILURE in 1h 00m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ce54dfb4ae214513b624d2223d686d80 : FAILURE in 53m 40s (non-voting)\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/17450e0a9c17446d9b87afde5bff3250 : FAILURE in 1h 14m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f8449f7301c14fc9976a4e5f22b045cd : SUCCESS in 54m 49s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/89825e3fb03347a18cb80ce827192002 : FAILURE in 46m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6459d0b495184fddb048efc62c991a72 : SUCCESS in 1h 38m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/76a7375e10e04d829814beed47095a72 : SUCCESS in 24m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c2976b95a6af4fb198bb94449d5a5ef0 : SUCCESS in 1h 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46756ac6ffdba31b2075507c1667ff01d48b24f6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-13 11:26:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/72/727072/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/56f8a68/ : SUCCESS in 2h 14m 10s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/72/727072/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/67aa3a3/ : SUCCESS in 1h 31m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2d1bfb3e249fe8c41f81553505ef40fdb7226466","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 06:01:40.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck\n\nLGTM, cephfs-native job logs are gone, lets run it again","accounts_in_message":[],"_revision_number":2},{"id":"971c2fcab22295960cd6d812eada56c42fcbc954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 10:55:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3046a98bdf0245378002fc46301d9fd3 : SUCCESS in 8m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5a3d3c7c11ce4f44ab342f9eda860228 : RETRY_LIMIT in 1h 01m 24s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e561c5a7096749d58868c96aa6bc78e0 : SUCCESS in 44m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0c8bd9dc2011449fbc67f89bbc5c3bf0 : FAILURE in 1h 49m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5c4779dc6918404a8004b080b2ef2616 : SUCCESS in 56m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2bab2cf10bcb4bb18712c799b9a56197 : SUCCESS in 55m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3863f95552d244fca0a3a96c59d3d9b7 : FAILURE in 2h 03m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5a76a08dc51143bbbe8c88ae06683d7d : SUCCESS in 1h 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"223e967bf6e06ef65ec8acd78424db1c345461ce","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-07-08 11:05:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31f4b21eb54397f3e9b15b21d122d9ae31c2f8a0","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-08 11:31:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nOK, we have cephfs-native logs, where the new test ran and passed.  It looks good and the lvm failure is unrelated.","accounts_in_message":[],"_revision_number":2},{"id":"5e7d229c454f634e6807635b7f2308e8c32f1c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 14:14:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43daf8bb805e4eaf9cb65820e3f815be : SUCCESS in 4m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46a9697200314cb69e6d1b743a131ada : FAILURE in 1h 20m 58s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e4ee5edf83c647b4a18dd62ec0a361b0 : SUCCESS in 46m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/78ae8e90136043adac313c4d003a9dd0 : FAILURE in 1h 48m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5d24eca334f74d47b9bb98882e423ea3 : FAILURE in 1h 11m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9066c18f21ef4946a0365af53be2c571 : SUCCESS in 44m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ab94504b3e824b8fb8dc3288580bfa55 : FAILURE in 1h 55m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4be0cd57a99a499a82b11b075e0d8c1c : FAILURE in 1h 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ec0dd2862f793ad5f535ea8ffbed43321d07eb5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-08 14:33:52.000000000","message":"Patch Set 2:\n\nrecheck\n\nWe\u0027re down to intermittent test failures in lvm and cephfs-nfs jobs now.  They are unrelated to Lirpn\u0027s patch.","accounts_in_message":[],"_revision_number":2},{"id":"4b32c1f992967bff8e86bd31d55ab1bc7275463c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 17:51:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341830c7a79a49af8e59a42cc3a9d624 : SUCCESS in 4m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/406e51cf1d624f3f9ab5c191ed6b0571 : FAILURE in 1h 17m 55s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/37abc473fe104891a3b4c1700c59f904 : SUCCESS in 57m 11s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4934acd0b3504d40bf03a5d78e1d2b77 : FAILURE in 1h 53m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/336f3646b7e54d069c5a95a7420d332b : SUCCESS in 57m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/252bf415e73b42f29812e432985c7c1a : SUCCESS in 49m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0a39050b8bf349ea8c640e3cebe89b02 : FAILURE in 1h 46m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c3224be5c5f443efb928d1b86ed92f70 : SUCCESS in 1h 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d3d104d68c57024c59c6bc29937d58c5e772692","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-10 21:27:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5a155966dbdd7aa4c0e80663abaad0cbf314155c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:29:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7d30f8d17547af82d5610293a51626 : SUCCESS in 3m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d22a653b63f44bedb4451acd1eb3f0a9 : SUCCESS in 30m 42s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a032cd58f38c4457a8f519f94d345f8d : SUCCESS in 28m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/89d6c7cbc6ae4a64944e5b6187c69616 : SUCCESS in 1h 04m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b1304fcbcdd34428a1356777c27e8277 : SUCCESS in 53m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d6a447dd320d46c5a947607c40653967 : SUCCESS in 32m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f96d42e472ba4152ac003dcc505deb8a : FAILURE in 1h 17m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7ef9974977644629a48af36890e46193 : SUCCESS in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"871b4aa38e755c814ea8c17d99f1a911311a1532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:29:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b569ece7ba2fa7a96dfb5c5cdec6654f747a17d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 00:15:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"997cc681f1aadbceb3dde79dabf5e25347ef344c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 00:15:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c458ac67846b4ca9a88b92cec63b253e : SUCCESS in 4m 47s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8e28c63032da4530ab51ea0438fcf7fd : SUCCESS in 44m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1cba8cbcd18c40fb8f37327102dd73ad : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e53a7f6db7f449ee0bb310c2cca46a1e609d4ff","revisions":{"72dc8c310dedf39006ad3609f6c1fe39600d36a7":{"kind":"REWORK","_number":1,"created":"2020-05-12 07:08:41.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/72/727072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/727072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/1"}}},"commit":{"parents":[{"commit":"7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de","subject":"Add native zuulv3 job for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-12 06:35:38.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-12 06:57:27.000000000","tz":180},"subject":"[Negative] Test different tenants cannot use same cephx id","message":"[Negative] Test different tenants cannot use same cephx id\n\nThis test uses two tenants (alt and primary) and uses one\ncephx id. The test validates that a same cephx id can not\nbe used in different tenants.\n\nChange-Id: I8dede49a32f432d05637e63177f9f5feae245adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72dc8c310dedf39006ad3609f6c1fe39600d36a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/72dc8c310dedf39006ad3609f6c1fe39600d36a7"}]},"branch":"refs/heads/master"},"0e53a7f6db7f449ee0bb310c2cca46a1e609d4ff":{"kind":"REWORK","_number":2,"created":"2020-05-13 05:28:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/72/727072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/72/727072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/72/727072/2"}}},"commit":{"parents":[{"commit":"7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de","subject":"Add native zuulv3 job for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a8f5efb88aa1111dc1d176e12e3e8b989f0d6de"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-12 06:35:38.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-05-13 05:28:30.000000000","tz":180},"subject":"[Negative] Test different tenants cannot use same cephx id","message":"[Negative] Test different tenants cannot use same cephx id\n\nThis test uses two tenants (alt and primary) and uses one\ncephx id. The test validates that a same cephx id can not\nbe used in different tenants.\n\nChange-Id: I8dede49a32f432d05637e63177f9f5feae245adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e53a7f6db7f449ee0bb310c2cca46a1e609d4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e53a7f6db7f449ee0bb310c2cca46a1e609d4ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
