)]}'
{"id":"openstack%2Fmanila-tempest-plugin~729486","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic240622a5791cca0bfd8dd7d452a2ff70565055d","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic240622a5791cca0bfd8dd7d452a2ff70565055d","subject":"Fix pygments style","status":"NEW","created":"2020-05-20 02:22:33.000000000","updated":"2020-05-20 07:37:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ecc2b434f32a54032599e65d2cc03febef6a6df","_number":729486,"virtual_id_number":729486,"owner":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-20 07:37:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 07:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86179faf156b08ec6f328ae85c317510346f1485","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 02:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"248d323fe625295895253362396c4edf1e2c97d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 02:27:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b870f527b64583af62e87194ab1a2f : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"599d5bb00bb7cdf1f6224d33fd1005d9a5136f8e","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 07:31:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9350f1f51414983952a439d70f0543bdaad6593c","author":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"date":"2020-05-20 07:31:47.000000000","message":"Topic d0097249e77bfca285fa1603a71721b40a503926 removed","accounts_in_message":[],"_revision_number":2},{"id":"9ecc2b434f32a54032599e65d2cc03febef6a6df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 07:37:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d6e2de74b04e26b4391e9e9df2e158 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"acb53afb750c0350bfc8a53b80339b768b2b529c","revisions":{"d0097249e77bfca285fa1603a71721b40a503926":{"kind":"REWORK","_number":1,"created":"2020-05-20 02:22:33.000000000","uploader":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"ref":"refs/changes/86/729486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/729486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/1"}}},"commit":{"parents":[{"commit":"0e780261632e60b9acd2929583599f81cbb18b26","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e780261632e60b9acd2929583599f81cbb18b26"}]}],"author":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 02:21:07.000000000","tz":480},"committer":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 02:21:07.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nChange-Id: Ic240622a5791cca0bfd8dd7d452a2ff70565055d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0097249e77bfca285fa1603a71721b40a503926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d0097249e77bfca285fa1603a71721b40a503926"}]},"branch":"refs/heads/master"},"acb53afb750c0350bfc8a53b80339b768b2b529c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-20 07:31:15.000000000","uploader":{"_account_id":31446,"name":"fuzihao","email":"fuzihao@inspur.com","username":"fuzihao"},"ref":"refs/changes/86/729486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/729486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/729486/2"}}},"commit":{"parents":[{"commit":"0e780261632e60b9acd2929583599f81cbb18b26","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e780261632e60b9acd2929583599f81cbb18b26"}]}],"author":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 02:21:07.000000000","tz":480},"committer":{"name":"fuzihao","email":"fuzihao@inspur.com","date":"2020-05-20 07:30:37.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.\n\nChange-Id: Ic240622a5791cca0bfd8dd7d452a2ff70565055d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/acb53afb750c0350bfc8a53b80339b768b2b529c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/acb53afb750c0350bfc8a53b80339b768b2b529c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
