)]}'
{"id":"openstack%2Fmanila-tempest-plugin~730996","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I38136b8721b953fb8121a64d4c51849568091f0c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cephfs-native-scenario","hashtags":[],"change_id":"I38136b8721b953fb8121a64d4c51849568091f0c","subject":"[DNM] Run CephFS Native scenario tests","status":"ABANDONED","created":"2020-05-26 21:38:28.000000000","updated":"2020-06-16 02:19:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a37cde1ca8c4b6141fed3c1235201a812242e74","_number":730996,"virtual_id_number":730996,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-27 19:47:18.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2020-05-28 15:50:28.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-31 06:17:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"date":"2020-05-31 06:17:43.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 19:47:18.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-05-28 15:50:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-31 06:17:43.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2020-05-31 06:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5aa92720797cb8ec430b3a033550f5cca327ecf","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-26 21:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635e335fc360c7a6b2e5207303ac931d168547a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 21:38:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  The key \"MANILA_SERVICE_IMAGE_ENABLED\" appears more than once;\n  duplicate keys are not permitted.\n\nThe error appears in the following stanza:\n\n  SHARE_DRIVER: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_ENABLED_BACKENDS: cephfsnative\n          ENABLED_SHARE_PROTOCOLS: CEPHFS\n          MANILA_DEFAULT_SHARE_TYPE_EXTRA_SPECS: \u0027snapshot_support\u003dFalse\u0027\n          MANILA_SERVICE_IMAGE_ENABLED: false\n          MANILA_OPTGROUP_cephfsnative_share_driver: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_OPTGROUP_cephfsnative_driver_handles_share_servers: false\n          MANILA_OPTGROUP_cephfsnative_cephfs_conf_path: /etc/ceph/ceph.conf\n          MANILA_OPTGROUP_cephfsnative_cephfs_auth_id: manila\n          # NOTE(vkmc) Defaults to true, we be safe to remove\n          MANILA_SERVICE_IMAGE_ENABLED: true\n          MANILA_SERVICE_IMAGE_URL: https://tarballs.opendev.org/openstack/manila-image-elements/images/manila-service-image-cephfs-master.qcow2\n          MANILA_SERVICE_IMAGE_NAME: manila-service-image-cephfs-master\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 202, column 9","accounts_in_message":[],"_revision_number":1},{"id":"7a5ab2086533f8abbf468508c5ff316099e15686","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-26 21:40:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7373325c3e81adffc762c1c065d682138fedfe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 21:41:03.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  The key \"MANILA_SERVICE_IMAGE_ENABLED\" appears more than once;\n  duplicate keys are not permitted.\n\nThe error appears in the following stanza:\n\n  SHARE_DRIVER: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_ENABLED_BACKENDS: cephfsnative\n          ENABLED_SHARE_PROTOCOLS: CEPHFS\n          MANILA_DEFAULT_SHARE_TYPE_EXTRA_SPECS: \u0027snapshot_support\u003dFalse\u0027\n          MANILA_SERVICE_IMAGE_ENABLED: false\n          MANILA_OPTGROUP_cephfsnative_share_driver: manila.share.drivers.cephfs.driver.CephFSDriver\n          MANILA_OPTGROUP_cephfsnative_driver_handles_share_servers: false\n          MANILA_OPTGROUP_cephfsnative_cephfs_conf_path: /etc/ceph/ceph.conf\n          MANILA_OPTGROUP_cephfsnative_cephfs_auth_id: manila\n          # NOTE(vkmc) Defaults to true, we be safe to remove\n          MANILA_SERVICE_IMAGE_ENABLED: true\n          MANILA_SERVICE_IMAGE_URL: https://tarballs.opendev.org/openstack/manila-image-elements/images/manila-service-image-cephfs-master.qcow2\n          MANILA_SERVICE_IMAGE_NAME: manila-service-image-cephfs-master\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 202, column 9","accounts_in_message":[],"_revision_number":2},{"id":"4cb023c5ab28b129eb931c58df05818a0e7f9f9b","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-26 21:59:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea563f3a4567620c4a818a5f621150a1e009ca88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:14:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade1791b0a204f49b2fbf68aa784d134 : SUCCESS in 2m 58s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/15d1bbfefaf1478faa1d83e42c0d1ab4 : FAILURE in 58m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a6f304dff202df71014e7b65f58f84d06386d37","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-27 10:49:13.000000000","message":"Patch Set 3:\n\nstorage network must exist and be configured in tempest.conf for running scenario tests.","accounts_in_message":[],"_revision_number":3},{"id":"157dd0bec64dca82bd5ae6b17d41a133a5439018","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-27 11:02:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"897b51e73f23416c9bf7ed049ae0b45207e73631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:29:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91fe9f406de4aaab679b132ed06ed73 : SUCCESS in 3m 30s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/47dc92e0667c49359a6d6a1fb2b41455 : FAILURE in 1h 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f01ccc4198550125b43c054384bd9b659bb34226","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-27 12:29:18.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5e1ab5462eec87b62cdaa0a300fa6fb2fbcc14af","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-27 14:13:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fa47149daf5c0f162f917c30d9e02732d1472fa9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-27 14:39:43.000000000","message":"Patch Set 3:\n\nPlease ignore my last comment, I confused with our downstream requirement.","accounts_in_message":[],"_revision_number":3},{"id":"6aef8c5c95446ccc2b0d0fd4b4603e304a9672d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:40:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515df0a7736f4dff82c050f86c56a040 : SUCCESS in 4m 33s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3117a849143a453f86089aa9f5109f3e : FAILURE in 1h 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd56e7c25c4211950ba93fa152ba6abd8eacc513","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-27 15:40:22.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c4a380181b58ba7ee29f5f75502436858de8df05","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-05-27 17:36:54.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fc531bb4e793a64967ee6a229fb39810a24b0c6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 19:20:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b920b3e086d7445495694d00867154e9 : SUCCESS in 6m 39s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8095f95ca6d14839a7bcbc85ae0104db : FAILURE in 1h 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6ef2d29007160168c58164ae7ed75c408ba9fbc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-27 19:29:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/96/730996/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a892942/ : FAILURE in 4m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/96/730996/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/27e6ee5/ : FAILURE in 4m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"905514699b25730d0723d090cab31a401b023433","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-27 19:47:18.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"68a0f31f752eb04d2f1d283eba6d1dc7e1a64c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:58:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da03b91c63b142978ba6ade21224ae3c : SUCCESS in 3m 51s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0375b925534442cb8d1143e9f095b8d6 : SUCCESS in 1h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4edd7aec10cd94c47353822aa7c6f6bc171d9264","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-27 23:15:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/96/730996/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b44ae8c/ : SUCCESS in 2h 16m 51s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/96/730996/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/cffd095/ : SUCCESS in 1h 32m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"eca0d1867b06cc7481baf79be79836efe7035843","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-28 11:45:36.000000000","message":"Patch Set 4:\n\nChecking the latest update of https://review.opendev.org/#/c/725487\nIncludes all scenario tests.\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a9b9fecc9036528c958d94e1a8bfe306644f0e6e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-28 11:47:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1afb9ef18735e25c0efd6fcd870dcca6e1029cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:27:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a48ae4d6d1f43868e14ab57e8bb323b : SUCCESS in 3m 59s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ad8b6396492a4591bbee4c3b3aadcc8c : FAILURE in 1h 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3f860e7212f59bf8148a3fff677c9e53188fa834","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-28 15:50:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/96/730996/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/bc67342/ : SUCCESS in 2h 22m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/96/730996/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/726846c/ : SUCCESS in 1h 33m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8dac55349bec05496e794f959203430c43419288","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-05-31 06:17:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c4c3cff35eba64d8c925cc3bab37e4380a476326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-31 06:17:57.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"299b992eb43cc2cec4fe26d0364f49a24f4a95fa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-16 02:19:17.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/732014/ has enabled scenario tests with ceph native..","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c0463b4c610508eda2dc96c47a0e854baa648a37","revisions":{"c1e7239ff66b400486eef63912b1b40b90bf7f45":{"kind":"REWORK","_number":1,"created":"2020-05-26 21:38:28.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/96/730996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/730996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/1"}}},"commit":{"parents":[{"commit":"c35102ba7ff8ae4b334501ab100bd57f59f23c55","subject":"Add native zuulv3 job for cephfs native","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c35102ba7ff8ae4b334501ab100bd57f59f23c55"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:37:24.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:37:24.000000000","tz":-180},"subject":"[DNM] Run CephFS Native scenario tests","message":"[DNM] Run CephFS Native scenario tests\n\nDepends-On: https://review.opendev.org/#/c/725487\nChange-Id: I38136b8721b953fb8121a64d4c51849568091f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1e7239ff66b400486eef63912b1b40b90bf7f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1e7239ff66b400486eef63912b1b40b90bf7f45"}]},"branch":"refs/heads/master"},"19f721441aec296f4bb62393d0d93c54a633729a":{"kind":"REWORK","_number":2,"created":"2020-05-26 21:40:45.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/96/730996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/730996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/2"}}},"commit":{"parents":[{"commit":"c35102ba7ff8ae4b334501ab100bd57f59f23c55","subject":"Add native zuulv3 job for cephfs native","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c35102ba7ff8ae4b334501ab100bd57f59f23c55"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:37:24.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:40:28.000000000","tz":-180},"subject":"[DNM] Run CephFS Native scenario tests","message":"[DNM] Run CephFS Native scenario tests\n\nDepends-On: https://review.opendev.org/#/c/725487\nChange-Id: I38136b8721b953fb8121a64d4c51849568091f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19f721441aec296f4bb62393d0d93c54a633729a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19f721441aec296f4bb62393d0d93c54a633729a"}]},"branch":"refs/heads/master"},"ef49c70a830a2d0187156182391cbe043fdac51d":{"kind":"REWORK","_number":3,"created":"2020-05-26 21:59:42.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/96/730996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/730996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/3"}}},"commit":{"parents":[{"commit":"c35102ba7ff8ae4b334501ab100bd57f59f23c55","subject":"Add native zuulv3 job for cephfs native","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c35102ba7ff8ae4b334501ab100bd57f59f23c55"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:37:24.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:57:08.000000000","tz":-180},"subject":"[DNM] Run CephFS Native scenario tests","message":"[DNM] Run CephFS Native scenario tests\n\nDepends-On: https://review.opendev.org/#/c/725487\nChange-Id: I38136b8721b953fb8121a64d4c51849568091f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef49c70a830a2d0187156182391cbe043fdac51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef49c70a830a2d0187156182391cbe043fdac51d"}]},"branch":"refs/heads/master"},"c0463b4c610508eda2dc96c47a0e854baa648a37":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-27 17:36:54.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/96/730996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/730996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/730996/4"}}},"commit":{"parents":[{"commit":"c35102ba7ff8ae4b334501ab100bd57f59f23c55","subject":"Add native zuulv3 job for cephfs native","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c35102ba7ff8ae4b334501ab100bd57f59f23c55"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-26 21:37:24.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-05-27 17:36:25.000000000","tz":-180},"subject":"[DNM] Run CephFS Native scenario tests","message":"[DNM] Run CephFS Native scenario tests\n\nDepends-On: https://review.opendev.org/#/c/725487\nDepends-On: https://review.opendev.org/#/c/731284\nChange-Id: I38136b8721b953fb8121a64d4c51849568091f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0463b4c610508eda2dc96c47a0e854baa648a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0463b4c610508eda2dc96c47a0e854baa648a37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
