)]}'
{"id":"openstack%2Fmanila-tempest-plugin~731651","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iaef7873e7539045753bb5676fe93f4ded586b5ac","project":"openstack/manila-tempest-plugin","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"Iaef7873e7539045753bb5676fe93f4ded586b5ac","subject":"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs","status":"MERGED","created":"2020-05-29 05:26:07.000000000","updated":"2020-07-24 12:50:25.000000000","submitted":"2020-07-24 12:50:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731651-1595595025983-41dc37f0","meta_rev_id":"a61407db0b6295d80e0a7618909188ad2f5febb6","_number":731651,"virtual_id_number":731651,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-07-21 08:11:45.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-07-21 05:06:52.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-07-21 09:17:12.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 12:50:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-07-21 19:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-21 19:29:12.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-21 16:51:00.000000000","_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},{"value":0,"date":"2020-07-21 16:51:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-21 19:29:12.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2020-07-21 16:50:56.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 02:44:57.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-21 08:11:45.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-21 09:17:12.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-21 16:50:56.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-07-21 16:51:00.000000000","updated_by":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"reviewer":{"_account_id":30722,"name":"WooHyung Han","email":"techhanx@gmail.com","username":"whhan91"},"state":"REVIEWER"},{"updated":"2020-07-21 16:51:06.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-07-21 19:02:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-07-21 19:29:12.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-07-24 12:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfe450489130dc29cbaf310617557ca85c51828f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-29 05:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d67c8203c9221f7d086c0eaee8fe3c16bda18294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 05:26:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 231, column 16","accounts_in_message":[],"_revision_number":1},{"id":"320f244b1e20d828ef61072beb151e3a31dc0e45","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-29 05:27:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f35c663f751100db315698315915f3c2e3d9486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 05:27:48.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 235, column 7\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 236, column 7","accounts_in_message":[],"_revision_number":2},{"id":"d16a5d6ee005fba8a533e459929ddb5f5ec5f32b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-29 05:29:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"855a1d86e712fbea42066182ff9cf30ae4a6ebc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 05:57:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/159c2f10a0404df5afeb79ba33c3e1ef : FAILURE in 26m 28s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/bd8e9484e0e54ff58803891b2405bebb : FAILURE in 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"05aafeb279bb14052716e12235161fb0c6d3b757","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 18:18:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e27b88a1047517e3222346e6aeca7441f50cc3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:54:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fd7e1c6eaae148b891b227007b1481b2 : FAILURE in 29m 58s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3f4ec779b82746a8af21753c1f236c83 : FAILURE in 25m 49s","accounts_in_message":[],"_revision_number":4},{"id":"c8928337a38b03c413b5b81d2c64ba462b4cae03","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 19:47:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3d34d6ebf5365b0abd5fe45e67fea2bd61d201a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 20:25:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a3578a64771846b481e560c448ca041a : FAILURE in 25m 47s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/38a73678fbbb4965bf4d3eb44fd22e25 : FAILURE in 28m 11s","accounts_in_message":[],"_revision_number":5},{"id":"1000805616267851ae9d2af0cceaa3684a32010e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 20:42:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b3689587ca9bd47c9a36c11935feccb9baf20308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 21:18:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a29af7e0f378488cb83e9a6361f1489e : FAILURE in 33m 03s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9402cf520501410fa121476c2c732a3b : FAILURE in 25m 37s","accounts_in_message":[],"_revision_number":6},{"id":"80601784260c6c858aeffca289dea5182d1f2cbd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 21:38:25.000000000","message":"Topic changed from glusterfs to native-zuulv3-migration","accounts_in_message":[],"_revision_number":6},{"id":"6591848dde26a869e0bc582857b4e6f9c0b0297b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 21:38:28.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"407e21fd5f6337b3ee253e761b6ce16dfc11f785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 22:11:49.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/26d30fc37afa4784bb309cfe7a673629 : FAILURE in 22m 32s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a2273b18a1f9476088d23c5885bcdfdc : FAILURE in 26m 52s","accounts_in_message":[],"_revision_number":6},{"id":"faf2743c664031859ce2c6beec0eea0619642a7e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 22:44:07.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"dddda152e9c1b4e730e30e608b3be65142ec281e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 23:18:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b3548f33d83e4ac998a81e53e9377fb1 : FAILURE in 27m 58s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5647070656b84373abdc5b17696fe6a4 : FAILURE in 27m 01s","accounts_in_message":[],"_revision_number":7},{"id":"d36e4c9a1435ce013892af53c58a56742ac22747","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-17 00:20:13.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e3c8ac36397f2804a6a34e79f8ead90ce8d8d5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 01:00:43.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d7bf6744f28b4c40824bc3a8e445cc5d : FAILURE in 29m 22s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e79487bc84e54e3989190bbf76d69b06 : FAILURE in 25m 31s","accounts_in_message":[],"_revision_number":7},{"id":"26f064e9522582cf3517ff53ec174297b068994f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-17 02:44:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/51/731651/7/check/DellEMC_Unity_Manila_dhss_true/048c4c5/DellEMC_Unity_Manila_dhss_true/210 : FAILURE in 18m 25s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/51/731651/7/check/DellEMC_Unity_Manila_dhss_false/ef587e9/DellEMC_Unity_Manila_dhss_false/196 : SUCCESS in 49m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"029f45dd9b6dcd8e6d675c0dfebe195424c854b4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-17 16:30:34.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"08a0bba38714ea15b52476c97098b72c322f9348","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-17 16:30:37.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"b359d493d14faaade2e101ae2af99f8a483187eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 17:29:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/964b1cfdc02b471d9660e1bb39643c7c : FAILURE in 25m 40s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/54e27d94e1544983970316063d2679e9 : FAILURE in 24m 15s","accounts_in_message":[],"_revision_number":9},{"id":"0faa6f8af1cfd839eb66d0bb12bbcc461fdffd8c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-17 18:10:50.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"1664d1becffcddc55ba722c33fd4a36246d34035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 18:58:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9eceab6f8a7b4e18a5b2eaa6269bf36f : FAILURE in 40m 27s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/761c3b0edcc6480ca6bafe8d51faa500 : FAILURE in 43m 29s","accounts_in_message":[],"_revision_number":10},{"id":"cb5b27738ff1b4a93cd35d565304850d4f118e95","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-17 21:10:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1385bf676fb963ec59b3dd43b9aaa88e96805768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 21:38:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ee49634734e0437c892af11ab9584d53 : FAILURE in 18m 11s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b2d984c702da4427b83b72b8fea674b0 : FAILURE in 21m 57s","accounts_in_message":[],"_revision_number":11},{"id":"ba1149319953721b0a47686681c55005dc4ed69f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-18 04:15:00.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"3bfe90db5d6a7554fbc38545aa75898b748bc6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 04:30:05.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dd39439fe95447e090f3ccb77644f46a : FAILURE in 14m 00s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7865f4ca37b847b3bff48d761e4d2f72 : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":11},{"id":"8d4d8ca5752f3421f1f25d830215bee16673e6d8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-20 19:53:11.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"1b02221df5dcd6dca883107909cddd6b725f19d5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-20 20:04:42.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"dffbee9180c3110e7d4aeb52990f94c761884c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:12:57.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7b4de4b8b5714102b14ef3068b748a01 : SUCCESS in 26m 25s\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6e453b8f3c6743418301badd0d7e0475 : FAILURE in 1h 04m 01s","accounts_in_message":[],"_revision_number":12},{"id":"1584a3a7b76ce8ec163197373f53fdbac46f6769","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-20 22:04:20.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"527691ddd0918d39cfea6e96df76103484393a62","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-20 22:04:44.000000000","message":"Patch Set 13:\n\nrun experimental","accounts_in_message":[],"_revision_number":13},{"id":"43d5527b098bd2980e78802b708047c28c8f69c2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-20 22:05:12.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"d0a59ae54658be87996e1f6bdf22007b31b6a2d3","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 22:33:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\n\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/43d919ef135e48c3a399c020c34cf4dc : FAILURE in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f3a1953b0178657b8267b99d085fce5f61540598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 23:56:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b80b169ced24e5cb6dbcc00c51d09fe : SUCCESS in 2m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/68d20594dfcc421596a72b6f6f6091d5 : SUCCESS in 30m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b976a28e59fa411fa374e06a597507bd : SUCCESS in 33m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b91ad0a7cbed444689d3c9f59131aa81 : SUCCESS in 1h 42m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1a32c1d53b0e48a89c0678552c5906bd : SUCCESS in 1h 03m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ccc5d2cd5981487ba61640735234003d : FAILURE in 1h 02m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9d842d5ff4e54b8b83d542800c07277b : FAILURE in 47m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ac8b4167dd534a1d8aad12db5af255f2 : SUCCESS in 46m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0d70740395964b35ad018799d9589244 : FAILURE in 1h 50m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f235df86d3d44630ad3276bb4d6b8d82 : FAILURE in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"28fe6b693428308b9c69b9fbef357d9b37ef30fd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 00:57:39.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/51/731651/13/check/DellEMC_VNX_Manila/f890412/DellEMC_VNX_Manila/96 : FAILURE in 39m 20s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":13},{"id":"9a9e404c6ae142438ddaafe30df6f81844f5cec0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-21 02:05:33.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/51/731651/13/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6b79827/ : SUCCESS in 2h 09m 04s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/51/731651/13/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/7165fcd/ : SUCCESS in 1h 38m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"4f24d514bc2a064bbd1dbb8dcc1524066ee90ace","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-21 05:06:52.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"8959e9a65e9fbbd4902ff2a300daff4c9cef287f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 07:08:03.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade2ba2127e64f619c32555e6423b330 : SUCCESS in 3m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/07e14b7e3e124c60bea993021d0988dd : SUCCESS in 30m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5857aa9247b34e54afb05c1ab9fc08cb : SUCCESS in 29m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb160ae4a7604f63af593f936bde9537 : SUCCESS in 57m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/66cdec3050d74a9dbbbceb10516f0bf2 : SUCCESS in 1h 20m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db9f4d6c362c4b5db5619d388cdabbf2 : SUCCESS in 1h 00m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d1cb937ec0124de380f2fdd9cea6e8ed : SUCCESS in 55m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/47403d53ac4044c98437254c19372e6b : SUCCESS in 48m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b8d595cbc8994e13a859996c5e4902ca : FAILURE in 1h 55m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/91346e37a0b54f628bfd8ef6cbcc1af8 : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b8e0dfc3d9eb8050ff8207fc7c61b62de2bf54f4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-21 08:11:45.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/51/731651/13/check/DellEMC_VNX_Manila/5a06b98/DellEMC_VNX_Manila/100 : SUCCESS in 1h 00m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":13},{"id":"131e164e554577940a92095b74ef31f186c5cd08","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-21 09:17:12.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/51/731651/13/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c75f21d/ : FAILURE in 2h 08m 19s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/51/731651/13/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/29f3be6/ : SUCCESS in 1h 23m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"a1005e80d38db3a566a0c5a4698b7dba56783843","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-07-21 19:02:38.000000000","message":"Patch Set 13: Code-Review+2\n\nLooks good, Goutham! Thanks","accounts_in_message":[],"_revision_number":13},{"id":"3ad9b904ad2eeba565636aa330727a3589af4fb0","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-21 19:29:12.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\nThanks, Goutham!","accounts_in_message":[],"_revision_number":13},{"id":"9f599211173b6bd463788e816736ef9b73cd5e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:47:59.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"748f9787d7c9a0d2692baf5debffdb3a83d1c7cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:50:25.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f5e936ed9f4be299d7a7bd3b0ac3cc : SUCCESS in 4m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9d760b02717646e3875d5fbde1bd8576 : SUCCESS in 26m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea5b4f19c4144b15a0d4bbed4e5c16e4 : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/408c4b2789d549dba3e98834834fcc8c : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":13},{"id":"a61407db0b6295d80e0a7618909188ad2f5febb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:50:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"842d08b4c868f10cfd24416f4c39041127116ffc","revisions":{"380fa5ba38c580e21cbd9ec3eeac9f32431602d3":{"kind":"REWORK","_number":1,"created":"2020-05-29 05:26:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/1"}}},"commit":{"parents":[{"commit":"0a40f10dc81ae0c4d0d2ae056416870d0a76a84b","subject":"Merge \"Zuulv3 generic driver testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a40f10dc81ae0c4d0d2ae056416870d0a76a84b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/380fa5ba38c580e21cbd9ec3eeac9f32431602d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/380fa5ba38c580e21cbd9ec3eeac9f32431602d3"}]},"branch":"refs/heads/master"},"2179de93b75a8b3a4625acd14fd450ee1acd3afb":{"kind":"REWORK","_number":2,"created":"2020-05-29 05:27:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/2"}}},"commit":{"parents":[{"commit":"0a40f10dc81ae0c4d0d2ae056416870d0a76a84b","subject":"Merge \"Zuulv3 generic driver testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a40f10dc81ae0c4d0d2ae056416870d0a76a84b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:27:25.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2179de93b75a8b3a4625acd14fd450ee1acd3afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2179de93b75a8b3a4625acd14fd450ee1acd3afb"}]},"branch":"refs/heads/master"},"a1cb71d31ea0e30678646ad6325f4fdc8ced70df":{"kind":"REWORK","_number":3,"created":"2020-05-29 05:29:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/3"}}},"commit":{"parents":[{"commit":"0a40f10dc81ae0c4d0d2ae056416870d0a76a84b","subject":"Merge \"Zuulv3 generic driver testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a40f10dc81ae0c4d0d2ae056416870d0a76a84b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:29:30.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a1cb71d31ea0e30678646ad6325f4fdc8ced70df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a1cb71d31ea0e30678646ad6325f4fdc8ced70df"}]},"branch":"refs/heads/master"},"42bf52e93d1fa1572d7be172dc5b00312029eaa2":{"kind":"REWORK","_number":4,"created":"2020-07-16 18:18:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/4"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 18:17:47.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/42bf52e93d1fa1572d7be172dc5b00312029eaa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/42bf52e93d1fa1572d7be172dc5b00312029eaa2"}]},"branch":"refs/heads/master"},"751f07b0a58201bd8d69eee9782022683077abb3":{"kind":"REWORK","_number":5,"created":"2020-07-16 19:47:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/5"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 19:47:16.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/751f07b0a58201bd8d69eee9782022683077abb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/751f07b0a58201bd8d69eee9782022683077abb3"}]},"branch":"refs/heads/master"},"cae38835491bb84db6236516ca96b2440b895f95":{"kind":"REWORK","_number":6,"created":"2020-07-16 20:42:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/6"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 20:41:58.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nDepends-On: https://review.opendev.org/741542\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cae38835491bb84db6236516ca96b2440b895f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cae38835491bb84db6236516ca96b2440b895f95"}]},"branch":"refs/heads/master"},"06cf7abe89340a1a7eed1839d534f58317de21a9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-07-16 22:44:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/7"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 22:44:05.000000000","tz":0},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nDepends-On: https://review.opendev.org/741556/\nDepends-On: https://review.opendev.org/741542\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/06cf7abe89340a1a7eed1839d534f58317de21a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/06cf7abe89340a1a7eed1839d534f58317de21a9"}]},"branch":"refs/heads/master"},"947429a9771f059a858d18fa86ebe5470a73c0c3":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-07-17 16:30:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/8"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-17 16:30:28.000000000","tz":0},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/947429a9771f059a858d18fa86ebe5470a73c0c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/947429a9771f059a858d18fa86ebe5470a73c0c3"}]},"branch":"refs/heads/master"},"28e8e07185ba0b1730f588a5991b0e87cc3d27d1":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-07-17 16:30:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/9"}}},"commit":{"parents":[{"commit":"517a7a250770e4b43940eec637de97b126672a3c","subject":"Fix ReplicaQuotasNegativeTest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/517a7a250770e4b43940eec637de97b126672a3c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-17 16:30:37.000000000","tz":0},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/28e8e07185ba0b1730f588a5991b0e87cc3d27d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/28e8e07185ba0b1730f588a5991b0e87cc3d27d1"}]},"branch":"refs/heads/master"},"046e5bb9f7a571cbca8e06542a276d0b57b9f14c":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-07-17 18:10:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/10"}}},"commit":{"parents":[{"commit":"517a7a250770e4b43940eec637de97b126672a3c","subject":"Fix ReplicaQuotasNegativeTest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/517a7a250770e4b43940eec637de97b126672a3c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-17 18:10:48.000000000","tz":0},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nDepends-On: https://review.opendev.org/741696\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/046e5bb9f7a571cbca8e06542a276d0b57b9f14c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/046e5bb9f7a571cbca8e06542a276d0b57b9f14c"}]},"branch":"refs/heads/master"},"811a74c774712b738922a6e9306ed7ee8bb9251a":{"kind":"REWORK","_number":11,"created":"2020-07-17 21:10:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/11"}}},"commit":{"parents":[{"commit":"df804d802010ce2193f779b0dde20d8daa5f12ba","subject":"Add Zuulv3 native jobs for the dummy driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df804d802010ce2193f779b0dde20d8daa5f12ba"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-17 21:02:41.000000000","tz":-420},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nDepends-On: https://review.opendev.org/741696\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/811a74c774712b738922a6e9306ed7ee8bb9251a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/811a74c774712b738922a6e9306ed7ee8bb9251a"}]},"branch":"refs/heads/master"},"2ea19bf077adb5123e28806e7f9e84663e44d1f2":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2020-07-20 20:04:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/12"}}},"commit":{"parents":[{"commit":"df804d802010ce2193f779b0dde20d8daa5f12ba","subject":"Add Zuulv3 native jobs for the dummy driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df804d802010ce2193f779b0dde20d8daa5f12ba"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-20 20:04:41.000000000","tz":0},"subject":"WIP Zuulv3 native GlusterFS native, NFS jobs","message":"WIP Zuulv3 native GlusterFS native, NFS jobs\n\nDepends-On: https://review.opendev.org/742020\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ea19bf077adb5123e28806e7f9e84663e44d1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ea19bf077adb5123e28806e7f9e84663e44d1f2"}]},"branch":"refs/heads/master"},"842d08b4c868f10cfd24416f4c39041127116ffc":{"kind":"REWORK","_number":13,"created":"2020-07-20 22:04:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/731651/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/731651/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/731651/13"}}},"commit":{"parents":[{"commit":"df804d802010ce2193f779b0dde20d8daa5f12ba","subject":"Add Zuulv3 native jobs for the dummy driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df804d802010ce2193f779b0dde20d8daa5f12ba"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-29 05:25:21.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-20 22:03:39.000000000","tz":-420},"subject":"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs","message":"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\n\nGlusterFS jobs were included in this repository as\nnon-voting experimental jobs. However, they depend\non devstack-gate, a deprecated component. Converting\nthem to native zuulv3 style will allow ease of\nmaintenance. These jobs were only ever running \"api\"\ntests and were not testing integration with the rest\nof OpenStack, so to conserve resources, these jobs\ndeploy a minimal devstack with just keystone and manila\nenabled.\n\nThe GlusterFS-NFS job was fixed in the Train release,\nwe can promote it to the regular CI check queue,\nkeeping it non-voting at the moment.\n\nHowever, the GlusterFS-native job has been pretty\nfailure prone. The legacy DSVM job has no history of\nsucceeding as of this commit [1]. We\u0027ll leave it\nin the experimental queue until we triage what\u0027s\nwrong with the job.\n\nWhile NFS and Native testing was refreshed in the\nTrain release, \"heketi\" layout remains an unmaintained\ncomponent of the GlusterFS driver for Manila. This\ncommit deletes the experimental jobs for heketi.\nThese can be revived when a maintainer shows up,\nwilling to invest in the heketi layout for GlusterFS.\n\n[1] https://zuul.opendev.org/t/openstack/builds?job_name\u003dmanila-tempest-dsvm-glusterfs-native\nChange-Id: Iaef7873e7539045753bb5676fe93f4ded586b5ac\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/842d08b4c868f10cfd24416f4c39041127116ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/842d08b4c868f10cfd24416f4c39041127116ffc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
