)]}'
{"id":"openstack%2Fmanila-tempest-plugin~736806","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I6fb5a26d8e5c021226f02d496fe086cf345c1cd2","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/update-cephfs-drivers","hashtags":[],"change_id":"I6fb5a26d8e5c021226f02d496fe086cf345c1cd2","subject":"Enable snapshot tests in the CephFS job","status":"MERGED","created":"2020-06-18 19:15:15.000000000","updated":"2020-07-13 14:09:59.000000000","submitted":"2020-07-13 14:09:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736806-1594649399244-167245bb","meta_rev_id":"66d154bcab7377a3dcc53fc11fb750bb7e054a0a","_number":736806,"virtual_id_number":736806,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-07-13 13:14:50.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 14:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-07-13 10:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-07-13 10:34:01.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":2,"date":"2020-07-13 10:34:01.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-07-13 10:34:05.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 20:02:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-06-23 19:28:13.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-07-11 12:01:57.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-13 10:32:12.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 10:34:05.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-07-13 13:14:50.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 14:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"975b8cd6646d432ff12a3c4334d4e13e70311b02","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-18 19:15:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef74ce74d811733d1474bd01b58e863ada8ad523","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-18 19:16:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0721f8af82672fd69ae49c136ef9df0dd1cea391","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-18 20:10:33.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e2caa053d4d6389f5fca707d2b5ae1ad93ff5e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:10:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":3},{"id":"806d2b82f55914b09f11247e7c26c083467ad425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:10:45.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":3},{"id":"bee1cf1b7a1433eebdae5ab3e9f7ac197a633fb8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-18 20:21:45.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fb2101580e58a57cf3da34efdf9b23dba0054ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:59:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/27cc062e21ac4febb438f086c064a62f : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4269e7e6e2f0c077957924d059a630a5b9b46c7b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-19 16:19:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ca0b6d30a60e0662f1aadb2a5daee3307aaf849d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:28:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7805db5575654f3788991ed44b6b83b0 : FAILURE in 1h 06m 16s","accounts_in_message":[],"_revision_number":4},{"id":"9e3e25a258f2b0490a1fbd6c411671f5fe880302","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-19 19:58:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e11b55e6e8a66c00694d2f539d21d6158921d1d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:12:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e540459e5d547869f6ba13dd29b6c45 : SUCCESS in 3m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d20421048e69423da8b7162616e81222 : SUCCESS in 41m 49s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5c4cb59824844748bee0919d3df1f239 : SUCCESS in 44m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2023c9586c614d8e832c2f9d4a3d3679 : FAILURE in 1h 10m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d517b600f41d48e3b74c3dec7b758a7f : SUCCESS in 47m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/406e8488e9ad40e68f5a39d9f858355e : FAILURE in 48m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af00ab8123d34d1d8b15cd5aa7372a3f : FAILURE in 1h 12m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/00d7613c45204fc3a7414de1a1ba7cef : POST_FAILURE in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"79d2a31093d0da97859f22d179be7a0683f00aa9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-19 23:22:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/06/736806/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ee558b0/ : SUCCESS in 2h 09m 01s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/06/736806/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4166ff7/ : SUCCESS in 1h 36m 18s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"16b1ac9471b0c4c8bd750403f34b1fc300e2ef43","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-22 20:02:50.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM. Thanks, Goutham :)","accounts_in_message":[],"_revision_number":5},{"id":"ed8f50a4ee24c57bfdd2ecf7a27962cf297976a9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-06-23 19:28:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLooks good, thanks Goutham","accounts_in_message":[],"_revision_number":5},{"id":"10a302307010f88a03679416754162648f3c6ce8","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-07-07 00:04:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"abf0bdf84a2dbdf165ce509e184d8b77109fbb4f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-07 00:56:47.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/06/736806/6/check/DellEMC_Unity_Manila_dhss_true/2d0b5fc/DellEMC_Unity_Manila_dhss_true/132 : FAILURE in 51m 48s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/06/736806/6/check/DellEMC_Unity_Manila_dhss_false/3058e5b/DellEMC_Unity_Manila_dhss_false/130 : SUCCESS in 50m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"aaa1b03916d212ccd1ee7273da9b1428d29ae21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 01:52:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef43713f78b04a09811d9a7eb6d25139 : SUCCESS in 3m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/60671a686fec409e9f9226707ab03916 : FAILURE in 1h 00m 56s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3bfc0c44b8394ded93c96fa7a5f60a21 : SUCCESS in 40m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bf03e70060744f44bb00c3f9ad8c68bb : FAILURE in 1h 46m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/217b9ca3670f4277be52d5d1ae84ba31 : SUCCESS in 1h 08m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c96b4ed3e06c47aea588af39083d6252 : SUCCESS in 54m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c9609faff219427e9a0fd32025046af2 : FAILURE in 1h 45m 54s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4f98f69e3742455ca5d238d92a53a88b : FAILURE in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"44465363149d060c5d47aa81bea77d21446deae2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-07 17:20:07.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1e63be3057ed539806d5445ee5732553db7fe68e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 20:19:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006cde5f2a514a6f9cddbbde37cc392d : SUCCESS in 4m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ca4c9725e1134c62aac358a4ce723d90 : SUCCESS in 1h 12m 41s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a8dd9757776445b8a5cc25073dd6d0b2 : FAILURE in 47m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/184125366b304ec6b62d3d7f47853059 : FAILURE in 1h 50m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/517e70d1d8c0424eb121ec924676e6ff : SUCCESS in 1h 17m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3fb545186cee4d26963a9e67fbb71e74 : SUCCESS in 49m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/468a8cd755054b258f037e7a8f8cde9e : FAILURE in 2h 02m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fbc94a0164a64bf29656fa2b9d089441 : FAILURE in 1h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6846d865a6b23f7dc45e9a8fb2a63dea56419ebd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-07 23:28:04.000000000","message":"Patch Set 6:\n\nrecheck\n\nBase patch has been updated, should address the manila-tempest-minimal-dsvm-cephfs-nfs failure.","accounts_in_message":[],"_revision_number":6},{"id":"9268ac2f946fe2addc259fd495ac990b88b17d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 01:09:14.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"784a919808df99341003aca69ff88086cc591ffd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 01:17:10.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"92b59a01d989489cd06c3cfbd4dd0961c99ccbde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 03:24:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc09d3e719434e8793a5a939231846a8 : SUCCESS in 3m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/946f6e02312445d89ff4d424c5455c0b : SUCCESS in 1h 02m 47s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7c174ccab36b43888a579d24fb2bc9af : SUCCESS in 52m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58e4f54b69664035b8375d9046433038 : FAILURE in 2h 01m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/286681d1b6f04b78bd8b3edcff3792db : SUCCESS in 1h 19m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/63ee8eaaf53548b990a641c87006824f : SUCCESS in 48m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2a9826fe3a7142cebe4b6977c114f7c6 : FAILURE in 1h 47m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4e22553f22264294b73d11141113ed4c : FAILURE in 1h 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"592e5d19f4d9a2a73631e8a778c5a096733facdc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-08 04:43:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/06/736806/6/check/DellEMC_Unity_Manila_dhss_true/bfa17e3/DellEMC_Unity_Manila_dhss_true/146 : FAILURE in 56m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/06/736806/6/check/DellEMC_Unity_Manila_dhss_false/d7c2a54/DellEMC_Unity_Manila_dhss_false/142 : SUCCESS in 49m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"a9e603b622f37f949cbe0f8619e23ebaaef76937","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 05:16:01.000000000","message":"Patch Set 6:\n\nrecheck\n\nparent patch has been updated again.","accounts_in_message":[],"_revision_number":6},{"id":"9a9d8b0aaa51d227c906e05703d398160930f3cb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-08 05:33:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/06/736806/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a9ffff3/ : SUCCESS in 2h 08m 41s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/06/736806/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b3433e8/ : SUCCESS in 1h 24m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"6021f16982aa4a9901575dd4f0e8313cc6e1b3f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 07:46:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a1751de7db44bba511eb638020d006 : SUCCESS in 4m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/66c6783de3b34adeb666e484ed7f38a1 : FAILURE in 1h 15m 27s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/33d212c4c749490685826463a8b397c1 : SUCCESS in 46m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fb2a8208684649c9953235501ed91893 : FAILURE in 1h 15m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/85446ffd698d4b62be79cd666663128f : SUCCESS in 1h 42m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/872d2294743747c3941eb57d2296812a : SUCCESS in 48m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/279fbdbc6a39452aa79b007915417a1b : FAILURE in 2h 23m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/63243d433818425bb4417c3f40b38e75 : FAILURE in 1h 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7e351fa3c485f4e534f2cf90b2aeb210f46dd10b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-08 17:18:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"afd4daa89b59a19f2c0a47cca2bd337cc6529ed1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-10 21:27:22.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3313c38e917432d24de64ad43a04f635fee95089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 00:24:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aecc3ede44f4b09a4040b5981e2bbfc : SUCCESS in 4m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/94101191bc734de98d11cd58c6b65cb4 : SUCCESS in 57m 34s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/aa737f234c5c429db926f1eac94bf5e6 : SUCCESS in 46m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/06f33c0ae6e44e0e93e11aca79e36b18 : FAILURE in 1h 54m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9ffb48811e324741ab1aecc533700255 : FAILURE in 2h 33m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c59952d25764320beba72bd104fb451 : SUCCESS in 54m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d022ceee892645aa8cc115d12cf96a9e : FAILURE in 1h 32m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/41d1b3c67ace47f280f028cace2edcf4 : SUCCESS in 1h 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"52b93678f399a3eb0a7659eadc9b5ef126ffbc24","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-11 06:13:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/06/736806/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e91ee3f/ : SUCCESS in 2h 12m 43s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/06/736806/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/41edb3e/ : SUCCESS in 1h 22m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"729175225772cfbb2aab612de9032e1fcdccf96a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-11 11:06:25.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0dead88d3b19f03c92c3e743fb5015270eefe6b5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-11 11:06:35.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4d4287c78cca51229d58deedcf2fb10decce9ed2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-11 12:01:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/06/736806/7/check/DellEMC_Unity_Manila_dhss_true/ab0bfaa/DellEMC_Unity_Manila_dhss_true/164 : FAILURE in 43m 42s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/06/736806/7/check/DellEMC_Unity_Manila_dhss_false/5fbb5d1/DellEMC_Unity_Manila_dhss_false/148 : SUCCESS in 55m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"8aeeb268bee2bdbb25d6b1333a47f4eefc0775b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 13:02:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843f10ee6a6d41e59ac9383e9a20de9d : SUCCESS in 4m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b117b98f50a246aca9450eb2f15e6854 : SUCCESS in 37m 26s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/41c6bda392264dd99d5d24820bc35f79 : FAILURE in 36m 54s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2f0d8a1a19d64855b3bfdedbba58c868 : FAILURE in 1h 47m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d9c64014867d41cf9b1af7c71bca5963 : SUCCESS in 1h 11m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7644bf222494409781fd4176bc41c35a : SUCCESS in 41m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4fb71a13ce04d48abb10433a6acc0ef : FAILURE in 1h 51m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d8e64adb2b114cf692c1c35abf8b415a : SUCCESS in 1h 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0e7f25624efed2e838c721c00d17056c481ca28e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-11 13:32:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"9b445d0ce22654f5036cfcdddc70880304553315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 16:11:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd16076a88404eb08b0621e5cce78e41 : SUCCESS in 4m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4aaa782e206c496bbd4e4a2c91f1f05d : SUCCESS in 35m 25s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0b3cd46034454dc3be8a79a27ca43d51 : FAILURE in 48m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a455303686c44fa91f23595b490451a : SUCCESS in 1h 17m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c5f528d6e0c148fb9db4bd2d0e177574 : SUCCESS in 1h 39m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f27466f98cc0426380e9fb4764d161ba : SUCCESS in 49m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/06983bc4b0bf468c8cb9685b946a5b4d : FAILURE in 2h 37m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/28c9527e24b04131b714e75c3f37cc8e : SUCCESS in 1h 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"68dfc836ab55c22d0012cc2c62634d947f0b4975","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-13 05:17:03.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"47d3eea07c67595f316fbf38f49769929b026986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 08:23:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a528163782cf43b38b0ed4bf37cffae1 : SUCCESS in 3m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c86b32d913b84cf8b09fd2b47219cf45 : SUCCESS in 55m 19s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/afc9de58e64f46139d60a96a4a08d543 : SUCCESS in 47m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d350ee4bb2f242a9b2b79d2ae8512162 : FAILURE in 1h 44m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9dc886cf50454872bf976a4d5c604e89 : SUCCESS in 2h 18m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6b2a6232f0641409daffccf34d674bb : SUCCESS in 46m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9f6d091836684973a50e4531bfead3c9 : FAILURE in 3h 04m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/505c8edd7eff47d1b3ea817c9ac1d5cd : TIMED_OUT in 2h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2b0ec0aecc3eb905b8e43759cbd4b1c87b823416","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-13 08:23:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/06/736806/7/check/DellEMC_VNX_Manila/372c723/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"a40d1caa222fa0795f4dfde426b62811dd7e26dd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-13 10:32:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/06/736806/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a6c083e/ : SUCCESS in 2h 08m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/06/736806/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/322f16e/ : SUCCESS in 1h 05m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"c228ac246bacb2f09d1c086e9eb179ae1feaa3e8","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 10:34:01.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"d1974fbc28c141895bef51e374900994a009d424","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 10:34:05.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e39738e5a9c6a72eb0c3530fc683d403a0fd6d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 13:14:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0372fc42b6342149ba4739a2038eb84 : SUCCESS in 5m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f34a58f4a18848e8bd6b4425c08ddfe3 : SUCCESS in 29m 19s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f11496a4ca5a4339b825c2942aed2c70 : SUCCESS in 44m 03s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cd039eb0d3364018a1bc2e651b0c4854 : RETRY_LIMIT in 39m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/89dff74eca044e3d873db5083da51966 : SUCCESS in 50m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3c5135f8d45d48ebb5ccee64b32ef444 : SUCCESS in 46m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7341a0767cdb46298e1908b954fdf9c5 : FAILURE in 1h 44m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8b1ea4428f2f48009620e48961e5a071 : SUCCESS in 1h 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ecb6a3ce6527be546ec3581fd02c61cba2e42a38","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-13 13:14:50.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/06/736806/8/check/DellEMC_VNX_Manila/e557cdd/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"f4059668214f0bef00f18da4cbeb6eded9ac66f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 13:14:52.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fad070c8c1f0bca280119f7add729e334cdf0823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:09:58.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13dae7d275dd4f0ab7f3f0b963d083d6 : SUCCESS in 3m 40s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1e9ebeff44144d43b8cf68c8662de070 : SUCCESS in 39m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d175dfc91714722881068d56447b1c9 : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":8},{"id":"8d2fdb9caeb2ac3f5430768f178d8babe71a9adb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:09:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"2ed5826913da39453476a6eb1ab095900200e377","revisions":{"60ceed09a80a7478faee28c8a8f956c9de0cdb68":{"kind":"REWORK","_number":1,"created":"2020-06-18 19:15:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/736806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/1"}}},"commit":{"parents":[{"commit":"8257d9e9c095a6d145217217f5c84edfa29384bc","subject":"Merge \"Add CephFS Native scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8257d9e9c095a6d145217217f5c84edfa29384bc"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 19:14:53.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 19:14:53.000000000","tz":-420},"subject":"WIP: Enable snapshot tests in the CephFS job","message":"WIP: Enable snapshot tests in the CephFS job\n\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60ceed09a80a7478faee28c8a8f956c9de0cdb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60ceed09a80a7478faee28c8a8f956c9de0cdb68"}]},"branch":"refs/heads/master"},"9ed55a6d05249b5aa11588d3a68a88963286368d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-18 19:16:10.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/736806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/2"}}},"commit":{"parents":[{"commit":"8257d9e9c095a6d145217217f5c84edfa29384bc","subject":"Merge \"Add CephFS Native scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8257d9e9c095a6d145217217f5c84edfa29384bc"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 19:14:53.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 19:15:42.000000000","tz":-420},"subject":"WIP: Enable snapshot tests in the CephFS job","message":"WIP: Enable snapshot tests in the CephFS job\n\nDepends-On: https://review.opendev.org/676722\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ed55a6d05249b5aa11588d3a68a88963286368d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ed55a6d05249b5aa11588d3a68a88963286368d"}]},"branch":"refs/heads/master"},"03653cb9e89bec3cbec3ca485aff07ff957cd626":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-18 20:10:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/736806/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/3"}}},"commit":{"parents":[{"commit":"8257d9e9c095a6d145217217f5c84edfa29384bc","subject":"Merge \"Add CephFS Native scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8257d9e9c095a6d145217217f5c84edfa29384bc"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 19:14:53.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 20:10:32.000000000","tz":0},"subject":"WIP: Enable snapshot tests in the CephFS job","message":"WIP: Enable snapshot tests in the CephFS job\n\nDepends-On: https://review.opendev.org/736806\nDepends-On: https://review.opendev.org/676722\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03653cb9e89bec3cbec3ca485aff07ff957cd626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03653cb9e89bec3cbec3ca485aff07ff957cd626"}]},"branch":"refs/heads/master"},"cef2aaeffe1fd8265990451ce6abc0ba1300cca5":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-18 20:21:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/736806/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/4"}}},"commit":{"parents":[{"commit":"8257d9e9c095a6d145217217f5c84edfa29384bc","subject":"Merge \"Add CephFS Native scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8257d9e9c095a6d145217217f5c84edfa29384bc"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 19:14:53.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-18 20:21:43.000000000","tz":0},"subject":"WIP: Enable snapshot tests in the CephFS job","message":"WIP: Enable snapshot tests in the CephFS job\n\nDepends-On: https://review.opendev.org/676722\nDepends-On: https://review.opendev.org/736827\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cef2aaeffe1fd8265990451ce6abc0ba1300cca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cef2aaeffe1fd8265990451ce6abc0ba1300cca5"}]},"branch":"refs/heads/master"},"a3973a88dfff6b712cc85c8d694ed2e75d4f21b8":{"kind":"REWORK","_number":5,"created":"2020-06-19 19:58:47.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/736806/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/5"}}},"commit":{"parents":[{"commit":"f8894b6f0b3d98394d6c573a065ac10baa7e621c","subject":"Skip snapshot cloning if there is no capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f8894b6f0b3d98394d6c573a065ac10baa7e621c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 19:57:40.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 19:58:20.000000000","tz":-420},"subject":"Enable snapshot tests in the CephFS job","message":"Enable snapshot tests in the CephFS job\n\nThis is only for the Ceph Native job for now,\nthe new Zuulv3 job for Ceph-NFS will have\nsnapshot tests enabled when it lands, see [1]\n\n[1] https://review.opendev.org/726088/\n\nPartially-implements: bp update-cephfs-drivers\nDepends-On: https://review.opendev.org/676722/\nDepends-On: https://review.opendev.org/736827/\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3973a88dfff6b712cc85c8d694ed2e75d4f21b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3973a88dfff6b712cc85c8d694ed2e75d4f21b8"}]},"branch":"refs/heads/master"},"3237ba65c8c6310b336cafec12789a8d41f74c05":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-07 00:04:51.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/06/736806/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/6"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 19:57:40.000000000","tz":-420},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2020-07-07 00:04:51.000000000","tz":0},"subject":"Enable snapshot tests in the CephFS job","message":"Enable snapshot tests in the CephFS job\n\nThis is only for the Ceph Native job for now,\nthe new Zuulv3 job for Ceph-NFS will have\nsnapshot tests enabled when it lands, see [1]\n\n[1] https://review.opendev.org/726088/\n\nPartially-implements: bp update-cephfs-drivers\nDepends-On: https://review.opendev.org/676722/\nDepends-On: https://review.opendev.org/736827/\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3237ba65c8c6310b336cafec12789a8d41f74c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3237ba65c8c6310b336cafec12789a8d41f74c05"}]},"branch":"refs/heads/master"},"41501a51007a9b4573a2c87b2fcd235a771deb53":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-11 11:06:25.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/06/736806/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/7"}}},"commit":{"parents":[{"commit":"f3731e7a869a29d25cf70a488bc0605e3d29f64b","subject":"Merge \"[Negative] Test different tenants cannot use same cephx id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f3731e7a869a29d25cf70a488bc0605e3d29f64b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 19:57:40.000000000","tz":-420},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-11 11:06:25.000000000","tz":0},"subject":"Enable snapshot tests in the CephFS job","message":"Enable snapshot tests in the CephFS job\n\nThis is only for the Ceph Native job for now,\nthe new Zuulv3 job for Ceph-NFS will have\nsnapshot tests enabled when it lands, see [1]\n\n[1] https://review.opendev.org/726088/\n\nPartially-implements: bp update-cephfs-drivers\nDepends-On: https://review.opendev.org/676722/\nDepends-On: https://review.opendev.org/736827/\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/41501a51007a9b4573a2c87b2fcd235a771deb53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/41501a51007a9b4573a2c87b2fcd235a771deb53"}]},"branch":"refs/heads/master"},"2ed5826913da39453476a6eb1ab095900200e377":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-07-13 10:34:01.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/06/736806/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/736806/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/736806/8"}}},"commit":{"parents":[{"commit":"c80e538a47c0be47eea193deaa8ce7eb971792ae","subject":"Merge \"Test different users in same tenant can use same cephx id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c80e538a47c0be47eea193deaa8ce7eb971792ae"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 19:57:40.000000000","tz":-420},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-13 10:34:01.000000000","tz":0},"subject":"Enable snapshot tests in the CephFS job","message":"Enable snapshot tests in the CephFS job\n\nThis is only for the Ceph Native job for now,\nthe new Zuulv3 job for Ceph-NFS will have\nsnapshot tests enabled when it lands, see [1]\n\n[1] https://review.opendev.org/726088/\n\nPartially-implements: bp update-cephfs-drivers\nDepends-On: https://review.opendev.org/676722/\nDepends-On: https://review.opendev.org/736827/\nChange-Id: I6fb5a26d8e5c021226f02d496fe086cf345c1cd2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ed5826913da39453476a6eb1ab095900200e377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2ed5826913da39453476a6eb1ab095900200e377"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
