)]}'
{"id":"openstack%2Fmanila-tempest-plugin~737055","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia617d54951988f844a7da857a1c842b794191404","project":"openstack/manila-tempest-plugin","branch":"master","topic":"cephfs-snapshots","hashtags":[],"change_id":"Ia617d54951988f844a7da857a1c842b794191404","subject":"Skip snapshot cloning if there is no capability","status":"MERGED","created":"2020-06-19 17:44:39.000000000","updated":"2020-06-22 22:13:31.000000000","submitted":"2020-06-22 21:04:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737055-1592859887788-0147b0e7","meta_rev_id":"87b3979b58b2c5954ec22aa79fd06a68ec464f4c","_number":737055,"virtual_id_number":737055,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-06-22 20:48:29.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-06-22 16:56:20.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-06-22 19:16:27.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 21:04:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-22 22:13:31.000000000","post_submit":true,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-06-22 19:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-22 19:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2020-06-22 19:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2020-06-22 19:16:44.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 19:16:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 19:16:44.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-06-22 19:42:29.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-06-22 19:43:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-06-22 20:48:29.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 21:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 22:13:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"}],"messages":[{"id":"2803f5fb2eb6f0db177a7e75e62dad4a3be05f7b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-19 17:44:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a34ea1ebbdc9238b8f3be3a3598e793809120d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:18:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b42a0e2d070d484c8b663d84fda8d46e : SUCCESS in 6m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fe2d98cdc9814da58433b6a6ebe726a1 : FAILURE in 1h 16m 35s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8ff7a39be5954a4f980ae22c8f9517fc : SUCCESS in 51m 11s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c3152004de714bfab2a7f45eaface8fa : SUCCESS in 1h 13m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3f874144821149dcab7560e1351ed390 : FAILURE in 3h 23m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/32bf083480694a9c95606cb2bdb8889d : SUCCESS in 49m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8703c32d6b7c431d9250132fb982b10f : FAILURE in 1h 39m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b0a050bb83344f5fb2712fdbe2355679 : FAILURE in 1h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ed5b81df432febca8f1f8fc00c63bae2f43d4ef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-19 21:29:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76e5afd50614c14b240f7a9e46c390fc136e0bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 23:01:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b835b6b403174969b3692b0ec2e07191 : SUCCESS in 3m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fef55c17fec14fb185128e267a7db17e : FAILURE in 1h 11m 07s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/15bcf7583b054e879297db8c581ef70e : SUCCESS in 40m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cdea18ef1969460faef540d48085e3f6 : SUCCESS in 58m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aceea78027e64068bae39041964d9800 : SUCCESS in 1h 08m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3a9c2adb4b9b4813aa70b7f867eb01f3 : FAILURE in 1h 01m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c0be3a64b40e44fab437060e4a205b4d : FAILURE in 1h 30m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7026e20a9ba947f88467c58a83747d1c : SUCCESS in 1h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4967f5264d25e9e6fe12963ceed6ae2681a47bf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-22 16:56:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"162897950830a4ba12e093fd9d387f2c4122cff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:16:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9562a6d526924613ba47b8140fa6d27f : SUCCESS in 3m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d593604920c14c409887ed1b80201c6c : SUCCESS in 54m 10s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0d00ba14000c4b3db56a6f1537acdc5c : SUCCESS in 46m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c078ddb61f184b549be21e267a3ee363 : SUCCESS in 1h 04m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ee6de802c04b4febb18915445ffb1aa4 : SUCCESS in 1h 25m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/57e1de1fb89f4811b27d9d8cfe021af0 : SUCCESS in 49m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/33f1b3f101264ef28437f62de995995e : FAILURE in 2h 07m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f531b670621445c0a20e5d233bc47291 : SUCCESS in 1h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e13a70d2de44ea6be3bd324dd291433259c53a47","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-22 19:16:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/55/737055/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/58f2014/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/55/737055/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/80278a4/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"62d77df643aeba0712c6fb0c99324aea07dc2c95","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-06-22 19:42:29.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, LGTM!","accounts_in_message":[],"_revision_number":1},{"id":"f4da486ca011445dab972e2eeacdf7a6d468015b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-22 19:43:46.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. Thanks for improving this, Goutham :)","accounts_in_message":[],"_revision_number":1},{"id":"5f69b59d0622a91e7f5eeed0572bf188d97f335d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-22 19:43:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a491653391a585c0845231efb328d24cc8b6b648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:44:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf176c2c05a79e8acb5471e6d7924f22cfae298b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-06-22 20:48:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/55/737055/1/check/DellEMC_VNX_Manila/8c0bdcc/DellEMC_VNX_Manila/52 : FAILURE in 1h 02m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"a8bd88e56298d72036f60026ed1fbf78125f88f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 21:04:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79597442fe5a9e4411d19afde0bb7f09be0cfde3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 21:04:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737c193a3e194a27a136ce77b2930978 : SUCCESS in 5m 35s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5b9a4356405d43ef8c84b31f5e1b4adb : SUCCESS in 56m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6ddcaaa840604699a26037717c2a0feb : SUCCESS in 45m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b6b4f89e401d78b1d320db7edae52a504f5b2cb7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-22 22:13:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/55/737055/1/check/DellEMC_Unity_Manila_dhss_true/c950fee/DellEMC_Unity_Manila_dhss_true/81 : SUCCESS in 1h 47m 20s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/55/737055/1/check/DellEMC_Unity_Manila_dhss_false/714b272/DellEMC_Unity_Manila_dhss_false/80 : SUCCESS in 48m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8894b6f0b3d98394d6c573a065ac10baa7e621c","revisions":{"f8894b6f0b3d98394d6c573a065ac10baa7e621c":{"kind":"REWORK","_number":1,"created":"2020-06-19 17:44:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/55/737055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/55/737055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/55/737055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/55/737055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/55/737055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/55/737055/1"}}},"commit":{"parents":[{"commit":"8257d9e9c095a6d145217217f5c84edfa29384bc","subject":"Merge \"Add CephFS Native scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8257d9e9c095a6d145217217f5c84edfa29384bc"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 17:41:55.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-06-19 17:44:19.000000000","tz":-420},"subject":"Skip snapshot cloning if there is no capability","message":"Skip snapshot cloning if there is no capability\n\nShare drivers can support snapshots (\u0027snapshot_support\u0027),\nbut needn\u0027t support cloning snapshots into\nshares (\u0027create_share_from_snapshot\u0027). Tempest tests\nacknowledge this capability, however, a scenario test\nis missing an appropriate skip decorator.\n\nChange-Id: Ia617d54951988f844a7da857a1c842b794191404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f8894b6f0b3d98394d6c573a065ac10baa7e621c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f8894b6f0b3d98394d6c573a065ac10baa7e621c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
