)]}'
{"id":"openstack%2Fmanila-tempest-plugin~738724","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id5323a5c38842dbf706aec75459d2086ca142635","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1699856","hashtags":[],"change_id":"Id5323a5c38842dbf706aec75459d2086ca142635","subject":"Seek network resources from tempest","status":"MERGED","created":"2020-07-01 02:42:54.000000000","updated":"2020-07-14 09:58:47.000000000","submitted":"2020-07-14 09:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738724-1594720727706-c502fe0b","meta_rev_id":"1a2bdfe2b53e8b32fcce4af6bd5761fdbdbc70f3","_number":738724,"virtual_id_number":738724,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"date":"2020-07-14 02:24:01.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-07-14 06:19:14.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-07-14 04:33:26.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-07 05:44:45.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 09:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-13 20:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-07-08 15:48:38.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"date":"2020-07-13 05:01:00.000000000","_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-13 20:11:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 05:44:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-08 15:48:38.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-07-13 05:01:00.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2020-07-13 20:11:17.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-07-14 02:24:01.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-14 04:33:26.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-14 09:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"346d71d501e30a3b3b47de1ffecd73ec8c5dba6c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 02:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1992c3537d0961bcd0f18747e418356238b64039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 04:35:33.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c184318581349909b971c89539c2cd9 : FAILURE in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f29ea5fd9385460593f4579c40ef699d : SUCCESS in 1h 06m 11s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b6927444a31f4fdd856b667192db17d1 : SUCCESS in 42m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ae479c8f917241348d2d7451443f6531 : SUCCESS in 1h 02m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c3332d85d59f424b98aae3c7049472bb : SUCCESS in 1h 07m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/400dc3da209a481fb04229694cc3c0d7 : SUCCESS in 47m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8b9042320114445690db4872c1c0e271 : FAILURE in 1h 50m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c29cb7632f4042ec8f340746746819d2 : SUCCESS in 57m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3df84763a6d67a9f9a09daa5d4dd3255e82cf973","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-01 12:35:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThis is a very good optimization and cleanup. Thanks Goutham.","accounts_in_message":[],"_revision_number":1},{"id":"5a7dc8c0d75d4f265c05c53b9bc563fa7a158e3b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-01 12:35:32.000000000","message":"Patch Set 1:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":1},{"id":"4375b4cec6a078a77437208b9a54a74391b04fe4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 12:54:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"245bdfdc3ea393c4130f6986cf2de2b24421f29d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 12:54:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e468729b678e2154ca47e25ca598e2a5d0679f2e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 12:56:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60848b8c4b6859b6d35bcdee72ae56a23cc1af73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 15:02:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b070fcfec1410b8036558817af6cc0 : SUCCESS in 5m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/669c5aad06aa4dde8107db83e9d9fd49 : SUCCESS in 1h 07m 43s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e73beb46b3a54ccebb795f99482d1997 : FAILURE in 47m 54s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d68e1f7611ec4ef78bf9d0300ca87925 : SUCCESS in 58m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/93a2c1010f1143feb726a78b562ae8e4 : SUCCESS in 1h 13m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6e20a2e287a84a5cad2b567930a7f2d0 : SUCCESS in 58m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/255c47971caa4030913136723a212967 : FAILURE in 2h 01m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e0ae56f47e994994a4a4e3467902b957 : SUCCESS in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f45ddfb9cd6ad10e0d1bb9c1676cbd5ab4af3b3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 15:04:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c53e44f9b0515cdc38b90b3c4bb04194eeb608a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:37:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329882e830994e3b81a54ee15e45e366 : SUCCESS in 6m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b7ee0d407c22470ebe2950d476ea0166 : SUCCESS in 1h 06m 31s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/763e1512359e46439294de2df76e2584 : SUCCESS in 53m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e552eb4db07949709bbdd31afd3fa8e1 : FAILURE in 1h 12m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/47442593f6ec42e39191cf079ca4fe89 : FAILURE in 1h 04m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3e7a7fec5c86431690961ea042e62cab : FAILURE in 57m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/472c41ad4f984edebfcfc5ca1d8c9bbb : FAILURE in 2h 26m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2198862f0f074be7b8f3501a17936fc2 : SUCCESS in 50m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"888f85aaa651f4f5100bff3c66501f817bd662dc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-01 19:40:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/24/738724/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/fe31404/ : FAILURE in 1m 38s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/24/738724/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0317efa/ : SUCCESS in 1h 31m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"abe1ab9ad10a846c8171e23a8034d34799cd7b46","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 20:32:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f27ff3ab22aed27864900f4d01034d80b50b6c3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 20:33:32.000000000","message":"Patch Set 4:\n\nI\u0027ll handle share network cleanup in another patch - it can cleanly be separated from this work here..","accounts_in_message":[],"_revision_number":4},{"id":"81c278902e37c9f0d3788951dc0f5f06b23146ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 23:02:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235eb2714ab34bdf83ac27a0c67ef343 : SUCCESS in 5m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb4049201a2343ee8fa8b407693a86ca : FAILURE in 1h 20m 06s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/41ef782ac67e4bf88c9423c04b50da8d : SUCCESS in 43m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ea25c41755354000b1661ab05da90e05 : FAILURE in 1h 12m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d0f53ed0f1454e9a87dcd58186ee777b : FAILURE in 1h 26m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa57cd0791fc4a6d83bfb576e3978c3d : SUCCESS in 53m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9684452ed5624af68fc7d3ed5509fb32 : FAILURE in 2h 24m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9880cb530e2640d0b4cf368ee95a6ada : SUCCESS in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1aedcae9ecef81b473deaed47abc6ec883bffb8d","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-01 23:38:25.000000000","message":"Patch Set 4:\n\nzfsonlinux reported some weird errors[1], lets try again.\n\nrecheck\n\n[1] https://919e51072639993b5fa1-deafdf99ece277fbd0a62f16555071c1.ssl.cf1.rackcdn.com/738724/4/check/manila-tempest-plugin-zfsonlinux/ea25c41/testr_results.html","accounts_in_message":[],"_revision_number":4},{"id":"908c1890d77bea587e729eb48cdb6b3d300dbd3d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-02 05:45:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bffd7ba88ceeb410ee7c269330b2d470fef9a6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:53:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0961a07f994ec0a2d030ac0173781a : SUCCESS in 4m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a10ab7d64c674b3691e281aaeecc6a20 : RETRY_LIMIT in 55m 08s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c724428462ee4a6fa0c90c120a7ec116 : SUCCESS in 54m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4aba0ef7cda04a66aafaa4196e5a7658 : FAILURE in 1h 46m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ef4e99d43dfe48a9834546ade29e1788 : FAILURE in 1h 31m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5827fe95bf5c4efdbf137d3e1712e837 : SUCCESS in 1h 00m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0c2107111cbf45188b3a3312c3c09f9a : FAILURE in 2h 33m 30s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1fa0185731e045388bbce20c893784c2 : SUCCESS in 1h 08m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98f443ba3a26dd02e27aba6f4993efd359f4cce1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-07 04:52:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"54becefca7e2f9ceb39dff841af180f330172457","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-07 05:44:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/24/738724/5/check/DellEMC_Unity_Manila_dhss_true/1542109/DellEMC_Unity_Manila_dhss_true/134 : FAILURE in 51m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/24/738724/5/check/DellEMC_Unity_Manila_dhss_false/900ebb7/DellEMC_Unity_Manila_dhss_false/132 : SUCCESS in 46m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"919fd303cc866eae5abecf568ece84e1fd81512c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 06:44:45.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad17f77e5fe342edad3dc5ffb43ad810 : SUCCESS in 3m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/faaddb467548484587182b9788187834 : FAILURE in 1h 02m 48s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b496741504c844dba5bc3de8a2e070b3 : SUCCESS in 50m 08s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/682e2f575de74058aa257457bbb7cf33 : FAILURE in 1h 46m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6e394db57db742e4804c36fb108ed0b8 : SUCCESS in 51m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fbc83eff0d564f7a803b088d3b9cc47e : SUCCESS in 40m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a72e1cebfbd84872b42cd15f97933c80 : FAILURE in 1h 38m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5acb2c1d507443ad9ac520d0d15a7c9a : FAILURE in 47m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a4346527e468abccfdf7fc96e7a8149704228ca0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-07 23:32:21.000000000","message":"Patch Set 5:\n\nrecheck\n\nI\u0027ve seen this LVM job fail multiple times when running on rax-iad nodes - The logs are gone from the job, but, on debugging a noop change, we see this issue over and over where the node just freezes and is rebooted. I think rax-iad has an issue right now.. we\u0027ll reattempt a few times to see if this test job gets scheduled elsewhere.","accounts_in_message":[],"_revision_number":5},{"id":"a76e9a89046cfc824e9efe688722dc6a880597d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 02:42:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/742d0b320f0640cc88d440b77cc14563 : SUCCESS in 4m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bfb219c406d647a5bb52ddbae9157044 : RETRY_LIMIT in 50m 56s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e9430915efd84d2b8197a7cda933b71c : SUCCESS in 50m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8283a2f2454649f08dd293276efe8b73 : FAILURE in 1h 04m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/60d35418dd994c92ac51d808b4bea4e9 : SUCCESS in 1h 12m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/500a43b76883497ba1abe16e0da56df5 : SUCCESS in 52m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eeea687cc70b4f3cb57675ab79f7a84f : FAILURE in 1h 48m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d7363960ca164d81972b3fa4e3c789c6 : FAILURE in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"22798999119fec0af25d82b4588ae702b45e6fa0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 05:30:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"883b1fe12a8469befae77812ed89754dadb2eb9c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-08 05:30:35.000000000","message":"Patch Set 5:\n\nrun netapp-ci","accounts_in_message":[],"_revision_number":5},{"id":"d77ae77484db36372f49c1137ab0204d07f52e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:59:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38345789d6a4f8cb0fef558cd23ff8d : SUCCESS in 3m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ea96c1c5141648ccad484548cc69af9d : RETRY_LIMIT in 58m 48s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/1478821099f64acfbc33eb13764796cf : SUCCESS in 56m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6838122cdc9646d68d05c8025608c5c8 : RETRY_LIMIT in 50m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/79c218c60dd44a7a8f9dce1ef80a5943 : SUCCESS in 51m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bf0fe47d8a754b06a96f9889efee0633 : SUCCESS in 57m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2841a7c1d2c648f6a77bf64d936df162 : FAILURE in 1h 49m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c04486336cdf45a5bc20b1f4cb13428c : FAILURE in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1d88a3512ce78f6c684ea34283bacf29705adc8b","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-08 15:48:38.000000000","message":"Patch Set 5: Code-Review+2\n\nNice cleanup.\n\nGate is flaky and we likely have work to do before this can merge but I think all test failures are unrelated to this patch.","accounts_in_message":[],"_revision_number":5},{"id":"55a5df592aa5fd1f9b8ed5c963f7c98ddca80954","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-10 21:27:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"48515053d50b49e4f6af8a1d14ca49658dfc3684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:05:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad1180879994d44a3d7e70dbc1d34a6 : SUCCESS in 4m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e8ff04df185b470b809c5c598a268c83 : SUCCESS in 35m 39s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/55ec3fee46fe4b8691c77b4508c9e6d3 : FAILURE in 1h 24m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b567a8fba0ce4a8bb99d0f0f00df363c : FAILURE in 1h 31m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/68bcc85c6c42409b9f97196c7198fab8 : FAILURE in 1h 17m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/14d982346c74446c8ede2315ca04e24a : SUCCESS in 54m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cda99d65e0d74f068240b32574612457 : FAILURE in 1h 25m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7ab3526d37d446e996606e2c335c1523 : SUCCESS in 58m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9818be34b522591e6aa84fd5be7766360f7cbc98","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-13 05:00:50.000000000","message":"Patch Set 5:\n\nrecheck\n\nThe Dummy driver job broke badly but it should be nothing related to this patch;","accounts_in_message":[],"_revision_number":5},{"id":"5626e604a8ac715052f6431690bd9e0ef79d306c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-13 05:01:51.000000000","message":"Patch Set 5:\n\nHi, Sam, \n\nThis is a follow up patch that cleans up much of the share network stuff and makes things simpler - would be glad if you could take a look at this patch too!","accounts_in_message":[],"_revision_number":5},{"id":"58210f36a931aec0a176332bfcb5474c9c1811c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 08:08:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47cdf17f70f6489e82e4026377cbed6b : SUCCESS in 4m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3a457b9989974ebaba81de228dd53c80 : SUCCESS in 1h 27m 47s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ba4913296aca4ecab997f4161d5ebca6 : SUCCESS in 1h 39m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f00ada6d928a462e934745e14ffb481d : FAILURE in 2h 06m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c0385deda2fb4a95a4616e63d8bfa131 : FAILURE in 2h 11m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/42efd6531e8844eea013b874cfa8f2c5 : SUCCESS in 1h 39m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/35ebd7e5adfa4ef58e20bccff4bdb3da : TIMED_OUT in 3h 05m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/14cef8a952f1464b9b1c4016f67121e1 : FAILURE in 1h 47m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b4f00c879ae3dea2d88d3f46933935e89fba044e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-13 08:09:11.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/24/738724/5/check/DellEMC_VNX_Manila/21f1009/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"2bf954e6c6e724feee70e71867fb3bd3423087db","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-13 15:02:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/24/738724/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/9af3623/ : SUCCESS in 2h 00m 54s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/24/738724/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f4dfa83/ : SUCCESS in 3h 24m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"1f331cf3d27a4db26b26782cd871622e10f54e62","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-13 20:09:29.000000000","message":"Patch Set 5: Code-Review+2\n\nGreat job on this cleanup! Great improvement on removing this on-demand credentials that were responsible only for network creation and moving this duty to tempest framework.\nGate issues doesn\u0027t seems related to this change. \nThanks Goutham!","accounts_in_message":[],"_revision_number":5},{"id":"026771f956526c8ce23e99707f3ad32c554dc06e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-13 20:11:17.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0d3b95ae6a30ceeac3c36e40b3ece3d0d25c043d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:11:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b9cbdde960dd29e0e6e07666ee9e2517ecdad8bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:49:46.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc4b4583df040dd85f4997a4419c4dd : SUCCESS in 3m 40s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e3b5eab996fb40408ff26dc7d9cc65b2 : SUCCESS in 34m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e390eac9bb494c748adc645c3632b88f : FAILURE in 37m 18s","accounts_in_message":[],"_revision_number":5},{"id":"5de639207ce9543c5831d34b294714ea19c84065","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-13 21:10:15.000000000","message":"Patch Set 5:\n\nrecheck\n\nlvm job failed due to timeout[1]\n\n[1] https://a97f2a1acf6438ad06dd-fa9050fbccbc1b8e2fcd252255e35175.ssl.cf1.rackcdn.com/738724/5/gate/manila-tempest-plugin-lvm/e390eac/testr_results.html","accounts_in_message":[],"_revision_number":5},{"id":"7550b9eb97cd592c941c24af868a4804a456d0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 23:32:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ccd2f6f24a7433a9a52225499e94934 : SUCCESS in 3m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ecfa0d31169845e18dbb04059b1c3c60 : SUCCESS in 38m 24s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d1b6983a4c3c45c2bf637505b40b4869 : FAILURE in 57m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/129662418d2c40858a75dcd728d88f46 : FAILURE in 1h 33m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d4950aa171c342f8bd2503bfe4c8b065 : FAILURE in 1h 05m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/93d4195e57464034aeacb409a9a74957 : SUCCESS in 59m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/61b57b7815744a67a8f124d242500ca9 : SUCCESS in 47m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bf95bed82c124c0cb5eb9b237fa45897 : FAILURE in 2h 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"274b6ecb2ddfa2fd54a2fe57bddce1158e935a15","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-14 00:15:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"252e55ebee7c6f1c03184c0977d737ee00439e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 02:23:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7905af77c29a400e85e8a6c79acfab8e : SUCCESS in 2m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/25b420b3756346ab9c447795c35877b4 : SUCCESS in 36m 46s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/56dc6b71fde44bc382ade8dd5c37eeea : SUCCESS in 43m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1bd84679c6534948910c0603243c836b : FAILURE in 1h 52m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7a880fa40ce242c3a0a29a5046213de8 : FAILURE in 1h 06m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6faf8430c0e84c5390d01996d2d3296f : FAILURE in 57m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2236336f934b44daabab14c0ebf6bb57 : SUCCESS in 40m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d69a7454fa3f46c7aeb243cec9687627 : FAILURE in 2h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5cf08b16098d9b44dfff54871545de75d2ceed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 02:23:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"230d04caee8d31d4f59d892a41c902ad5a9b7e5c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-14 02:24:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/24/738724/5/check/DellEMC_VNX_Manila/302f611/DellEMC_VNX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d7b5f9d02b55cab6fe60e032cca4b8952ae46be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 03:19:06.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d382fcb0f4a44b1902b5714302c2ea0 : SUCCESS in 4m 25s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9e3350410bb645c083781cfb6213b50b : SUCCESS in 44m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/47c781a695ad4a52998dae13f863f5ee : FAILURE in 53m 27s","accounts_in_message":[],"_revision_number":5},{"id":"f56914bf507d9c711a19cbb5752365e6bdad5bf9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-14 04:33:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/24/738724/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2a35af3/ : FAILURE in 2h 09m 20s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/24/738724/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/193f91e/ : SUCCESS in 1h 33m 26s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"774defe558706f60c73710d22589750bfbd26234","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-14 06:19:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c104d11484784b0194b509f47a1c835e465e61f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:10:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86df0b05a4094df7ade8929968a8a6b9 : SUCCESS in 6m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e26f329945b41598e3eab35436a2ec7 : SUCCESS in 40m 55s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b1785b9884c449b8b9442f0f96478477 : SUCCESS in 39m 21s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/422b9293b453485c9a2a937f6845ce05 : SUCCESS in 45m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/da5fd09afa324526ae02c29733ee28f0 : FAILURE in 1h 03m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f99081877d1641cabdec23cc6f087ddc : SUCCESS in 55m 22s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2bf2f9353c6f49c6b77dd5d271200fa6 : SUCCESS in 46m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/10c2eaa5c90b4924a3bffc95303d25f5 : FAILURE in 1h 59m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"15d0b53583ec2b0a8084022f21292485864907dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:10:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"506f7cad114b525e04e5c3204e7a35a1caced8be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:58:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce581ade033545509c573df44ca114cb : SUCCESS in 5m 32s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2380b8eaab704b17bcb649330cad1f6f : SUCCESS in 44m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13192dc3f0e34057a0d85e72e5c83d55 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":5},{"id":"bd881bdc6385442b82eb70bb46035ce07e039936","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:58:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4a0b732aabb6130ab9216f318f454bd91398c3d1","revisions":{"0e1ce73348f7076b943a7c8ecc9f97e645c22e82":{"kind":"REWORK","_number":1,"created":"2020-07-01 02:42:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/24/738724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/738724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/1"}}},"commit":{"parents":[{"commit":"af004188c232c9837c2314a7261beb163d24c771","subject":"Add security service to share networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/af004188c232c9837c2314a7261beb163d24c771"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 02:42:45.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 02:42:45.000000000","tz":-420},"subject":"Seek network resources from tempest","message":"Seek network resources from tempest\n\nIn order to do DHSS\u003dTrue testing, we need\nshare networks. Share networks are really\nuseful only when testing with self-service\nproject/tenant neutron networks.\n\nManila tempest tests are executed\nby using different projects - these can\nbe created on the fly using tempest\u0027s dynamic\ntest credential generation framework. We\nhave always controlled what type of\nprojects/credentials are created by tempest\nto run a test class when using dynamic\ncredential generation. However, we were\nso far not asking tempest to create network\nresources (neutron network, subnet, router, etc)\nwhen creating these credentials.\n\nWe can do so now, and delete all the code we\nhad to create networks ourselves.\n\nWith this change, we will raise a SkipException\nif tests are supposed to be run with neutron\nnetworks, but we somehow can\u0027t find them.\nThis error can ONLY occur if:\n- tempest is mis-configured, or\n- pre-provisioned test credentials are used and\n  networks aren\u0027t created beforehand.\n\nChange-Id: Id5323a5c38842dbf706aec75459d2086ca142635\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e1ce73348f7076b943a7c8ecc9f97e645c22e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e1ce73348f7076b943a7c8ecc9f97e645c22e82"}]},"branch":"refs/heads/master"},"a7483fe234d43332a5c6b48764f2804626134243":{"kind":"REWORK","_number":2,"created":"2020-07-01 12:54:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/24/738724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/738724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/2"}}},"commit":{"parents":[{"commit":"78073c5b683ae4cf69af2fbcef50d71c1d829bc5","subject":"Add security service to share networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78073c5b683ae4cf69af2fbcef50d71c1d829bc5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 02:42:45.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 12:53:48.000000000","tz":-420},"subject":"Seek network resources from tempest","message":"Seek network resources from tempest\n\nIn order to do DHSS\u003dTrue testing, we need\nshare networks. Share networks are really\nuseful only when testing with self-service\nproject/tenant neutron networks.\n\nManila tempest tests are executed\nby using different projects - these can\nbe created on the fly using tempest\u0027s dynamic\ntest credential generation framework. We\nhave always controlled what type of\nprojects/credentials are created by tempest\nto run a test class when using dynamic\ncredential generation. However, we were\nso far not asking tempest to create network\nresources (neutron network, subnet, router, etc)\nwhen creating these credentials.\n\nWe can do so now, and delete all the code we\nhad to create neutron networks ourselves.\n\nWith this change, we will raise a \"skipException\"\nif tests are required to use neutron networks\n(configuration option:\n\"[share]/create_networks_when_multitenancy_enabled\u003dTrue\"),\nbut we somehow can\u0027t find them. This error can ONLY occur if:\n- tempest is mis-configured, or\n- pre-provisioned test credentials are used and\n  networks aren\u0027t created beforehand.\n\nChange-Id: Id5323a5c38842dbf706aec75459d2086ca142635\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a7483fe234d43332a5c6b48764f2804626134243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a7483fe234d43332a5c6b48764f2804626134243"}]},"branch":"refs/heads/master"},"6f608b188538b2e0a1f772d674bb7c6c28151159":{"kind":"REWORK","_number":3,"created":"2020-07-01 12:56:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/24/738724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/738724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/3"}}},"commit":{"parents":[{"commit":"78073c5b683ae4cf69af2fbcef50d71c1d829bc5","subject":"Add security service to share networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78073c5b683ae4cf69af2fbcef50d71c1d829bc5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 02:42:45.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 12:56:00.000000000","tz":-420},"subject":"Seek network resources from tempest","message":"Seek network resources from tempest\n\nIn order to do DHSS\u003dTrue testing, we need\nshare networks. Share networks are really\nuseful only when testing with self-service\nproject/tenant neutron networks.\n\nManila tempest tests are executed\nby using different projects - these can\nbe created on the fly using tempest\u0027s dynamic\ntest credential generation framework. We\nhave always controlled what type of\nprojects/credentials are created by tempest\nto run a test class when using dynamic\ncredential generation. However, we were\nso far not asking tempest to create network\nresources (neutron network, subnet, router, etc)\nwhen creating these credentials.\n\nWe can do so now, and delete all the code we\nhad to create neutron networks ourselves.\n\nWith this change, we will raise a \"skipException\"\nif tests are required to use neutron networks\n(configuration option:\n\"[share]/create_networks_when_multitenancy_enabled\u003dTrue\"),\nbut we somehow can\u0027t find them. This error can ONLY occur if:\n- tempest is mis-configured, or\n- pre-provisioned test credentials are used and\n  networks aren\u0027t created beforehand.\n\nChange-Id: Id5323a5c38842dbf706aec75459d2086ca142635\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f608b188538b2e0a1f772d674bb7c6c28151159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f608b188538b2e0a1f772d674bb7c6c28151159"}]},"branch":"refs/heads/master"},"02f212f812abe8bb6f8761cf98423bd01b7b4ea8":{"kind":"REWORK","_number":4,"created":"2020-07-01 20:32:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/24/738724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/738724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/4"}}},"commit":{"parents":[{"commit":"78073c5b683ae4cf69af2fbcef50d71c1d829bc5","subject":"Add security service to share networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78073c5b683ae4cf69af2fbcef50d71c1d829bc5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 02:42:45.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 20:31:48.000000000","tz":-420},"subject":"Seek network resources from tempest","message":"Seek network resources from tempest\n\nIn order to do DHSS\u003dTrue testing, we need\nshare networks. Share networks are really\nuseful only when testing with self-service\nproject/tenant neutron networks.\n\nManila tempest tests are executed\nby using different projects - these can\nbe created on the fly using tempest\u0027s dynamic\ntest credential generation framework. We\nhave always controlled what type of\nprojects/credentials are created by tempest\nto run a test class when using dynamic\ncredential generation. However, we were\nso far not asking tempest to create network\nresources (neutron network, subnet, router, etc)\nwhen creating these credentials.\n\nWe can do so now, and delete all the code we\nhad to create neutron networks ourselves.\n\nWith this change, we will raise a \"skipException\"\nif tests are required to use neutron networks\n(configuration option:\n\"[share]/create_networks_when_multitenancy_enabled\u003dTrue\"),\nbut we somehow can\u0027t find them. This error can ONLY occur if:\n- tempest is mis-configured, or\n- pre-provisioned test credentials are used and\n  networks aren\u0027t created beforehand.\n\nChange-Id: Id5323a5c38842dbf706aec75459d2086ca142635\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/02f212f812abe8bb6f8761cf98423bd01b7b4ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/02f212f812abe8bb6f8761cf98423bd01b7b4ea8"}]},"branch":"refs/heads/master"},"4a0b732aabb6130ab9216f318f454bd91398c3d1":{"kind":"REWORK","_number":5,"created":"2020-07-07 04:52:49.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/24/738724/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/738724/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/738724/5"}}},"commit":{"parents":[{"commit":"1753a59b43448a77165103558d8d7a318b83f173","subject":"Add security service to share networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1753a59b43448a77165103558d8d7a318b83f173"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 02:42:45.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-07 04:52:28.000000000","tz":-420},"subject":"Seek network resources from tempest","message":"Seek network resources from tempest\n\nIn order to do DHSS\u003dTrue testing, we need\nshare networks. Share networks are really\nuseful only when testing with self-service\nproject/tenant neutron networks.\n\nManila tempest tests are executed\nby using different projects - these can\nbe created on the fly using tempest\u0027s dynamic\ntest credential generation framework. We\nhave always controlled what type of\nprojects/credentials are created by tempest\nto run a test class when using dynamic\ncredential generation. However, we were\nso far not asking tempest to create network\nresources (neutron network, subnet, router, etc)\nwhen creating these credentials.\n\nWe can do so now, and delete all the code we\nhad to create neutron networks ourselves.\n\nWith this change, we will raise a \"skipException\"\nif tests are required to use neutron networks\n(configuration option:\n\"[share]/create_networks_when_multitenancy_enabled\u003dTrue\"),\nbut we somehow can\u0027t find them. This error can ONLY occur if:\n- tempest is mis-configured, or\n- pre-provisioned test credentials are used and\n  networks aren\u0027t created beforehand.\n\nChange-Id: Id5323a5c38842dbf706aec75459d2086ca142635\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a0b732aabb6130ab9216f318f454bd91398c3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a0b732aabb6130ab9216f318f454bd91398c3d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
