)]}'
{"id":"openstack%2Fmanila-tempest-plugin~738937","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie090c6e56211b651d0dd48c2021e922c3562576d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1699856","hashtags":[],"change_id":"Ie090c6e56211b651d0dd48c2021e922c3562576d","subject":"WIP Cleanup share servers on share network deletion","status":"NEW","created":"2020-07-01 20:48:54.000000000","updated":"2020-08-21 17:43:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58aa51cf4bb0de724cc0ac9d029d2bd83531c8d1","_number":738937,"virtual_id_number":738937,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-21 17:14:54.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"date":"2020-08-21 17:43:25.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-08-21 17:07:06.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-21 16:02:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 13:18:03.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-08-21 16:02:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 17:07:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-08-21 17:14:54.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 17:43:25.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c95b176c80955dd2a40042823e4a25b93cab9790","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 20:48:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3ba6a194bd41b3bd2a314f0c46ebc00af216417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 23:00:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f9d92f54c241c190f49439105821c9 : SUCCESS in 4m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5cf8a732912241b5b756ecf4721fbea3 : FAILURE in 1h 18m 53s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d701369c56b9456eb26a8eaf19a98c13 : FAILURE in 48m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a4166f0d3ef4404a6b6bb23c440266f : FAILURE in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/533a6a6d0d77427abede859b6c26fb38 : FAILURE in 1h 26m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/35582bcb7b5f40948ef4e5b3fa3a0060 : FAILURE in 59m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a84610b200c944adaf030e0a84e594a4 : FAILURE in 2h 05m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9fe11194ec504fe283cd837f7e11f9fd : SUCCESS in 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb0b4eb9bec7d06d157818288ad2aa3dca42b4aa","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-21 13:18:03.000000000","message":"Patch Set 1:\n\nrecheck\n\nlogs are no longer available to debug","accounts_in_message":[],"_revision_number":1},{"id":"b0015443fe391aab70f7e49f6941c19601eb037c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 13:18:14.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"34400afb144e0b35a3df3be647adee6dc12ed5d3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-21 13:58:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9bac987f5f137a3e290b27f453d3c1a9f46b8344","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-21 15:05:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/37/738937/2/check/DellEMC_Unity_Manila_dhss_true/06119e6/DellEMC_Unity_Manila_dhss_true/299 : FAILURE in 1h 06m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/37/738937/2/check/DellEMC_Unity_Manila_dhss_false/6618187/DellEMC_Unity_Manila_dhss_false/282 : SUCCESS in 59m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"689c2cadbe5434f4c7e6f39218875bde2ade3e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:02:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4534a3aba44b53915721040ccdb3b1 : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e575e0aadad4ac5b33487b1499142ac : SUCCESS in 33m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71686b42d646478fb5ed2ad01399b0ff : SUCCESS in 40m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/66bfb3b5d1c44425a8120f6f26eb3dfd : SUCCESS in 1h 30m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7ef215b38f3c453a95895cd756342041 : SUCCESS in 1h 14m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6b39790acfdd4f7b89239ed5e7233663 : FAILURE in 1h 36m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/80f15884129f45b8869d7775374e63ec : FAILURE in 1h 17m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9b2a9f76fee04c5b8811479ced9c9169 : FAILURE in 49m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2cf63a10b09745039d3c931c74f893fd : FAILURE in 2h 02m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0cd8c054d7b443f0a4e31e904d147dca : SUCCESS in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62687c2055b6c74e9dbd5b46134141a841994f95","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-21 17:07:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/37/738937/2/check/DellEMC_Unity_Manila_dhss_true/c835ae2/DellEMC_Unity_Manila_dhss_true/300 : FAILURE in 1h 04m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/37/738937/2/check/DellEMC_Unity_Manila_dhss_false/a8dd2ca/DellEMC_Unity_Manila_dhss_false/283 : SUCCESS in 57m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fd3d5328cdc0b5d837e94525224dc3592a0b1841","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-21 17:14:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/37/738937/2/check/DellEMC_VNX_Manila/c8c70b4/DellEMC_VNX_Manila/175 : FAILURE in 1h 11m 46s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"482bada256d1078d25f1a3aa02ee3b11015f3219","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-21 17:43:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/37/738937/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8064e04/ : FAILURE in 1h 40m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/37/738937/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e26d3fd/ : SUCCESS in 1h 35m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51db55f1fccc96e899f87a34238c5ff737fd5699","revisions":{"87fc9a07fd1754f9d3b6a73be36a534c6e6867b0":{"kind":"REWORK","_number":1,"created":"2020-07-01 20:48:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/37/738937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/37/738937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/1"}}},"commit":{"parents":[{"commit":"02f212f812abe8bb6f8761cf98423bd01b7b4ea8","subject":"Seek network resources from tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/02f212f812abe8bb6f8761cf98423bd01b7b4ea8"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 20:48:42.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 20:48:42.000000000","tz":-420},"subject":"WIP Cleanup share servers on share network deletion","message":"WIP Cleanup share servers on share network deletion\n\nChange-Id: Ie090c6e56211b651d0dd48c2021e922c3562576d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/87fc9a07fd1754f9d3b6a73be36a534c6e6867b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/87fc9a07fd1754f9d3b6a73be36a534c6e6867b0"}]},"branch":"refs/heads/master"},"51db55f1fccc96e899f87a34238c5ff737fd5699":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-21 13:58:48.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/37/738937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/37/738937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/37/738937/2"}}},"commit":{"parents":[{"commit":"1369e54a2d8c4874f2b90f98c77e115349efb22a","subject":"Fix skip condition for \"test_default_share_group_type_applied\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1369e54a2d8c4874f2b90f98c77e115349efb22a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-01 20:48:42.000000000","tz":-420},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-21 13:58:48.000000000","tz":0},"subject":"WIP Cleanup share servers on share network deletion","message":"WIP Cleanup share servers on share network deletion\n\nChange-Id: Ie090c6e56211b651d0dd48c2021e922c3562576d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/51db55f1fccc96e899f87a34238c5ff737fd5699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/51db55f1fccc96e899f87a34238c5ff737fd5699"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
