)]}'
{"id":"openstack%2Fmanila-tempest-plugin~739832","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I047397a797403f9e40ba29103bc9fdbe71ead852","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I047397a797403f9e40ba29103bc9fdbe71ead852","subject":"WIP - should fail","status":"ABANDONED","created":"2020-07-07 19:08:55.000000000","updated":"2020-07-13 17:10:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c802c096e4f9d9cf3d533f02e2421aebde57ca7d","_number":739832,"virtual_id_number":739832,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-07 22:33:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-07-07 21:04:04.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 21:04:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-07 22:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6b1e701215dfd4813e6cc96d4e298b7b0118736","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-07 19:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a324f15e288b46b6216d4c619aba4ba7f3bd2ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-07 21:04:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/32/739832/1/check/DellEMC_Unity_Manila_dhss_true/71cc895/DellEMC_Unity_Manila_dhss_true/141 : FAILURE in 55m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/32/739832/1/check/DellEMC_Unity_Manila_dhss_false/d963ef1/DellEMC_Unity_Manila_dhss_false/137 : FAILURE in 50m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1c35dbf85d435a8f9d18cf8c0328abac346833b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 22:33:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c2630d37f8249b5ae4906759b02001d : SUCCESS in 4m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8b50366a1f3c4203985deba7a1d43605 : FAILURE in 1h 14m 45s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f6a430fef2b54494b6b2046c0da19383 : FAILURE in 37m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/866cdd80ff5140c4b8979a18da94f231 : RETRY_LIMIT in 54m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ecc2d16edca848209437998b715f915f : FAILURE in 1h 23m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/765c155a5e114ccbbffcdde70e6373af : FAILURE in 51m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1a45b1adde23485bbd018b2591b62f36 : FAILURE in 1h 55m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/190fff2e7bb3497cafa1e0729e4552d6 : FAILURE in 1h 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c802c096e4f9d9cf3d533f02e2421aebde57ca7d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-13 17:10:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5471524568889f3cd597e5c191ed02cb0d3f62a","revisions":{"b5471524568889f3cd597e5c191ed02cb0d3f62a":{"kind":"REWORK","_number":1,"created":"2020-07-07 19:08:55.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/32/739832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/32/739832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/739832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/739832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/739832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/32/739832/1"}}},"commit":{"parents":[{"commit":"1ba5e647ae73bf16dffb1ccc416d7b560c845d5c","subject":"Merge \"Stop forcing creation of \"isolated\" credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ba5e647ae73bf16dffb1ccc416d7b560c845d5c"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-07 19:08:41.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-07 19:08:41.000000000","tz":-240},"subject":"WIP - should fail","message":"WIP - should fail\n\nChange-Id: I047397a797403f9e40ba29103bc9fdbe71ead852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5471524568889f3cd597e5c191ed02cb0d3f62a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5471524568889f3cd597e5c191ed02cb0d3f62a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
