)]}'
{"id":"openstack%2Fmanila-tempest-plugin~741376","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4fd282673074887cada729b6e4491ee493e230cc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-tenant-id","hashtags":[],"change_id":"I4fd282673074887cada729b6e4491ee493e230cc","subject":"Fix ReplicaQuotasNegativeTest","status":"MERGED","created":"2020-07-16 05:50:52.000000000","updated":"2020-07-16 19:34:09.000000000","submitted":"2020-07-16 19:34:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741376-1594928049042-56289871","meta_rev_id":"11fc4cc5e6b61be7e3679783f17baa195cad7038","_number":741376,"virtual_id_number":741376,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-07-16 18:09:24.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-07-16 10:32:50.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-16 12:12:06.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 19:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-07-16 17:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-07-16 18:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2020-07-16 18:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 10:32:50.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-16 12:12:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-16 17:57:42.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-07-16 18:41:38.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-07-16 19:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfaaaf775e8402c5de23b420bd87b3e41e6997da","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 05:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2764a55e16115b2e9e8095f464a4e00067881117","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 06:48:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d72cbeffd4fddae4e56c5368047fb3f07b98d5d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:28:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75353bb272e34fc182ba1dd96f442f37 : SUCCESS in 7m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5f6a58ea4a2f4ebf91ba441280d60937 : SUCCESS in 40m 11s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8577241d0534420780b5b0cc46feb987 : SUCCESS in 43m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/11f74af4416b4194b09bea69fd3580e2 : SUCCESS in 1h 10m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/945268fe7cae4599b0886c0fdad9554d : SUCCESS in 1h 20m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9d4200746fb549eaa21b616368846849 : SUCCESS in 48m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1d7f6dc0ac7244409c733ce2d74af505 : SUCCESS in 47m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ac9e39d8350e4464ab641f6654d77416 : FAILURE in 1h 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a353280cec905b31a9eced5c0615d3ff1ff21136","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-16 10:32:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/76/741376/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c005f92/ : SUCCESS in 2h 04m 01s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/76/741376/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/37c188c/ : SUCCESS in 1h 24m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d350b573d9900f21633bfc8ce980415b9598ef70","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-16 12:12:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/76/741376/2/check/DellEMC_Unity_Manila_dhss_true/7487587/DellEMC_Unity_Manila_dhss_true/198 : FAILURE in 1h 11m 50s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/76/741376/2/check/DellEMC_Unity_Manila_dhss_false/b8d456a/DellEMC_Unity_Manila_dhss_false/183 : SUCCESS in 56m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c4a6cf00e1395485b713093379b06d11008d4d4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-07-16 17:57:42.000000000","message":"Patch Set 2: Code-Review+2\n\nFix looking good. Thanks, Goutham.\nIs there a launchpad bug for this issue? Or there is no need to have one? :)","accounts_in_message":[],"_revision_number":2},{"id":"d7c8a82dd683166cc397600f4d5100681f537f6e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 18:09:24.000000000","message":"Patch Set 2:\n\n\u003e Is there a launchpad bug for this issue? Or there is no need to have one? :)\n\nWe haven\u0027t released manila-tempest-plugin yet - lets fix it before anyone notices","accounts_in_message":[],"_revision_number":2},{"id":"d14f33b62671c86b552fec791ca507ad576bd770","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-07-16 18:15:31.000000000","message":"Patch Set 2:\n\n\u003e \u003e Is there a launchpad bug for this issue? Or there is no need to\n \u003e have one? :)\n \u003e \n \u003e We haven\u0027t released manila-tempest-plugin yet - lets fix it before\n \u003e anyone notices\n\nAlright! Thanks","accounts_in_message":[],"_revision_number":2},{"id":"450ca5af167b42b0ea2b4abb0d964ab1bfe46a95","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-16 18:41:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"e51d23c235bfd6dac152c8a208ce9fc579513809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:42:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b09ed5fd92f9e86d478deff7ac8e496767977123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 19:34:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62a4adaefe549a5b3524decc9ee33a1 : SUCCESS in 7m 47s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e95605261dbd4a09a5eb4b30de06f468 : SUCCESS in 47m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/91b4df2e35c046faa64c878532b0bf17 : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9eb19a6a213f006bcc47df956e1546972ace6694","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 19:34:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"517a7a250770e4b43940eec637de97b126672a3c","revisions":{"e9b488c6d4d070472ecb57fba5532cfda4a01bed":{"kind":"REWORK","_number":1,"created":"2020-07-16 05:50:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/741376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/741376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/1"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 05:48:58.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 05:48:58.000000000","tz":-420},"subject":"Fix more usage of tenant_id in tests","message":"Fix more usage of tenant_id in tests\n\nCommit 38544589c4a5c39e78f5399fbb2f4ccf92fb7b91\nfixed the use of tenant_id in the rules\nand quota tests. There is one more usage\nin the ReplicaQuotasNegativeTest class to fix.\n\nChange-Id: I4fd282673074887cada729b6e4491ee493e230cc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e9b488c6d4d070472ecb57fba5532cfda4a01bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e9b488c6d4d070472ecb57fba5532cfda4a01bed"}]},"branch":"refs/heads/master"},"517a7a250770e4b43940eec637de97b126672a3c":{"kind":"REWORK","_number":2,"created":"2020-07-16 06:48:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/76/741376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/76/741376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/76/741376/2"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 06:48:10.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 06:48:10.000000000","tz":-420},"subject":"Fix ReplicaQuotasNegativeTest","message":"Fix ReplicaQuotasNegativeTest\n\nThis test was not being executed in the CI\ndue to recent changes in manila-tempest-plugin\nand a few bugs went unnoticed.\n\nChange-Id: I4fd282673074887cada729b6e4491ee493e230cc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/517a7a250770e4b43940eec637de97b126672a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/517a7a250770e4b43940eec637de97b126672a3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
