)]}'
{"id":"openstack%2Fmanila-tempest-plugin~741384","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iefcacfb83262eb8441fd524b4703491980b6a9d7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1886988","hashtags":[],"change_id":"Iefcacfb83262eb8441fd524b4703491980b6a9d7","subject":"[ci] Re-enable scenario tests in the LVM job","status":"MERGED","created":"2020-07-16 06:42:20.000000000","updated":"2020-07-18 13:34:26.000000000","submitted":"2020-07-18 13:34:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741384-1595079266039-bfbdc47d","meta_rev_id":"ccbad2287833ab77baba94728703fdecc529836d","_number":741384,"virtual_id_number":741384,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-07-17 20:05:31.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-07-17 20:53:21.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 13:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-18 11:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"date":"2020-07-18 04:17:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-18 11:11:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 18:56:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-07-17 20:05:31.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-17 20:53:21.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-18 04:17:14.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-07-18 11:11:54.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-07-18 13:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82a3c8ded6a7f375ed7fbbc7de0a3790bcf4ff71","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 06:42:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"325edaffaddec53d1a3cfbf08c25af550eb77ba2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 06:42:33.000000000","message":"Topic changed from bug/1886668 to bug/1886988","accounts_in_message":[],"_revision_number":1},{"id":"273723eefd0cccafe7a9826b997b84f66461dc17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b0eacd96a2bd436eaeefc68661b7cba9 : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":1},{"id":"707aec94cad16aa53d9c9ceac0b7aae5b257c801","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-16 11:00:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/84/741384/1/check/DellEMC_Unity_Manila_dhss_true/4dec79f/DellEMC_Unity_Manila_dhss_true/197 : FAILURE in 1h 13m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/84/741384/1/check/DellEMC_Unity_Manila_dhss_false/c3e4fe4/DellEMC_Unity_Manila_dhss_false/182 : SUCCESS in 58m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6e878144646c668dbbbf29e02f55821469c6a4e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-16 11:15:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/84/741384/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/859b0ee/ : FAILURE in 1h 29m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/84/741384/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/da835f5/ : SUCCESS in 1h 34m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bb6319471e9c8ba15d03e518c94df94424644a63","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 15:57:14.000000000","message":"Patch Set 1:\n\nrecheck\n\nThis landed on rax-ord, and passed with kernel version 4.15.0-111. \n\nLet\u0027s try again and see this wasn\u0027t a fluke.","accounts_in_message":[],"_revision_number":1},{"id":"82dee70c76af5904c933a92bd00c65a6325c844b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 17:14:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fa060ce43583419ba122a5227a900c71 : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4e048014dd6d687d22daaf1418451f6cacf2a707","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-16 17:38:15.000000000","message":"Patch Set 1:\n\nrecheck\n\nLanded on OVH","accounts_in_message":[],"_revision_number":1},{"id":"837c6709d1e17570aa576fe2c365f1c00bae9051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:52:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/141eeff4ebdf44aa906b63a21381cbbb : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"823bb89649a42102d046758f366aa7dd619dad6f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-07-16 18:56:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/84/741384/1/check/DellEMC_Unity_Manila_dhss_true/75daba0/DellEMC_Unity_Manila_dhss_true/202 : FAILURE in 2m 31s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/84/741384/1/check/DellEMC_Unity_Manila_dhss_false/2df0a3e/DellEMC_Unity_Manila_dhss_false/187 : SUCCESS in 54m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"69b3eecd33f3c5fcfec9b810132274f103efa7d0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-16 19:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/84/741384/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/151a107/ : SUCCESS in 2h 09m 47s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/84/741384/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4e0ea51/ : SUCCESS in 1h 27m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f7b98ad593ac6f80b8ce43233d17dba0d4a49042","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-17 17:24:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"49f8d5cbde9999b5e3014c278b84da3cbc9476ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 19:39:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68013c56b2424b6985f3ac9b3b514838 : SUCCESS in 4m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b4fc1672fefa4ab88754ab7e4df3c92f : SUCCESS in 58m 22s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e74bc1fc072540728e53012e11667754 : SUCCESS in 40m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/96074ddc31c04b3db022aa90fab2c90c : SUCCESS in 1h 01m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d9d802921eac4d6599902dd042674749 : SUCCESS in 1h 03m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/944ea0cc4d2344b7a1f94c8fb99c8d54 : SUCCESS in 1h 03m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c85c24242167432c972ad3a6e08510b6 : SUCCESS in 49m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e3b7592bb238495abefafff2b42f5654 : FAILURE in 2h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5965f72c1c9319effeda11ab2acf3c0bbe26d611","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-17 20:05:31.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/84/741384/2/check/DellEMC_VNX_Manila/51a14bc/DellEMC_VNX_Manila/85 : FAILURE in 25m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e7082dc63e946eacfbbfeaa28a02feaeaa4658ab","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-17 20:53:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/84/741384/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2e7e6d3/ : FAILURE in 1h 13m 15s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/84/741384/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1f378a3/ : FAILURE in 23m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"132b4f1bba8c6d8fcf3469a63eb65b4efeab45e5","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-18 11:11:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThe kernel update fixes the reboot issue for both repos[1].\nThanks for all the efforts Goutham and Tom.\nLGTM.\n\n[1] https://review.opendev.org/740551","accounts_in_message":[],"_revision_number":2},{"id":"0be0abe062de59e713623e628ef073e6634e9bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 11:12:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86dd35d8a698a09494702b2e454f49853a7c399e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 13:34:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5718ce1de2434bb4fbcb3f1ace09a3 : SUCCESS in 3m 23s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6fd45f364c0d43b188b450963f664877 : SUCCESS in 1h 24m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ed1f29b9a0264c0bb06e2353bd641cb1 : SUCCESS in 1h 53m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ccbad2287833ab77baba94728703fdecc529836d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 13:34:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"170cc45947459384b62a7f899f8c65fa7081bb3e","revisions":{"f57fee857eb32d777449ea77e1774b8d97a343b0":{"kind":"REWORK","_number":1,"created":"2020-07-16 06:42:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/84/741384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/84/741384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/1"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 06:42:05.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 06:42:05.000000000","tz":-420},"subject":"WIP - Re-enable scenario tests","message":"WIP - Re-enable scenario tests\n\nChange-Id: Iefcacfb83262eb8441fd524b4703491980b6a9d7\nRelated-Bug: #1886668\nRelated-Bug: #1886988\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f57fee857eb32d777449ea77e1774b8d97a343b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f57fee857eb32d777449ea77e1774b8d97a343b0"}]},"branch":"refs/heads/master"},"170cc45947459384b62a7f899f8c65fa7081bb3e":{"kind":"REWORK","_number":2,"created":"2020-07-17 17:24:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/84/741384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/84/741384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/84/741384/2"}}},"commit":{"parents":[{"commit":"38544589c4a5c39e78f5399fbb2f4ccf92fb7b91","subject":"Fix usage of tenant_id in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/38544589c4a5c39e78f5399fbb2f4ccf92fb7b91"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-16 06:42:05.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-17 17:24:06.000000000","tz":0},"subject":"[ci] Re-enable scenario tests in the LVM job","message":"[ci] Re-enable scenario tests in the LVM job\n\nBug #1886668 has now been addressed in ubuntu, and\nthe new kernel no longer suffers from the problem\nthat led to mid-test reboots like the previous ones\ndid. So we can safely re-enable scenario tests in the\ngating LVM job.\n\n\nChange-Id: Iefcacfb83262eb8441fd524b4703491980b6a9d7\nRelated-Bug: #1886668\nCloses-Bug: #1886988\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/170cc45947459384b62a7f899f8c65fa7081bb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/170cc45947459384b62a7f899f8c65fa7081bb3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
