)]}'
{"id":"openstack%2Fmanila-tempest-plugin~741929","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1706022d0189e199ba47e296d8084526eb890e20","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1883506","hashtags":[],"change_id":"I1706022d0189e199ba47e296d8084526eb890e20","subject":"allow delete share in manage_error status","status":"ABANDONED","created":"2020-07-20 12:29:16.000000000","updated":"2020-09-25 06:09:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4bed8f48bd57b27e0f59118a93b9b831ac0d6f9","_number":741929,"virtual_id_number":741929,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-16 10:42:20.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"date":"2020-09-16 07:13:27.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"date":"2020-09-16 10:52:33.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-09-16 10:12:42.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-16 09:17:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-09-24 18:08:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 09:17:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 10:12:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-09-16 10:42:20.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-16 10:52:33.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 18:08:51.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"}],"messages":[{"id":"d31d59fb61f155444fff822fde223091e78aadf7","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-07-20 12:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44cdfb042c301ced0a3acc85b34a6e97aca4e5aa","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-07-20 12:32:05.000000000","message":"Topic set to bug/1883506","accounts_in_message":[],"_revision_number":1},{"id":"5e5308fbc37d688e6611761f887b893064ac7e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 14:31:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e2a6c01a5c48ecbb30d2e833d609d5 : SUCCESS in 3m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1d6cbce36b2d451d824e587a5e0672f3 : SUCCESS in 1h 05m 17s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/60c429e862a4474fbd89492db4fbc44a : SUCCESS in 44m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a49364e78adf41dca85af8bbcab46adc : SUCCESS in 1h 04m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1c8be7ae84ad4eeead81c1e99f7a9dbd : SUCCESS in 42m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e01f54a382ab460e9f9d36970ad14c09 : FAILURE in 57m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3f6b4d74b87e42fc847540e2bfe1063a : SUCCESS in 41m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ea193bc2aad14c6391ffbda1697f48e8 : FAILURE in 2h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52c019c97bb70caec1ab7f4f7a2dde196bffd44d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-07-20 15:37:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/29/741929/1/check/DellEMC_VNX_Manila/4f5e15f/DellEMC_VNX_Manila/93 : FAILURE in 37m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"605c1a0749d1c083f2bcdd17371b64fdd5a2da3c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-20 16:33:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/29/741929/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/42548b0/ : SUCCESS in 2h 01m 42s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/29/741929/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2896fd1/ : SUCCESS in 1h 29m 36s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea9b0a00df548249611385a2f63acd15a0dda4e8","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-15 06:39:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2add9f9b51a189bf085403703ee629ace934afe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-15 08:13:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/29/741929/2/check/DellEMC_Unity_Manila_dhss_true/a2e21b9/DellEMC_Unity_Manila_dhss_true/411 : FAILURE in 51m 33s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/29/741929/2/check/DellEMC_Unity_Manila_dhss_false/d4fd880/DellEMC_Unity_Manila_dhss_false/392 : SUCCESS in 47m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"546ca02019a26a473f47665a4166545e921399a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:59:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc89ed0e72d94f4bb46f93b4d6b3dfed : FAILURE in 4m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b49df71272354fdebe4a87fbad172c08 : FAILURE in 14m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e24ba8360c964673b705c9cfe8d0dae5 : FAILURE in 15m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b626dbfded4c428c8ff779ab7e507311 : SUCCESS in 1h 36m 47s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f49dcc82aa7746f3b1795d79427def87 : FAILURE in 1h 04m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aac7491332994d1b89d5de6f491fa024 : SUCCESS in 1h 07m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/03413567055c4709a3c8772598518d9f : FAILURE in 1h 06m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/613ed47a8d78451f8147d6973ab988e2 : SUCCESS in 46m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a399c11fb4df4562b51cea865406e21a : FAILURE in 2h 14m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b56d5081de484a04bf095fa82ae286f8 : SUCCESS in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9373203d19d03a4d9e98985fa9242d2602520f17","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-15 11:18:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"235eda1875e8f5bec9c92d142b2725f90b4c6758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 12:06:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4aacb28f8b743f39e025cd23b4cfd7e : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9e2d9ab832bc48ec847718e27e200aee : POST_FAILURE in 33m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2ccdba996c68438bad3b403a67db3e55 : POST_FAILURE in 13m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/79618bcaac304ec7a8381c3392b0b719 : FAILURE in 11m 25s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/67bf4f2728e34c1eb3c5e3b19573b5c2 : FAILURE in 14m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1481a38cc7d848569d11ed3990129a52 : FAILURE in 12m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a984081335f2442a9c42aa74f48bb23d : FAILURE in 18m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/af2dd51d5d234ed8bd0ed0be2ba8d92e : FAILURE in 10m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/39350a7021834465a6bbea31723b929f : FAILURE in 16m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2dbbf3555f20466f9d0c04059748b0d6 : FAILURE in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"374b91c53844b495fbd3df3c182d5366b89e1206","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-15 12:14:17.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/29/741929/3/check/DellEMC_Unity_Manila_dhss_true/720c597/DellEMC_Unity_Manila_dhss_true/415 : FAILURE in 55m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/29/741929/3/check/DellEMC_Unity_Manila_dhss_false/2dc8fa3/DellEMC_Unity_Manila_dhss_false/396 : SUCCESS in 45m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8f5cd5eba083c2cf539ff868edb9752b454de47a","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-16 01:02:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fcff5bd6644dfb4287e6322479dfa026d409fbaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:09:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e4014588374c5a93a7dd4a831347d7 : SUCCESS in 3m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/27927befb75f47158bfd95c581b15c36 : SUCCESS in 30m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fba4d7217de0459db8017413159c7142 : FAILURE in 40m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a32b162f1fad41a39e9b60de86ec78b8 : SUCCESS in 1h 13m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6b1bce0fa5564420b74a67fd12d37c25 : FAILURE in 1h 05m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/76cfcc93b4354513a15edf25b6aec342 : SUCCESS in 1h 23m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a4ac9a5cedef4ffa9dec3ae1c532426d : FAILURE in 1h 07m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aca90136b4e84e18828ac30278544768 : SUCCESS in 44m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e5a08ac70fc246139336098869ede74a : FAILURE in 2h 01m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/344c7aff13694e9f97af8c7b952a3cde : SUCCESS in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a40c55047489164ab64f302c55e56cb2943e23ef","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-16 07:13:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2effa79fda49fcc34ac695dd008d76e63edcebe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 09:17:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c326721289430d9b1c1e506bf16c52 : SUCCESS in 4m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61f91cd3a04e4fd6872c13a17a5e21b4 : SUCCESS in 39m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d7e741e001814dc9b7d0e715f7c93510 : SUCCESS in 44m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3ecc27c7087c4ab0867cfdc912cc4656 : SUCCESS in 1h 10m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/176b052122c644e18f2b289a61920c0e : FAILURE in 1h 15m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/125d41cb78b44f55b6e441f82bf00dcd : SUCCESS in 1h 07m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/69a8dc11fc5347c5863e397da2ff486d : FAILURE in 52m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0c647eda541d45deb6a96c7c97c13747 : SUCCESS in 37m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1e30564c36df4102b53683676908705d : FAILURE in 1h 51m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/55aed5ffa753436f834f8004bd305efd : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ab71c36aed372ff54cbaa1395812cc24a926070e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-16 10:12:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/29/741929/3/check/DellEMC_Unity_Manila_dhss_true/59f7e31/DellEMC_Unity_Manila_dhss_true/423 : FAILURE in 55m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/29/741929/3/check/DellEMC_Unity_Manila_dhss_false/f11f789/DellEMC_Unity_Manila_dhss_false/404 : SUCCESS in 48m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5d7b1a41c276b80281fdb22cede8ab9d0af1cd48","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-16 10:42:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/29/741929/3/check/DellEMC_VNX_Manila/ae0881e/DellEMC_VNX_Manila/232 : SUCCESS in 1h 24m 56s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"57b89f099e386d91f13d0e50e180fb06a37d7d94","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-16 10:52:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/29/741929/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b7636c0/ : FAILURE in 1h 05m 03s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/29/741929/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/00c9188/ : SUCCESS in 1h 35m 06s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"87b65f491d7ac40e4d34e9d9dbf68540bb17c591","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2020-09-24 18:08:51.000000000","message":"Patch Set 3: Code-Review-1\n\nSee https://review.opendev.org/#/c/741588/8/manila/share/api.py@1107","accounts_in_message":[],"_revision_number":3},{"id":"a4bed8f48bd57b27e0f59118a93b9b831ac0d6f9","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-09-25 06:09:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b579e224b8eecf3a7f627fa96e8a8877c72a784","revisions":{"37579c3098f7bdfd0798c0d115b89412f29649a7":{"kind":"REWORK","_number":1,"created":"2020-07-20 12:29:16.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/29/741929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/741929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/1"}}},"commit":{"parents":[{"commit":"e3d7ef58de6650cfad0c2e07a6990b74dc903043","subject":"Merge \"[ci] Re-enable scenario tests in the LVM job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3d7ef58de6650cfad0c2e07a6990b74dc903043"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-07-20 12:26:56.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-07-20 12:27:16.000000000","tz":480},"subject":"allow delete share in manage_error status","message":"allow delete share in manage_error status\n\nif failed to manage a share, we should allow to\ndelete the manage_error share.so remove the test\nin share_manage_negative.\n\nChange-Id: I1706022d0189e199ba47e296d8084526eb890e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37579c3098f7bdfd0798c0d115b89412f29649a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/37579c3098f7bdfd0798c0d115b89412f29649a7"}]},"branch":"refs/heads/master"},"d680c4621a761664e3e9da7458f8c3b293fd7661":{"kind":"REWORK","_number":2,"created":"2020-09-15 06:39:36.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/29/741929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/741929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/2"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-07-20 12:26:56.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-15 06:39:12.000000000","tz":480},"subject":"allow delete share in manage_error status","message":"allow delete share in manage_error status\n\nif failed to manage a share, we should allow to\ndelete the manage_error share.so remove the test\nin share_manage_negative.\n\nChange-Id: I1706022d0189e199ba47e296d8084526eb890e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d680c4621a761664e3e9da7458f8c3b293fd7661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d680c4621a761664e3e9da7458f8c3b293fd7661"}]},"branch":"refs/heads/master"},"6b579e224b8eecf3a7f627fa96e8a8877c72a784":{"kind":"REWORK","_number":3,"created":"2020-09-15 11:18:25.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/29/741929/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/741929/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/741929/3"}}},"commit":{"parents":[{"commit":"2e12123d97d04113fdc0f23458fa27ee18978da0","subject":"Merge \"Add tests for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e12123d97d04113fdc0f23458fa27ee18978da0"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-07-20 12:26:56.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-09-15 11:17:59.000000000","tz":480},"subject":"allow delete share in manage_error status","message":"allow delete share in manage_error status\n\nif failed to manage a share, we should allow to\ndelete the manage_error share.so remove the test\nin share_manage_negative.\n\nChange-Id: I1706022d0189e199ba47e296d8084526eb890e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b579e224b8eecf3a7f627fa96e8a8877c72a784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b579e224b8eecf3a7f627fa96e8a8877c72a784"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
