)]}'
{"id":"openstack%2Fmanila-tempest-plugin~743057","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ife4fd77a48b4aab1d9a6ecf3117f1d963b55c2fb","project":"openstack/manila-tempest-plugin","branch":"master","topic":"DNM","hashtags":[],"change_id":"Ife4fd77a48b4aab1d9a6ecf3117f1d963b55c2fb","subject":"DNM","status":"ABANDONED","created":"2020-07-26 12:59:13.000000000","updated":"2020-07-26 15:56:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2d38c6f34f2bb5ccb736e986345c0ba0142114a","_number":743057,"virtual_id_number":743057,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-26 14:58:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-26 14:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c815b6373d6a77996f5ea8199922148167e05e0c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-07-26 12:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"362ad114b4a387f1e982d5f52648cde28621a9bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 14:58:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d8cd6aa507408fbe5a3d8e844c9a7d : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f6c3bb57c2d1448dabd49189a0b8c9b1 : SUCCESS in 29m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6836791de0d0454f9f649b19b03319ab : SUCCESS in 29m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/28ee769e3acc4316856d573910eb7020 : SUCCESS in 58m 18s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aeeb9ab4d9c44f2791955c892b418451 : SUCCESS in 1h 09m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/34bdf753d3434f8d901e6f81ada41715 : SUCCESS in 56m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e20aecfbf9cb40e2b465e8812540e101 : FAILURE in 58m 22s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d7b5405bce54942beac7166cf6de36b : SUCCESS in 44m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/005fffde837046478a7447db59828996 : FAILURE in 1h 57m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1c18baa4d7da40f8bfdba1304eddb999 : SUCCESS in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7acc79ebf7268acee87093f94e94588dc5c2049","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-07-26 15:56:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c22d3c5a41962901fa915d16f8c820b63b24643","revisions":{"6c22d3c5a41962901fa915d16f8c820b63b24643":{"kind":"REWORK","_number":1,"created":"2020-07-26 12:59:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/57/743057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/57/743057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/743057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/743057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/743057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/57/743057/1"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-07-26 12:59:02.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-07-26 12:59:02.000000000","tz":180},"subject":"DNM","message":"DNM\n\nChange-Id: Ife4fd77a48b4aab1d9a6ecf3117f1d963b55c2fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c22d3c5a41962901fa915d16f8c820b63b24643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c22d3c5a41962901fa915d16f8c820b63b24643"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
