)]}'
{"id":"openstack%2Fmanila-tempest-plugin~744398","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I95e667745d872a12391c6713b70b4842d7b70984","project":"openstack/manila-tempest-plugin","branch":"master","topic":"share_servers","hashtags":[],"change_id":"I95e667745d872a12391c6713b70b4842d7b70984","subject":"Ensure share server is created when driver handles share servers","status":"MERGED","created":"2020-08-03 06:32:08.000000000","updated":"2020-08-21 14:57:45.000000000","submitted":"2020-08-21 14:57:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744398-1598021865771-01c56cfd","meta_rev_id":"ee9bd83d1000670bd2e4c2ed6b5368e3332cf69d","_number":744398,"virtual_id_number":744398,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-17 08:17:49.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-08-17 10:14:01.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-08-17 09:17:48.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 14:57:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-21 13:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-08-17 15:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 13:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 06:32:55.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-08-03 06:33:10.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-08-03 06:33:22.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-08-17 08:17:49.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-17 09:17:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-08-17 10:14:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-17 15:40:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-08-21 13:28:22.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-08-21 14:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3a202b870dd2597ed13efd9fa538db1c5ebd50c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-03 06:32:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b8be35c424962431d5a98327e39cddb7998964b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-03 07:27:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/98/744398/1/check/DellEMC_Unity_Manila_dhss_true/1b47a9b/DellEMC_Unity_Manila_dhss_true/247 : FAILURE in 55m 35s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/98/744398/1/check/DellEMC_Unity_Manila_dhss_false/1845d0c/DellEMC_Unity_Manila_dhss_false/232 : SUCCESS in 50m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"73f995a112347c19d7510c5cdd0fc66060677045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 09:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4553fc30bfbb4d12a3da02b06c737ae7 : SUCCESS in 3m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/07a03a70debf47d49672e1990962882f : SUCCESS in 29m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/397e7599bb774eee880376e79bfba522 : SUCCESS in 31m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1446e6d7ee6641a19057fb2c4e120d0e : SUCCESS in 1h 03m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c806923a1c04606841b1c8b37fa33e7 : FAILURE in 1h 02m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48810d0b4cbc41648557bac69424c718 : SUCCESS in 52m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/29710f9b546f4f61b3012a7525197ddb : FAILURE in 1h 15m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/327458afd2834a5d8ea6b37c51dd9bc3 : SUCCESS in 41m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ad8c62e470d4482db15eebf5684d28a1 : FAILURE in 1h 47m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6e12090d88aa47f19a71ade94ced353b : FAILURE in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7dbb7ecc098a4d7ab74d3fc0be3a6423eef6c2e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-03 09:58:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/98/744398/1/check/DellEMC_Unity_Manila_dhss_true/d3d0402/DellEMC_Unity_Manila_dhss_true/248 : FAILURE in 54m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/98/744398/1/check/DellEMC_Unity_Manila_dhss_false/acd2ad2/DellEMC_Unity_Manila_dhss_false/233 : SUCCESS in 50m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a27365c5b4eb4ba17c46ba33d42e7f8ef6729de1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-03 10:07:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/98/744398/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/cb4dede/ : FAILURE in 1h 03m 07s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/98/744398/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/c9ca5b4/ : FAILURE in 28m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9e800364c7e371bd688f0e4e55e6f370895eaec4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-03 10:08:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/98/744398/1/check/DellEMC_VNX_Manila/f968eb9/DellEMC_VNX_Manila/129 : SUCCESS in 1h 04m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a1fc323ba5f57a8e7ef95a0d97d88c1071ff5a5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-03 17:48:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ba6f29a267759eb7012db60971ba8e1710cad39","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-03 18:10:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56b216afdf3960ea87f0b1225c865e5dd7e72ac5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-03 19:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4faf54439afa8b56ce3b76d57b3ee4b945a63dea","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-04 05:15:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51f558af975c1be3d2233d203935f165d3b7acc9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-04 05:17:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ca9859498325e517625f3144922b31fd9fe9189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 07:14:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45f0507545b49788be325fcfb2567e3 : SUCCESS in 2m 45s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4e24bbd593764f34b9fc377eb144835a : FAILURE in 33m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/365b33d9d7254d59a1a203f8b15143da : FAILURE in 33m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7e1f8d73818940ba966e63a08153cac9 : FAILURE in 1h 03m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b1cfb1b9d2074e64b5e728e5c9042865 : FAILURE in 55m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/985e09e52aed4118ab5eb88052648828 : FAILURE in 57m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/36cd7af167e64baeae7b733ea3ff3361 : FAILURE in 1h 02m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3e538e997a82479fb338e711462c0c29 : FAILURE in 43m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/81edd08a510841e4a8fbd05f92c77df4 : FAILURE in 1h 55m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/348a2c82885343a8b4955040d1f93c40 : FAILURE in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"578c3cbf659acc797ab8ff6dac9eff761d57a168","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-04 08:12:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3a24c43d4471b8380060141cdd3d9ff51749a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 10:24:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e823a1e6b7f340b88cf540b2907f018a : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3e324e2e2a90448f9e6c76504275921e : SUCCESS in 32m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/24a2f1cf43c94ec1821b60509d4add39 : SUCCESS in 35m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/16ce5ec8f68c461d8eb460de60d8590e : SUCCESS in 1h 28m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a3011d8d2a8045c092e206e03366f228 : FAILURE in 1h 11m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6827b3d9617a406ead1b84b3cd3cb900 : SUCCESS in 1h 03m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2fa37f60e017475ca8ed64b3317269b4 : SUCCESS in 1h 00m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b83baf2646e44f4484ba6bf0fefb8921 : SUCCESS in 50m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/44d22b7809f64cf2aec2914b273831fe : FAILURE in 2h 06m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5bc39edfa58245d29a6b365d508beaa4 : SUCCESS in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d32448209131c909a532a1a7effe59dfdcaaf434","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-04 11:25:49.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/98/744398/3/check/DellEMC_VNX_Manila/5ecd58a/DellEMC_VNX_Manila/132 : SUCCESS in 1h 01m 24s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"9c93f37924efa9d787e2ea8df5a61e973388632c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-04 12:31:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/98/744398/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f7dcae9/ : SUCCESS in 2h 06m 34s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/98/744398/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5015fbb/ : SUCCESS in 1h 36m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b2e052bd451098198a1e949d4878e317c57ac2da","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-12 23:56:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nLGTM, thank you..","accounts_in_message":[],"_revision_number":3},{"id":"97b132fef2cd1bfa84c81adefd1c40baac3f3145","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-17 05:58:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02ba177f78586840e8c87f8c9f0745f4bdfd6015","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-17 06:01:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae5d73c29b6164750d613e0d077e58dcdee3a5d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-17 07:13:38.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/98/744398/4/check/DellEMC_Unity_Manila_dhss_true/d06daed/DellEMC_Unity_Manila_dhss_true/272 : FAILURE in 1h 14m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/98/744398/4/check/DellEMC_Unity_Manila_dhss_false/3fdd3fe/DellEMC_Unity_Manila_dhss_false/256 : SUCCESS in 53m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c52e91c0c56bb2a1a0fd9e0fc058e271189d8818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 08:02:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da3bc8082bb2491da2497908cfc531cd : SUCCESS in 5m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/db168b01e59542f8997d7b43fe92e817 : SUCCESS in 30m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/de0a0bef6da3454da78a2f6f23a57a46 : SUCCESS in 31m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/05e3c115d1774582aa7c7462534e21f7 : SUCCESS in 1h 02m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7d658eee4d834d42bbc38e096dc83796 : SUCCESS in 1h 04m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/253d02be9b7540e09ce96e53478e60a5 : FAILURE in 1h 50m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a382fcc135de4126b16f81070fcb29dc : SUCCESS in 56m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2f941881f7324058bf4741a9eb5df668 : SUCCESS in 53m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/986123f990a24afd97efbe6cb6960c86 : FAILURE in 1h 38m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/636a1e61d01b4adcb20827c3265e956d : SUCCESS in 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4f0d8a8e6b9ada2a0494ef312030fdb1fa3048e0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-17 08:17:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/98/744398/4/check/DellEMC_VNX_Manila/7bba9b9/DellEMC_VNX_Manila/154 : FAILURE in 14m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"9dd9fc3589f56f1aaccbe1b76cd8a29502ed6f8e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-17 09:17:48.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/98/744398/4/check/DellEMC_Unity_Manila_dhss_true/f59bb43/DellEMC_Unity_Manila_dhss_true/273 : FAILURE in 1h 14m 50s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/98/744398/4/check/DellEMC_Unity_Manila_dhss_false/2475c81/DellEMC_Unity_Manila_dhss_false/257 : SUCCESS in 1h 02m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"22b23a38e47fb448baa6bca3b755e5605a6772d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-17 10:14:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/98/744398/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b9b9869/ : SUCCESS in 2h 09m 46s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/98/744398/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4440299/ : SUCCESS in 1h 37m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"f285a0b76559f7d1fe87d06886bd812836f8f4b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-17 15:40:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cdc517fc176902c1ca5d93b31b8cc04772feba03","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-21 13:28:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks Liron, LGTM!","accounts_in_message":[],"_revision_number":4},{"id":"b2ee14a2632199dd74e314fb0506c84b300bc953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 13:28:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7fe8b6b13459bb662be273edafa5b31a9a94755d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:57:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3e99ce0fe00a4f559c9b9bdee6e219ae288b84d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:57:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e760d8f1218c4ae28c8fd43e668538a1 : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d1111b5065b249d29d7f6c419f1aee6b : SUCCESS in 29m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bc7372ff7bcf473eab8b76292b98b41d : SUCCESS in 29m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c9525c0c4d3a48bdb1e3e1fd578eca43 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6db73a34356364fb9c09e977b51e686201bf2b28","revisions":{"6a9eddf4ed84bda9c76bafa3301d4600ef85fd91":{"kind":"REWORK","_number":1,"created":"2020-08-03 06:32:08.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/98/744398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/98/744398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/1"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-03 06:27:23.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-03 06:27:23.000000000","tz":180},"subject":"Ensure share server is created when driver handles share servers","message":"Ensure share server is created when driver handles share servers\n\nThis test will ensure that when a share creation request is handled,\nif the driver has the \"driver handles share servers\" option enabled,\nthat a share server will be created, otherwise, not.\n\nChange-Id: I95e667745d872a12391c6713b70b4842d7b70984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a9eddf4ed84bda9c76bafa3301d4600ef85fd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a9eddf4ed84bda9c76bafa3301d4600ef85fd91"}]},"branch":"refs/heads/master"},"c17a6392d5076664d14229620f984cc6e66dd89b":{"kind":"REWORK","_number":2,"created":"2020-08-04 05:17:21.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/98/744398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/98/744398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/2"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-03 06:27:23.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 05:16:55.000000000","tz":180},"subject":"Ensure share server is created when driver handles share servers","message":"Ensure share server is created when driver handles share servers\n\nThis test will ensure that when a share creation request is handled,\nif the driver has the \"driver handles share servers\" option enabled,\nthat a share server will be created, otherwise, not.\n\nChange-Id: I95e667745d872a12391c6713b70b4842d7b70984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c17a6392d5076664d14229620f984cc6e66dd89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c17a6392d5076664d14229620f984cc6e66dd89b"}]},"branch":"refs/heads/master"},"9ec92ce05f6aa9196f5cb3799b2ed691da9d871b":{"kind":"REWORK","_number":3,"created":"2020-08-04 08:12:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/98/744398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/98/744398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/3"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-03 06:27:23.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:12:06.000000000","tz":180},"subject":"Ensure share server is created when driver handles share servers","message":"Ensure share server is created when driver handles share servers\n\nThis test will ensure that when a share creation request is handled,\nif the driver has the \"driver handles share servers\" option enabled,\nthat a share server will be created, otherwise, not.\n\nChange-Id: I95e667745d872a12391c6713b70b4842d7b70984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ec92ce05f6aa9196f5cb3799b2ed691da9d871b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ec92ce05f6aa9196f5cb3799b2ed691da9d871b"}]},"branch":"refs/heads/master"},"6db73a34356364fb9c09e977b51e686201bf2b28":{"kind":"REWORK","_number":4,"created":"2020-08-17 05:58:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/98/744398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/98/744398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/744398/4"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-03 06:27:23.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-17 05:58:33.000000000","tz":180},"subject":"Ensure share server is created when driver handles share servers","message":"Ensure share server is created when driver handles share servers\n\nThis test will ensure that when a share creation request is handled,\nif the driver has the \"driver handles share servers\" option enabled,\nthat a share server will be created, otherwise, not.\n\nChange-Id: I95e667745d872a12391c6713b70b4842d7b70984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6db73a34356364fb9c09e977b51e686201bf2b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6db73a34356364fb9c09e977b51e686201bf2b28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
