)]}'
{"id":"openstack%2Fmanila-tempest-plugin~744660","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4feabcd4db583d213e0e4d414eee267b0ddb8870","project":"openstack/manila-tempest-plugin","branch":"master","topic":"import_order","hashtags":[],"change_id":"I4feabcd4db583d213e0e4d414eee267b0ddb8870","subject":"Add flake8-import-order extension to tox.ini","status":"MERGED","created":"2020-08-04 08:23:12.000000000","updated":"2020-09-01 15:12:39.000000000","submitted":"2020-09-01 15:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":78,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744660-1598973159495-42a114cc","meta_rev_id":"8fef9331248dbf10cddad043851a49efbc2f6828","_number":744660,"virtual_id_number":744660,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-18 10:23:18.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"date":"2020-08-18 09:31:46.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-08-18 09:21:50.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 15:12:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"date":"2020-08-18 07:29:57.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-18 17:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-01 14:00:23.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 14:00:23.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 08:24:33.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2020-08-04 15:09:04.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-08-18 09:21:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-08-18 09:31:46.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 10:23:18.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-18 17:34:36.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-09-01 14:00:23.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-01 15:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0c00e853e94fa2982ad4080037f0b2855f2c499","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-04 08:23:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e9b2f1d1c4f72857717703ad1464d7a8b1ab244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 08:32:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d353ad2951694222a517015b38efb19b : FAILURE in 4m 15s\n\nWarning:\n  Comments left for invalid file manila_tempest_tests/plugin.py\n  Comments left for invalid file manila_tempest_tests/utils.py\n  Comments left for invalid file manila_tempest_tests/services/share/v2/json/shares_client.py\n  Comments left for invalid file manila_tempest_tests/services/share/json/shares_client.py\n  Comments left for invalid file manila_tempest_tests/common/remote_client.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_network_subnets.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_networks_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_snapshot_rules_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_actions_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_access_rules_metadata.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_type_availability_zones.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_from_snapshot_across_pools.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_revert_to_snapshot.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_extensions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_metadata_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_type_availability_zones_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_types_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_limits.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_replication_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_security_services_mapping_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_metadata.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_access_rules_metadata_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_scheduler_stats_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_group_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_rules.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_quotas.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_rules_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_availability_zones_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_security_services_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_snapshot_rules.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_quotas_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_security_services_mapping.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_security_services.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_microversions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_availability_zones.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_replication_snapshots.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_replication_export_locations.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_shares.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_groups.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_replication.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_revert_to_snapshot_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_replication_export_locations_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/base.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_network_subnets_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_networks.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_groups_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_user_messages.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_group_types_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_group_types.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_multi_backend.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_servers_manage.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_services_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_export_locations_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_migration_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_export_locations.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_types_extra_specs.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_types_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_shares_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_admin_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_manage.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_snapshot_export_locations.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_scheduler_stats.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_instances.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_replication_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_admin_actions_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_servers_manage_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_user_messages_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_quotas.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_snapshot_instances_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_snapshot_instances.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_quotas_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_snapshot_manage_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_servers.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_security_services.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_snapshot_manage.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_types.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_types_extra_specs_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_instances_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_manage_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_snapshot_export_locations_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_services.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_groups.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_replication.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_migration.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_networks.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_servers_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_groups_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_shrink.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_extend.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/manager_share.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_basic_ops.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/manager.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_manage_unmanage.py","accounts_in_message":[],"_revision_number":1},{"id":"828192cdd2364c8618e822c4f5f298deb464751d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-08-04 15:09:04.000000000","message":"Patch Set 1:\n\nMaybe we can try with https://pypi.org/project/isort/","accounts_in_message":[],"_revision_number":1},{"id":"7fdf101f83db02007db454949d8373f2ff2f3ace","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-04 18:17:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThanks for starting on this, lets fix this and test it properly - I don\u0027t believe we have a ton of transgressions..","accounts_in_message":[],"_revision_number":1},{"id":"bc160e0f1c9cef0c87eb61f047f060b5ac87fff8","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-05 04:37:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aee7c7d484f3cc0572b0824bf85cb9bd4bf4891f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 04:42:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3250d6fa52844ccdac779a3abc96f08c : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad69bbc3e0a44ecbd87fbf3713bdd2d : FAILURE in 3m 50s\n\nWarning:\n  Comments left for invalid file manila_tempest_tests/utils.py\n  Comments left for invalid file manila_tempest_tests/services/share/json/shares_client.py\n  Comments left for invalid file manila_tempest_tests/services/share/v2/json/shares_client.py\n  Comments left for invalid file manila_tempest_tests/common/remote_client.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_shrink.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/manager_share.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_basic_ops.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_extend.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/manager.py\n  Comments left for invalid file manila_tempest_tests/tests/scenario/test_share_manage_unmanage.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_types_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_network_subnets_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_quotas.py\n  Comments left for invalid file manila_tempest_tests/tests/api/base.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_revert_to_snapshot_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_share_groups_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_rules.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_revert_to_snapshot.py\n  Comments left for invalid file manila_tempest_tests/tests/api/test_snapshot_rules.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_quotas_negative.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_snapshot_manage.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_admin_actions.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_manage.py\n  Comments left for invalid file manila_tempest_tests/tests/api/admin/test_share_group_types.py","accounts_in_message":[],"_revision_number":2},{"id":"c4fca13496a32db26dfed727a7c02153fb406032","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-05 04:58:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2e90139e884a9b218ce957438b6a7284eb34dac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 06:49:44.000000000","message":"Patch Set 3: Verified-1\n\n(44 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/58dbbc2e70b34b5586a9d216c843d067 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6d210f348b433eb8f82262f3e59409 : FAILURE in 3m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7a8ae8df90e94b4184ba02ee6eedbd69 : SUCCESS in 36m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a1c9aa6c39df44fdb1b21133298a5b03 : SUCCESS in 28m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9565e0cd110a4efea20847e12036f68d : SUCCESS in 1h 18m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6fe41e0f028a435eb5a17deedc76de64 : FAILURE in 59m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8aa55d46c3c54eda970dda81bc5d2f31 : SUCCESS in 51m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e9e519d3b3e24f46b0a4072e73a7cafe : FAILURE in 54m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1124b4e966ae4a3a918066b7e0b69059 : SUCCESS in 40m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/05e3934341fa43cfaf9d68813020e673 : FAILURE in 1h 50m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ff393033c70f4f01bc89d5f480eb5287 : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e979724ac4dea90dc1f379f571bbb5d7df14b3ce","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-05 09:15:59.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e Maybe we can try with https://pypi.org/project/isort/\n\nRight, there is another one https://pypi.org/project/reorder-python-imports/\nBut there are still errors. Maybe I didn\u0027t use it properly for our requirements.","accounts_in_message":[],"_revision_number":1},{"id":"a7f03c2d4588d61b7b54fc725c3b2a484a9cc3f2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-05 09:17:19.000000000","message":"Patch Set 3:\n\nPS3 is after using isort tool.","accounts_in_message":[],"_revision_number":3},{"id":"3de33581bda1909409321798fe4f68a6a5b7c5e9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-05 17:54:18.000000000","message":"Patch Set 3:\n\n\u003e PS3 is after using isort tool.\n\nHmmm, manual fixups may be better then..","accounts_in_message":[],"_revision_number":3},{"id":"4bedb145c90f9e00060ae1f38b035a8be3d48d94","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-17 11:08:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebecef8054be0a78563119bb22eecac5bf713402","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-17 11:49:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6278c49a99325fd6228f8dfe800a5263a3e89155","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-17 12:27:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d67936d19e6b0088835f2db2665e5aa585509054","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-17 13:45:56.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/60/744660/6/check/DellEMC_Unity_Manila_dhss_true/aa35a05/DellEMC_Unity_Manila_dhss_true/276 : FAILURE in 1h 18m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/60/744660/6/check/DellEMC_Unity_Manila_dhss_false/3af946b/DellEMC_Unity_Manila_dhss_false/260 : SUCCESS in 53m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"c2c66178920c96f10db97a155f72b22ae6e12df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 14:37:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/040f7f9b42f349fa85be8e9a021affe2 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102467d3d6d74c1f92246289483c743f : SUCCESS in 4m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6850591d18be46bdb7514b5ebcb6fcb3 : SUCCESS in 27m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/63ca3ea7853742e3bdfde163ed20e216 : SUCCESS in 36m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a536ba742bfd4fffb6d994883ac58f68 : SUCCESS in 1h 00m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9acccbc0bdc7403a969ed5d9f86950ec : FAILURE in 1h 19m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b8f969358a07418c88ef16c1721d2a99 : FAILURE in 1h 26m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2c760e644eec40f5af9cd8d26766c252 : FAILURE in 1h 04m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/baf44b7cea4b4886bf2dd7c57478f314 : SUCCESS in 40m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87eba10180044b22aa360e3e79064b72 : FAILURE in 2h 08m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/18aaef54d1814bbc9033489ad9232c95 : SUCCESS in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ddedbfbe018cedeaed8bc4a3ce5549da9149ba1d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-17 14:55:45.000000000","message":"Patch Set 6: Code-Review-1\n\n(30 comments)\n\nThanks, this is a great improvement - but, there are several issues - and I think this may be the case even with files you\u0027ve not included in this change. I know its a lot of manual work to get these in order, I\u0027m okay with splitting the change into multiple changes... wdyt?","accounts_in_message":[],"_revision_number":6},{"id":"dc6aa541dbcc79552deaa8227abf7d3267ee8da9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-17 15:41:50.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/60/744660/6/check/DellEMC_VNX_Manila/69a7b98/DellEMC_VNX_Manila/155 : FAILURE in 1h 04m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"9d1f0cda74507d4ae4f7ed4745631ee80aab99e1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-17 17:24:10.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/60/744660/6/check/DellEMC_Unity_Manila_dhss_true/ad1afe0/DellEMC_Unity_Manila_dhss_true/279 : FAILURE in 1h 09m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/60/744660/6/check/DellEMC_Unity_Manila_dhss_false/817667d/DellEMC_Unity_Manila_dhss_false/263 : SUCCESS in 51m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"4886ab5318eb5abab58d6ebadb999590230b79d0","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-18 07:01:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eeb400a871c53b2b291a586d038879f408d010cf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-08-18 07:29:57.000000000","message":"Patch Set 7:\n\nThanks for your comments.\nI wanted to isolate errors which is why I ignored I202.\nI think and hope this iteration will work.","accounts_in_message":[],"_revision_number":7},{"id":"db6e1fdf5a02ef84cacb199bc6df90890c82e4d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-18 08:43:16.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/60/744660/7/check/DellEMC_Unity_Manila_dhss_true/731f5da/DellEMC_Unity_Manila_dhss_true/284 : FAILURE in 1h 19m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/60/744660/7/check/DellEMC_Unity_Manila_dhss_false/41dd9b3/DellEMC_Unity_Manila_dhss_false/268 : SUCCESS in 55m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"e88fdd9d2bef8cd3465e0f0a820f9ae9601473f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 09:05:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c0f6bd39f6f43978c59a739b167c0ac : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26fd5190418c49e8b6029a55f15ce64f : SUCCESS in 4m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e746add1ae2041e3a98f3ca691679208 : SUCCESS in 31m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/02f0086403e74ac6b0d6f7deb83ea266 : SUCCESS in 34m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dfa53fca7abc45959ef4259e68b87d89 : SUCCESS in 1h 01m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5cd3cb9d75cf42a5b9ee1f77783bbc6a : SUCCESS in 1h 04m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/945518f419b34c28b2eea14f3f53a05e : FAILURE in 1h 32m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/27812266777b4fdbb9769ff05db6ed8a : SUCCESS in 1h 07m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/24116d94faf44400b833b99e78b78ab9 : SUCCESS in 45m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b47f5e83ffe74a01849a74b5863d636f : FAILURE in 1h 48m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7dedec59be7b41eabffddabc2ac10f9f : SUCCESS in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"65cc1a844617d0236435e621014cf22e0e27e77f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-08-18 09:21:50.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/60/744660/7/check/DellEMC_Unity_Manila_dhss_true/5e6b479/DellEMC_Unity_Manila_dhss_true/285 : FAILURE in 16m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/60/744660/7/check/DellEMC_Unity_Manila_dhss_false/6a0c8e5/DellEMC_Unity_Manila_dhss_false/269 : FAILURE in 16m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"7ab8ba8268f637170e5de1333aa9f1e22d9891e8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-18 09:31:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/60/744660/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8f6906b/ : FAILURE in 26m 28s\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/60/744660/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a6f5ee6/ : FAILURE in 1m 12s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"3411eea150d0a1c49878b1134a0276572554e1a4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-18 10:23:18.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/60/744660/7/check/DellEMC_VNX_Manila/7ca425d/DellEMC_VNX_Manila/156 : SUCCESS in 1h 18m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"97bd0de113c8a9ff4349656246227af398cb8d3c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-18 17:34:36.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":7},{"id":"705e423060b977f12fe4d026c3a38e0d66e828ff","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-01 14:00:23.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c9381cd5c0850520406f45fff8c52eed929528ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:00:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8da5b3b55f8034445818ea7548f3c4cffa7d91db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:12:39.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/607b994e04324ac8ad8d6fb707aa35e0 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db3af83b60d24e489f9b845fed3955b2 : SUCCESS in 5m 32s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/39c76b10e2a24d4fa15abba7f0b7f23d : SUCCESS in 34m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f159ae5df2b04de78c045d2143d3ca7b : SUCCESS in 41m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2fc4ed76fa3943c68034ec8a84c9eb1e : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":7},{"id":"c8f402852f7ac667a5b28ba5b2dba4e894bde298","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:12:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1d1461d23220dbd2a5d3420f7802f51992c994f7","revisions":{"21ec07c8ad27d1bffc9a78f7e511e3b95beeb493":{"kind":"REWORK","_number":1,"created":"2020-08-04 08:23:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/1"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21ec07c8ad27d1bffc9a78f7e511e3b95beeb493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/21ec07c8ad27d1bffc9a78f7e511e3b95beeb493"}]},"branch":"refs/heads/master"},"0e5d11b0f61af5b82f8be50545006d2f0ebb6f42":{"kind":"REWORK","_number":2,"created":"2020-08-05 04:37:30.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/2"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 04:37:14.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e5d11b0f61af5b82f8be50545006d2f0ebb6f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e5d11b0f61af5b82f8be50545006d2f0ebb6f42"}]},"branch":"refs/heads/master"},"860c134643d24ea5cc10706a64ab7f62bb7c376c":{"kind":"REWORK","_number":3,"created":"2020-08-05 04:58:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/3"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-05 04:58:00.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/860c134643d24ea5cc10706a64ab7f62bb7c376c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/860c134643d24ea5cc10706a64ab7f62bb7c376c"}]},"branch":"refs/heads/master"},"4af397cbdd122b1a206083445e68a23e071b4fb7":{"kind":"REWORK","_number":4,"created":"2020-08-17 11:08:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/4"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-17 11:08:10.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4af397cbdd122b1a206083445e68a23e071b4fb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4af397cbdd122b1a206083445e68a23e071b4fb7"}]},"branch":"refs/heads/master"},"7efa6b0a1994abb2f98c20ba30afa3cb8999acc1":{"kind":"REWORK","_number":5,"created":"2020-08-17 11:49:09.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/5"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-17 11:48:56.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7efa6b0a1994abb2f98c20ba30afa3cb8999acc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7efa6b0a1994abb2f98c20ba30afa3cb8999acc1"}]},"branch":"refs/heads/master"},"22f4215823882d94ebdcec2c8e00369a452e13ea":{"kind":"REWORK","_number":6,"created":"2020-08-17 12:27:11.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/6"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-17 12:26:59.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22f4215823882d94ebdcec2c8e00369a452e13ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22f4215823882d94ebdcec2c8e00369a452e13ea"}]},"branch":"refs/heads/master"},"1d1461d23220dbd2a5d3420f7802f51992c994f7":{"kind":"REWORK","_number":7,"created":"2020-08-18 07:01:15.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/60/744660/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/744660/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/744660/7"}}},"commit":{"parents":[{"commit":"7674f966482117ead95534875fa8116674f37f81","subject":"Merge \"Zuulv3 native \"standalone\" GlusterFS native, NFS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7674f966482117ead95534875fa8116674f37f81"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-04 08:19:11.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-08-18 07:01:00.000000000","tz":180},"subject":"Add flake8-import-order extension to tox.ini","message":"Add flake8-import-order extension to tox.ini\n\nThis package adds 3 new flake8 warnings\n\n* ``I100``: Your import statements are in the wrong order.\n* ``I101``: The names in your from import are in the wrong order.\n* ``I201``: Missing newline between sections or imports.\n\nChange-Id: I4feabcd4db583d213e0e4d414eee267b0ddb8870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1d1461d23220dbd2a5d3420f7802f51992c994f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1d1461d23220dbd2a5d3420f7802f51992c994f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
